)]}'
{"id":"openstack%2Frequirements~988925","triplet_id":"openstack%2Frequirements~master~Ib14fa7c6e15a3c37053c0c2b5f0dd5cdf7e0378d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib14fa7c6e15a3c37053c0c2b5f0dd5cdf7e0378d","subject":"update constraint for oslo.service to new release 4.6.0","status":"MERGED","created":"2026-05-18 09:43:04.000000000","updated":"2026-05-18 20:45:28.000000000","submitted":"2026-05-18 20:43:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988925-new-release","meta_rev_id":"a204427250537667da7cc4fa83356e57f701d145","_number":988925,"virtual_id_number":988925,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-18 20:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-18 17:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-18 17:21:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 11:27:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-18 17:21:31.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"43bd85a0473545f66f7f0e8b8a4af6f406dbec1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-18 09:43:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"515b815a100a208f47a855a09153af0ce13b6304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 11:27:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ee5343204c04676a41669bfd17215e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad4c945e3c04cbfbb60fb4b2641020e : SUCCESS in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283c2e92ab294058a208e6374b061c23 : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25ba143579cd430b94ab576d8f8d6d92 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/582008781ccf49be9530f15abe902aed : SUCCESS in 2m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1dd025b8917743f7a737ff64e8016589 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a6f34abeb894a499cced8ff579ad0ae : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fccbcdf8c93b4460994d89b6a3b4335b : SUCCESS in 2m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9d4252d0e2f24c90bab90befff22de37 : SUCCESS in 5m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d6d9e6ef99734f308996496abb1e230c : SUCCESS in 3m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d1127feab96c45908d047677fb9660c8 : SUCCESS in 8m 48s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/d6fbbd91c3534213815623e884558f81 : SUCCESS in 4m 23s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/effacfbd3b1f4f6a8af47dfa3c8c0d84 : SUCCESS in 4m 43s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/4c9cf9ea4d0b40d2baf8044a48334064 : SUCCESS in 5m 11s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/f8267a7df84a4da6879e8d969e36759e : SUCCESS in 10m 58s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/5baaaddba0664102beaa25d25538d46b : SUCCESS in 7m 28s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/fefca78222f84fad9171d490926a9086 : SUCCESS in 5m 21s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/23aa4924358745a3ba7b257d7d2dfa7e : SUCCESS in 4m 23s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/220fb174dddf4eb08197685d747a871c : SUCCESS in 7m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c4fb4fdacbb74d9782d3e4cbac66fa50 : SUCCESS in 6m 28s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/59cd793a05c645ee86e524520dd96188 : SUCCESS in 8m 10s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/dd363bdb43214e5197c9f8f32b8dc5c9 : SUCCESS in 11m 14s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/73a0f11805354db2997b82d11911e5a5 : SUCCESS in 2m 41s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/81edbc552edf454d81a17d7990dfe9d4 : SUCCESS in 9m 31s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/e636da188fc14d8ea52d6d403483ad44 : SUCCESS in 4m 02s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/c49754e0d5dd4c50ab3ff13abd1e6c16 : SUCCESS in 4m 29s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/b4c93982de924b6a852b0dc7fc05a4ea : SUCCESS in 11m 24s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/911b8adc6ca14dfd811ff5a1869f0807 : SUCCESS in 24m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/be033a73472445c4a022d963efe8d2f7 : SUCCESS in 3m 52s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/9cbc7ee199c644d28a2d87d99fab9d7c : SUCCESS in 14m 25s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/e447dfa5efa74310b85165bba680875a : SUCCESS in 4m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2842478f1dff436c97cf28af7ddcdac4 : SUCCESS in 1m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/866161a2146643ec8c9b6d3240aa1dda : SUCCESS in 25m 26s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/b559a6a760e14a41905ca4c17b9ffba9 : SUCCESS in 6m 40s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/25e2c2dcb1da4d34b99b44c73a13b694 : SUCCESS in 2m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a52c3073a12c4188a217489dbea8dd78 : SUCCESS in 6m 52s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/0391fcddb4524618931b196239ea829c : SUCCESS in 7m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/abdba37ed37e4f9e8a6b0c3e550ec067 : SUCCESS in 5m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/939be18a90934f72953dfba459dee83c : SUCCESS in 1h 10m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4993124c7db94985b823052c183ab222 : SUCCESS in 1h 41m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a2932e5beebfe5d1ba3531a59a96248d1d257999","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-18 17:21:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09553dff614f2c09886dd42d3962790cd725d578","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 17:22:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf420f39b1a578f23e9bcdf76d88963eac015e26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 20:43:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69f404aed09848a49e07b1cb957819df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3903beadc73d4aa0a80bf2afeac3097b : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85b983c4bf744090b4118d36a604a349 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd942472822d4ed9a3806bcb32562f6c : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6095f774ea034b85b7c7a2841eb36ec9 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/363128930bc440f2a8e7ef1e156dafd1 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5bb8093744cd4c2d9545e8e829b2fdb1 : SUCCESS in 2m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/972b6c8e21124941a1b939dbfd70bd3c : SUCCESS in 7m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d42b389761b84363b3446594f3210573 : SUCCESS in 5m 54s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d1fb2ecd253d45e29e7eecf7c211623a : SUCCESS in 6m 10s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/be2d48a9ae454bf880c08842a80c7315 : SUCCESS in 11m 05s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/422c94a95029499b8502f3f561f07fb3 : SUCCESS in 4m 59s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/71dd6b3de4c840cb8a3cfaf32fe1ffcb : SUCCESS in 5m 31s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e65d5d7d71a64fbe8fa573ca9a82f3ba : SUCCESS in 8m 07s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/49d9b3f4ba57435c97ecc115b316d660 : SUCCESS in 13m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ec5d6c4fcfa48b09ebf9b7f6ad6a7f8 : SUCCESS in 6m 04s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/3905e4a5ddcb4f2ab17b95961f84dff1 : SUCCESS in 5m 33s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0ba557f51d814a038be7938fc218bcd2 : SUCCESS in 11m 22s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/90346850fb2d464faabbfa7c879d611c : SUCCESS in 4m 07s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/a5134dd1e80345ee9c063e12c92b2d27 : SUCCESS in 7m 13s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/e394c77b9b8e43b4a3f4eb50e6c5d8ae : SUCCESS in 21m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6c8ade38c6f148d6a73f0c15fa2389b3 : SUCCESS in 7m 09s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/359c352d11bb4579b5e24412129a92ef : SUCCESS in 13m 35s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/b9943014f3c24811a31b5c4329eef526 : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f033574fbece423fb5451b9ecb2fd58b : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b76765a3ff44447e92d50cd9557aef05 : SUCCESS in 23m 52s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/3bef3c07a1ba49d8b23c04870a6eb0d3 : SUCCESS in 5m 12s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/9f94c2ccd5ea433a8e65e22d7d0b1315 : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/df06e9875acc41598735fcae2c55f4c1 : SUCCESS in 7m 25s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/74d103e4bcba41659f37d73d031afbe6 : SUCCESS in 19m 52s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ba8b283d9d704141923148d3c8efa4cb : SUCCESS in 6m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/790820bd910a4abda7a588d699dd3952 : SUCCESS in 54m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95b10e56c1c9480daf34a5ed3a3b8014 : SUCCESS in 2h 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"995c666f5b9d505f1f945cc6e1b889661ba96518","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 20:43:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a204427250537667da7cc4fa83356e57f701d145","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 20:45:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa48533ff9f44b54aff84d0fc9f8b1e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df48b16ab9d74b3bbb5fb58c64553e46 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70ac3ab81656ce9eeb5e55d1a91d3268ffb0ab44","revisions":{"70ac3ab81656ce9eeb5e55d1a91d3268ffb0ab44":{"kind":"REWORK","_number":1,"created":"2026-05-18 09:43:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/25/988925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/25/988925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/988925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/988925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/988925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/25/988925/1"}}},"commit":{"parents":[{"commit":"016f93e7ade7ccd65f23b34f720439a65fccb772","subject":"Merge \"update constraint for python-neutronclient to new release 12.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/016f93e7ade7ccd65f23b34f720439a65fccb772"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-18 09:33:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-18 09:33:02.000000000","tz":0},"subject":"update constraint for oslo.service to new release 4.6.0","message":"update constraint for oslo.service to new release 4.6.0\n\nmeta: version: 4.6.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iede52b38fbcd20d82fecdb01520530d3336f9146\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib14fa7c6e15a3c37053c0c2b5f0dd5cdf7e0378d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70ac3ab81656ce9eeb5e55d1a91d3268ffb0ab44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70ac3ab81656ce9eeb5e55d1a91d3268ffb0ab44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
