)]}'
{"id":"openstack%2Frequirements~988931","triplet_id":"openstack%2Frequirements~master~Ibbc4e9341da93f46434b84b043367b500668d416","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-05-19 10:18:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbc4e9341da93f46434b84b043367b500668d416","subject":"update constraint for oslo.log to new release 8.2.0","status":"MERGED","created":"2026-05-18 10:31:46.000000000","updated":"2026-05-19 10:22:10.000000000","submitted":"2026-05-19 10:18:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988931-new-release","meta_rev_id":"0a348cee7fccdb581a1311508e54c28e062399a1","_number":988931,"virtual_id_number":988931,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 10:18:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-19 07:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-19 07:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 12:04:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 07:10:02.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"8eb09d5b4c7e0fb06666ebfa41df0f88ca46b6ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-18 10:31:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"967417eb7eb39e6bf63aad953757cf666d66f98d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 12:04:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f42366a3a384e9695d0082e26506219\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/267556987f01452ebae06aadb2e21460 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9129c32a779242be8d6370816d0dd7de : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a7d5235fd59d4f0bbb440b30d9763ac3 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98b02ac6691a415d90dd50a98dad6305 : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eb22cd01ae4e425595e60ebe827a1a10 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca5e0c207f1b4c898498951937cf46ef : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/37cfd9910d374660a5cadc9ee09faa6f : SUCCESS in 4m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d1fbc771f6f14bae8ee867f49ba8e324 : SUCCESS in 5m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/674ab85a26d443589c9c648b88dc1805 : SUCCESS in 5m 26s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4d8d24abbdf149e48b1e11bf48cd1ae5 : SUCCESS in 3m 19s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/17b0e17067804da887effe5dda0acd7c : SUCCESS in 2m 40s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/76b1f079ea1348fe99cd7d9b71020f17 : SUCCESS in 5m 32s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/31c4b3e510154fd383c78c84338ea77b : SUCCESS in 5m 06s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/4854758484104796903dfc3efdccaf7a : SUCCESS in 10m 01s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/b8016fa295014453a014bcfaa4d14808 : SUCCESS in 4m 38s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/595839a97d4042959d190cfc298bce50 : SUCCESS in 2m 39s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/294e8b36f923452cad5b12329506b57d : SUCCESS in 3m 20s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a91c28773054429abb9e68fa4c188d4e : SUCCESS in 15m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cdabe3aea0f84b3996ccdff109b37662 : SUCCESS in 6m 20s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b234ba84400d462cb80588e68e4e5bda : SUCCESS in 6m 29s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0bfd3eb545254454bc2ea4830c44e804 : SUCCESS in 9m 04s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/28cf2adde9684b65830c23ead5445f44 : SUCCESS in 3m 33s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/ff0f048f2b1643b99ce968ccde92d4b4 : SUCCESS in 7m 53s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/f3f9be15c4be437994919a94bd929000 : SUCCESS in 2m 26s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/bda2e76994434f01a6d9e67337f406a8 : SUCCESS in 2m 38s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/32e9f450b94543e68dab94b49f1fbf23 : SUCCESS in 10m 21s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/3140ecb771754fb681bcd9f62ee1061f : SUCCESS in 24m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/063efda31ee74fe4938cb1fce33aa7a1 : SUCCESS in 7m 19s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/5f29263d77ea4883b6bd96c85f2ed36a : SUCCESS in 14m 50s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/6958aaac43df432a893de4d9b0928d36 : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e766f718f5c0443888323c2fc13eb2da : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a346abd5423465585d20d24ac051faf : SUCCESS in 23m 14s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/960e08f9a2fc4548aa5ff84699331d63 : SUCCESS in 5m 05s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f75e37247e744b0ca558c737b2041b4c : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7dbe6f86b6164511834caa74bf93f249 : SUCCESS in 6m 42s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/c28195fb69fd403d838d1768791ff75e : SUCCESS in 19m 31s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4922a45b72324afcb9f2c2abbbb13e9f : SUCCESS in 3m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a72bc8dd32c44787a2a38a34a6a5da52 : SUCCESS in 1h 01m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/590d96a7a30f4dbbb55de1610f0fc963 : SUCCESS in 1h 29m 45s","accounts_in_message":[],"_revision_number":1},{"id":"cdaee9264d285609d895d1afa6cb57d5f31327b8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 07:10:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"046b4508c09a3c85b4675e2ce11e79547b902147","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:11:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5729126856ac5384edc93e3d0f567457b20040b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 10:18:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb51d3c7452c40a69a2ab155a86ffba7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3c27a273d646d3a82e0a733b579b1b : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd777cb8837a4868b490305dd1129df6 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ae4c45b32254b4ead8ee684ed46ca31 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1819ce0db4444020a7f576e875336a31 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/28861bee1e9e4c9dbf4814c8da68d939 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48e31cc24f4b477b80ad55ff0bf0f606 : SUCCESS in 3m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a456be9aee724a95bb59ae0475ba7661 : SUCCESS in 4m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/31cf4f9577f948fab44e75564509e61c : SUCCESS in 5m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2b33cc5496d8483481c3119b79753e71 : SUCCESS in 5m 32s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/a740c49da245474999417dd97ee3a1e7 : SUCCESS in 9m 41s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/1e346dc268c347db888fd6f4f6fc1eb1 : SUCCESS in 4m 51s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/f853038ad37140b3aebfbada1e4ecd2b : SUCCESS in 6m 49s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/487a896f757542ddb2f4ba07b75a6d33 : SUCCESS in 6m 52s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/86198735b8854ba9be228b5f65a6ceff : SUCCESS in 16m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a4d807ff496486bbf9c96b01c778607 : SUCCESS in 4m 44s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/4ecc61ba47de431e86b6f6fcbe050cea : SUCCESS in 7m 00s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/df098c79623e4de9bd59a96b2fb18072 : SUCCESS in 6m 56s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e5a2a6156ef645299f5a02180202a659 : SUCCESS in 2m 41s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/1da81eda213f4af7b77319068bc5cd25 : SUCCESS in 9m 23s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c9e256a106214108bb5aa7f5a32acca2 : SUCCESS in 24m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8a0334be46e44f169d638f47cd48e56f : SUCCESS in 6m 41s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/4864303fa3d24842bab2129c3613d50b : SUCCESS in 13m 37s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/2ec3e6ffa6e9409ca66b3dd88a66884b : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/060b5666e80547a5ac6cf58cea03b5f0 : SUCCESS in 2m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5197e995b90c4ab488a616c3ef8b3d9c : SUCCESS in 24m 27s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/d8eb001baee24a6bac65625ddd590644 : SUCCESS in 5m 34s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/7477cb4c6c7444e09b061ca06d119351 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e33cb4300cd466e81c95b0599eb2e07 : SUCCESS in 7m 07s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/21a27b8630f84165b966e0d1366bb879 : SUCCESS in 18m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/904d708c3cd541c28e97278a4ead00a6 : SUCCESS in 6m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57335a3aa9434c23bbd00c255eec2bf0 : SUCCESS in 38m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3b60b563e6142ba8c1c799d17257a6f : SUCCESS in 1h 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"90300282615d774a2d6f42c57c5adddb7fe58c51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 10:18:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0a348cee7fccdb581a1311508e54c28e062399a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 10:22:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe837d64a5504f06bd51e1f921d73c21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45afd2bb771e4ce4a60f51cb4124fadf : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8557c203f2f77165240cbfbf90edddd488f8883","revisions":{"d8557c203f2f77165240cbfbf90edddd488f8883":{"kind":"REWORK","_number":1,"created":"2026-05-18 10:31:46.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/31/988931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/988931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/988931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/988931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/988931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/988931/1"}}},"commit":{"parents":[{"commit":"016f93e7ade7ccd65f23b34f720439a65fccb772","subject":"Merge \"update constraint for python-neutronclient to new release 12.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/016f93e7ade7ccd65f23b34f720439a65fccb772"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:31:41.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:31:41.000000000","tz":120},"subject":"update constraint for oslo.log to new release 8.2.0","message":"update constraint for oslo.log to new release 8.2.0\n\nmeta: version: 8.2.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I63c964639c6b3ace519f44d374587c99105ddb14\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ibbc4e9341da93f46434b84b043367b500668d416\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8557c203f2f77165240cbfbf90edddd488f8883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8557c203f2f77165240cbfbf90edddd488f8883"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
