)]}'
{"id":"openstack%2Frequirements~988933","triplet_id":"openstack%2Frequirements~master~I30ee45377aee4f19a36137b0d010946b69a5c18a","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-05-20 01:16:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I30ee45377aee4f19a36137b0d010946b69a5c18a","subject":"update constraint for oslo.limit to new release 2.11.0","status":"MERGED","created":"2026-05-18 10:41:36.000000000","updated":"2026-05-20 01:17:30.000000000","submitted":"2026-05-20 01:16:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988933-new-release","meta_rev_id":"a3fa4d36ccc5dd0f07b397017d0cb2302def3686","_number":988933,"virtual_id_number":988933,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-20 01:16:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-19 07:10:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-19 07:10:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 11:38:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 07:10:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"4042e076b32649ebd2bdf9c4f1ed85377804e674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-18 10:41:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9a9b91f5b2aba6cb7e18eba1a971e2631aa0f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 11:38:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3afc098ca3a3430dbe87944441d68412\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2f04f8125be4d34bd3f706058d9dce4 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4066dedc780640d8a20382c46f7516ac : SUCCESS in 2m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1abd5a7af6ca403d8e5e30f9eca3513a : SUCCESS in 2m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9d5ad9cb4214ca4968066be4d313ed2 : SUCCESS in 3m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/48431055ec414804b6754567d22b95fb : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db5a8cbce19f415fa78521a139bd74bb : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5ba56230d394ee3a03ce401996ab215 : SUCCESS in 4m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fe6493a9e0854d5cafdd238965f627ae : SUCCESS in 5m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dd12cfc4d44e4174abe570d07fa7ec56 : SUCCESS in 6m 40s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e43b8b894e36481a99b383b65307378f : SUCCESS in 5m 33s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/707792a04c3e4ecfb0d22ae04b3f0b1d : SUCCESS in 4m 03s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/3f3de7cddda146d6a50f3c005e63cc52 : SUCCESS in 3m 33s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/603032d220a247f597f2a0bb1d774895 : SUCCESS in 4m 44s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/145270e4dd31458bb0f65edeb439cae8 : SUCCESS in 10m 35s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c498cb4df4904638b9cab4ab58c8d4d0 : SUCCESS in 4m 05s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/f4c503238f5d4acda5ce6f1f29d44fef : SUCCESS in 2m 52s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/eef5bcead6734652a55ed49ff246b437 : SUCCESS in 7m 12s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/3ac38aa4a31c4f96afefdb3d6dbbc3b8 : SUCCESS in 15m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7ffd70871404de880f6eb7da803c976 : SUCCESS in 7m 23s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/8a5474e1a75f4a7a96cc3498e8cbb139 : SUCCESS in 4m 21s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/fb63a72ba4d04868a79803788b9d0f1a : SUCCESS in 10m 38s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/32b91c9d608741af8eb9ea089e7b0c28 : SUCCESS in 2m 15s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/9cfdcc8afa5f42569c3107242295a9de : SUCCESS in 7m 03s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/0814cd81d12a406d9b9796c2452c1d52 : SUCCESS in 3m 37s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/f38a9f8674904f2ba5e0df43966ab955 : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/84758309333f44ae8caeae20c37e6ccc : SUCCESS in 10m 51s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/702d5268ef184c6bbfebe975fbee9fbe : SUCCESS in 21m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3d85a9cb46964809b1dc63d1fd6443b0 : SUCCESS in 5m 54s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/68bfafd9b919451aa9ee2e973e92c631 : SUCCESS in 8m 05s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/419f285905b64936bba5ba44f98a9e86 : SUCCESS in 2m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a281d215f5ee41db9d1771f6c121d9bc : SUCCESS in 2m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a626587b3154e06ad78da3b411c7c92 : SUCCESS in 24m 49s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4e7da0e5972e4a1e8a8db7cd7cea5e0c : SUCCESS in 6m 28s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/c1ef97f9bdbd43029b46729b35ce8983 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c772d920395849399742fe17e50ead6f : SUCCESS in 6m 45s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/2f871a8ed65f4d6f9ce2d1ba937a4530 : SUCCESS in 19m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/99501680fe5b49caa6a465971c0b4028 : SUCCESS in 5m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae4aa2ca79cf48789b48d63a19b97c83 : SUCCESS in 39m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05d8c1c555ec4520b0dd1013c8701096 : SUCCESS in 55m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4df239da652b8e68f410d97730ce7e457fd644d5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 07:10:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c2bab07d55e39df0a850823247caf58222ee12b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:11:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd676b8be745c262117998942cfbe9d1870ecb58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 10:43:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02e65a561c2e4fed9b9b3c9725685e79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb23b6f13e544e3599e7c76de4e565f4 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b279c4c9234f0582ed0670d966aab6 : SUCCESS in 2m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89bdb7afae67471a9b4e1a3baf9c24c4 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dadba3b759204601b8608da32f02515e : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c992afecb4740d097b1b8d2cdde9535 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b8df006ac744ad8a2b7c4cba1337472 : SUCCESS in 3m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/127cf3347e794c96a4aec721f93c089b : SUCCESS in 7m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/71b95b1cd9714476a2b1a42b01160e95 : SUCCESS in 7m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5445711b942848fcbc5b1c425f0c17f1 : SUCCESS in 5m 43s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/5b63caf47b4243bf93874b408c549ce7 : TIMED_OUT in 41m 10s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/6499b7bcd4ad4548872485acadeb5e52 : SUCCESS in 5m 07s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/e6a628c52e924d2d82ee2d8b55481968 : SUCCESS in 6m 35s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/21a66dcc24e44c9bba7f9b646c249ed0 : SUCCESS in 6m 16s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/08f593ccb905491fb13b459b3c420f7e : SUCCESS in 14m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/21c02d61d5884e79b1b526b0630ed7f7 : SUCCESS in 6m 19s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/c6398fce415c4d54a895bda534270b39 : SUCCESS in 7m 42s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/52cb7eb3898145f2b2b0a58e86a6b7ef : SUCCESS in 11m 30s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/b4c6ab02865d4945852e973d59bae557 : SUCCESS in 2m 49s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/93325cb292004ec29e823d2c5d20c58f : SUCCESS in 8m 26s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/cc80f555717d4650b776fa8d5a41b03e : SUCCESS in 25m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5e3a20c582b44261a68c45124e83fa57 : SUCCESS in 6m 18s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a25b5719c04f4a89be1aa85805e31f40 : SUCCESS in 13m 25s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/162da07756af467f95e7b28f37da8e4b : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2232becd91cc4335a5302116745482b0 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e659fe32ad3473aaed9eda3c114cf95 : SUCCESS in 25m 44s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/7d19b6d08e8c47e780b2e68ce1079de6 : SUCCESS in 6m 28s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/de0f4438417e461d971c2bbf705e6455 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f0008d907c3b4e3f93eb55c21521e3e7 : SUCCESS in 6m 58s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/8e4816ab1d2f4c01a648b93d023b606e : SUCCESS in 20m 38s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ecb5ebc1fdd94c3d9789c48b559aede1 : SUCCESS in 5m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8467d45161d440fa84b0f6bea307b7cc : SUCCESS in 1h 04m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/049c71e628b84575a1d7563050a3e3d8 : SUCCESS in 2h 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3219e238ab0e9010d71db38dbb5640dbd1cafee5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 11:44:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c14981a699084d4a96f8f37c2f396ee637819c54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 12:49:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6404683eb1bd46ef84837b379a7441b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd8fc654f6a45da84a9d307e61d7f1f : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b3769f86ec45c780d61458e8ab9a65 : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9c1ac51251d47cb9170641ec84c789e : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71c34f20c13042b8b6e55aee7bba2abc : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/650bfdd251e448e9addde4c30564d110 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f21ff6c207c545cf8417a40ee2222fc6 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16d9c525287643cbad9f5bdc6b2c6834 : SUCCESS in 2m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e3b8122cfe0c439ca0fc1b823f676288 : SUCCESS in 5m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7cf90ddb004e430abff0441393be5d64 : SUCCESS in 5m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e924432a710347ee8204e19c2ffd75b0 : SUCCESS in 4m 23s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/2b405ddcd62845079b60e2dda8159983 : SUCCESS in 3m 58s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/50b0a006fbe74ea59bd35a9574657ebd : SUCCESS in 6m 01s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/929982093c7b4ffea124d8b62ffcde2a : SUCCESS in 5m 26s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/a67711a9352b499ca14b3a88041ae6a4 : SUCCESS in 6m 43s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/80531dcd622a41488f21adec4f39a07b : SUCCESS in 4m 44s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/483fc5fa3522415a8dd7b34643e95e5c : SUCCESS in 2m 36s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/1093c802587e4f90a38a783d95803e51 : SUCCESS in 6m 44s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b4c429d90b58407fa0d05a92f57f75d7 : SUCCESS in 13m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ecd4a45c446c49948dfc24d52d66a106 : SUCCESS in 6m 54s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/27fbf3c139da458e871533c1c6215931 : SUCCESS in 8m 07s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/152d0343c7bd44b2baad620ba16f20d7 : SUCCESS in 11m 14s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e32757d38aa54b1aa225816452a4c59e : SUCCESS in 2m 19s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/7198074bdea048cfb7fceb3933363c6d : SUCCESS in 9m 14s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/5e3cd85f4a7c416097349f15acc161a3 : SUCCESS in 3m 49s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/fc382ff8b5be490c9755dfd4c22d6942 : SUCCESS in 4m 30s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/f2da825f043e48f18436305ef176a13d : SUCCESS in 10m 21s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/18778195a5df4c958b314433a2595cfd : SUCCESS in 27m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e88900912b084393bdcc03ecf1960798 : SUCCESS in 5m 00s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/961aa4605a0a42ac887588a7d26147b1 : SUCCESS in 14m 01s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/6ac800baa8514418a3e1bb6ea0f37b8f : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c34ec1cca474444594cad7866310bd6c : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57088618e20b4e1e8df1622af58907ae : SUCCESS in 26m 37s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/f3f840248edc419e9e5e430e6bd310cc : SUCCESS in 4m 30s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/cdc0fc6c6a874155a4a825d62d4103b9 : SUCCESS in 2m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8639c638216149a8acccb55a4efb1f17 : SUCCESS in 6m 18s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/c404f80d21444e5baa9b42aa7af3368b : SUCCESS in 18m 33s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0d4d37fe9e4942d7a95136cd61384bc5 : SUCCESS in 4m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ba9bc369c5c4c1e80b4ce40f745b576 : SUCCESS in 47m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e206a56544b446aac6b63ed3225af28 : FAILURE in 1h 03m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7f8ee07949b62ccceaa7984c912b1e84b8532195","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 22:11:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4914cff51790d1b264573a1e154e7d4cd505109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 23:22:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/530ef6e1ac51429aae5d7ba9e0e676fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eae8bbf8b7648f5b83cb1b782ebea81 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e90699b882a24785982e25c7c056f064 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3f0f9f30e01478797fa095cdd9cf35b : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5120f326e94a40f9baa0d7a28180a59b : SUCCESS in 5m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ac094f82caa4f6eba38f4d3fdd3e144 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7135283e56094f5090ddb9412a21f9bd : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6601818aefd046f880a21a918e1c7c0a : SUCCESS in 2m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/92c75fbaaac248e6bd45867eee853087 : SUCCESS in 5m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4651deaf0e4d4baba9e839905fa79ca8 : SUCCESS in 5m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5c94872e47d0431ab9d0b76fb1274916 : SUCCESS in 5m 20s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/09a8e801efe546f384b6b4ac9c9ac89e : SUCCESS in 6m 14s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/e4ceb11a1d374dbcaba372e940379a13 : SUCCESS in 6m 03s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/c30eea71f97f449e9aa78501ab81ae79 : SUCCESS in 4m 55s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/ace28e18a5814b9f86c20557a77be32e : SUCCESS in 10m 24s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/6cca0a054667456ba8be99ccc4e9cac4 : SUCCESS in 6m 19s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/66808b065ce74054ab68ac2d50af6a1e : SUCCESS in 4m 44s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/187d0175f77e40738949209a704c284a : SUCCESS in 3m 44s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/7f4f82947abb4e3b84965a9eb8a6d928 : SUCCESS in 6m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8fc2822755f8425a91bf6a8966941e92 : SUCCESS in 6m 52s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/ce2c926ba157460aa0dfc9cee7d9a3d2 : SUCCESS in 6m 56s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/199c8a5e699b4874b1633195c5d1ce82 : SUCCESS in 7m 13s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/ae80acbf375e4b84950ff5d7767f5ab8 : SUCCESS in 2m 10s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/2300d6f355214ebaa7725282719bce46 : SUCCESS in 7m 19s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/6d129ea47c0d433894fb1c7974e62ed8 : SUCCESS in 5m 57s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/dbaa2d82a47140d5988b943568ab93ee : SUCCESS in 3m 18s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/0e83b31055c94e44a8e438ac3184de6d : SUCCESS in 10m 32s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/0bfd7c8daa56475e96d22daf7430d5e7 : SUCCESS in 26m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f21d5667b53a4b2a9cfb79c061d418cd : SUCCESS in 7m 35s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/993f58bd513243e7ba301b652bd5af7a : SUCCESS in 9m 25s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/6b0ba5235471429ab3a8d73b2daedbdb : SUCCESS in 2m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e021fe11f9844aacbf49838a6245007a : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/425f007a67fd4285bf7b92a4bf7d8da9 : SUCCESS in 21m 20s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/32207941a607496c934fe30c30bfb87f : SUCCESS in 2m 53s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/8c31d3dfe4f14ee5b5c8ce82b1e5e8da : SUCCESS in 3m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ef15a048573744ada560a08ef763b38f : SUCCESS in 5m 01s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/2ef72a16393b4bb28bca181ab5720b63 : SUCCESS in 12m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6e6691f2c9894116b9ecca5cfa79f228 : SUCCESS in 3m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b8346db26cf416c887dbc2e0bad0505 : SUCCESS in 1h 09m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90d4cc501ded466eb4c30c2f5a70645e : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4ad2c415b09ced771fdada653134d4772733d6f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 23:23:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae0b96a9c2f4077df9df207e4e734a113afdc635","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 01:16:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48ae21e889cb4e139fc0620f2a8719c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9246da423548c5bce3e6bad012060b : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccedfecf39d74ad4aa024694d0ebc4af : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a681c76b96314f9fad0f521041e2561c : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/514889701f2340f096ff201ee9867152 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34de5c54cdd54f3b8f66e071845a5569 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f075f72381a74ac1aa07314d5b311c73 : SUCCESS in 3m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e028875f8578468da5b983c7a6a2b836 : SUCCESS in 3m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9000711b73a64c73861cf0ddee34ac48 : SUCCESS in 6m 19s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/adfef0e1bc1d4256b7ad8cc11f4e0111 : SUCCESS in 5m 29s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/3cc38cf160ff41b8b613df78c22b42b2 : SUCCESS in 10m 19s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/11913a419c9f4c779e52a2840c20fe65 : SUCCESS in 4m 41s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/4b805ead720147fa8b34551d91dc587c : SUCCESS in 6m 25s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/080ba03632d54712b8b490fc6b80b512 : SUCCESS in 6m 47s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/e59f0fb4cddc47d081db1a146c6fa217 : SUCCESS in 7m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/16f902703cb34269a5fa4bb0a28fb954 : SUCCESS in 6m 59s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b90cf1bd4b7a42f48bbff69847159717 : SUCCESS in 9m 30s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/b8dedfb867154d96ac07f8701904a838 : SUCCESS in 11m 23s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/9afb9f943af24ce980ff63d3e2a18f5e : SUCCESS in 1m 48s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/2ad329f0651540318d9a2c6ce0801abe : SUCCESS in 8m 13s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/2be403f597ef4de78b704239f7ab2ab5 : SUCCESS in 28m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8cdf6222bbb5457a97ebc281ba9e77d8 : SUCCESS in 3m 33s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/9b6e1e9604294330a04ec6875ecf6d88 : SUCCESS in 14m 01s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/82a9f425a542484f8fbd560303f9f4eb : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a91db9427074127b1cdb938d279a4a8 : SUCCESS in 2m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f676dc91bb041f1b599a78c9d625825 : SUCCESS in 22m 48s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0a34aa761d3246fe953878928e2b5753 : SUCCESS in 4m 53s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/2198392fc6d94f0aaa35709f834c61fe : SUCCESS in 3m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/422800f451ac42b2abf5385c3b7a7214 : SUCCESS in 3m 22s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/0237d5521b2a4aa2bbbcedcdbc045d36 : SUCCESS in 8m 24s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2413b8d1c0524509b4dbfded700d3dc4 : SUCCESS in 5m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/916f581cc85b48e390f12f8f13ba56e0 : SUCCESS in 1h 06m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/529dab94491747e5a027ddfff4e73076 : SUCCESS in 1h 48m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c46c7a6cd88eb9357dd34729c60274348a6be151","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 01:16:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a3fa4d36ccc5dd0f07b397017d0cb2302def3686","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 01:17:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab4273f2e12420381ed2edbb8e7f95a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2155ffb55e5430c9e87c5d6339ab194 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63dca17c14c809775ca033df2e72c768b65eb7c3","revisions":{"63dca17c14c809775ca033df2e72c768b65eb7c3":{"kind":"REWORK","_number":1,"created":"2026-05-18 10:41:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/33/988933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/988933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/988933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/988933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/988933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/988933/1"}}},"commit":{"parents":[{"commit":"016f93e7ade7ccd65f23b34f720439a65fccb772","subject":"Merge \"update constraint for python-neutronclient to new release 12.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/016f93e7ade7ccd65f23b34f720439a65fccb772"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:41:31.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:41:31.000000000","tz":120},"subject":"update constraint for oslo.limit to new release 2.11.0","message":"update constraint for oslo.limit to new release 2.11.0\n\nmeta: version: 2.11.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I49df64e42768c1ac53f02d8dac66b7387587efe4\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I30ee45377aee4f19a36137b0d010946b69a5c18a\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63dca17c14c809775ca033df2e72c768b65eb7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63dca17c14c809775ca033df2e72c768b65eb7c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
