)]}'
{"id":"openstack%2Frequirements~988934","triplet_id":"openstack%2Frequirements~master~I29a0566320b7012530b70184789dd54d7145b823","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-05-20 09:25:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I29a0566320b7012530b70184789dd54d7145b823","subject":"update constraint for oslo.i18n to new release 6.8.0","status":"MERGED","created":"2026-05-18 10:42:53.000000000","updated":"2026-05-20 09:26:39.000000000","submitted":"2026-05-20 09:25:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988934-new-release","meta_rev_id":"7bae9f4c4077a3f338bebcc04713920a270088fd","_number":988934,"virtual_id_number":988934,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-20 09:25:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-19 07:09:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-20 07:43:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 12:31:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 07:09:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-19 16:59:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"deaeca488f1ffb1b4233eb1a2384875e149e5424","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-18 10:42:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e6cf68fbcfb62ed26e44033709e12207f71521f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 12:31:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25a7cd8e93164b0cac3444fc249cb669\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5d36f66ee064f1b8c332801c03ac6bf : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/309515a9abb344c6a2fe71aecae3ca4e : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f0462455d064a9ab9cae58366b8f868 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5421a0a12d541bf8c8332a366712b6f : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d4509cd755da44bd936131fab0b0675c : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4ecbf3e873040fe8772b1a3742e4c19 : SUCCESS in 2m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3853edb18d21401c87a3b0e794dd4e41 : SUCCESS in 2m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7d4218a6206c4bb098c35ddb0f4c131d : SUCCESS in 5m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d4eb41db90ae4d49ae8821c52bf51204 : SUCCESS in 6m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7c2cc26244b341a0ade260b54f78faa9 : SUCCESS in 5m 20s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/93899d22923f4a82844579d9442937de : SUCCESS in 5m 23s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/5c1fe7b5f06746c08a6fe6d73f80ef1a : SUCCESS in 5m 13s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/b4251856cdf54e6f8f9a8f1ba889fa8c : SUCCESS in 4m 46s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/449e5a71b39f42afbc4404ec22478453 : SUCCESS in 11m 45s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/db97da57b14643029f9b7b27a8e7b798 : SUCCESS in 3m 09s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d7013707deb54475ab65560bc05602bb : SUCCESS in 3m 27s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ab7459ccfc61442dab075a6eecd45b3a : SUCCESS in 6m 11s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/aaeef5573dab4de9aa2cdb2ae65aefba : SUCCESS in 15m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bed60cf37a4c4e81ac628ed3edaed63a : SUCCESS in 6m 28s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/3327678410c349af842afb42084729b9 : SUCCESS in 7m 09s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/4bc58ccfe06342e28de1fcb9ab9f8785 : SUCCESS in 11m 36s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/9eb8217816f1492390d8dd8897e21a53 : SUCCESS in 2m 33s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/acbfba0921e841af9f49cc115b0d06a0 : SUCCESS in 9m 13s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/b492f1f7dca84b4aaae5d507f482d3ee : SUCCESS in 4m 47s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/4121c9c2cc9e4339915505a8cce09889 : SUCCESS in 6m 03s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/32bc820d9b5b411aa84cab7b3aaa6a66 : SUCCESS in 10m 17s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/bfd3240cd2e3419d9f3426cf9390b881 : SUCCESS in 20m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cb0939caa0e54ad189f6ddf322886189 : SUCCESS in 8m 31s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/10b1c542145c4801af12e5b0db5deced : SUCCESS in 15m 14s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/dba7a95b787e4861aba0f719a95f5069 : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6ec69db8e4c34a7991297b03aea4fa02 : SUCCESS in 2m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a3fcb3ecdfa4dbd964d26c5f5f5b508 : SUCCESS in 29m 17s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/5fb09d47242248deb81615e48d569cba : SUCCESS in 5m 40s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/6eac01a8505645d793ac83ffc81003ad : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/06772a098eed4045b62863f6f89b3518 : SUCCESS in 6m 40s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/fa41587c045a4ce78a2d173e502b3548 : SUCCESS in 13m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/efe63af553b54dafa25bda2e99777e60 : SUCCESS in 2m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5a4d3207f98463e8a5ae66c8fcb4eb6 : SUCCESS in 1h 09m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87057417292440839aeaede016c26e03 : SUCCESS in 1h 45m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0685823e9f0b0fbf632a42542263dc0d834e6464","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 07:09:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3331750607243eac64cae57aee77b6d94bee56c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:10:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1df8e6bfbefc2d73eca18557717d0e8a85da11c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:32:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccf584c8989d425a93e6f6e8ebbfe27d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b7d3854093a475598d672442010c874 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbffdb7c82d44df7ba57a8bede598088 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6809a2a1ba84225ab160dd9023e7eff : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2a669bba11f42808a149716dab794a5 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29a9e53a703c4c0e976a4f0da1b4e387 : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4d3dff8a0d54a4c803cc4f046a35f78 : SUCCESS in 4m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/98a51febfe7c41e183fe4436ed8f1655 : SUCCESS in 6m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fa572da477774fc8be6a15b68778aa09 : SUCCESS in 3m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4a88acdf885f4e238c714c1bb21d657a : SUCCESS in 6m 20s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/be077e86ab694fb592cf78eadb430b6d : TIMED_OUT in 42m 00s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/729ae64bab4f4966820fba5960640104 : SUCCESS in 5m 42s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/221d777ed9124e7cb994dd3af0972d52 : SUCCESS in 7m 25s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/bc53ef2c720b445e86373f9d48a4b52a : SUCCESS in 7m 06s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/de4904f52818491ea0987827a822f7a7 : SUCCESS in 7m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7826dad137534d7d888d2f4a5f9a93ef : SUCCESS in 7m 58s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/1dcf82eac8e74443886e5eb37ca6c3be : SUCCESS in 4m 56s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/10abf990067948a6b6ed1c610008aa67 : SUCCESS in 11m 44s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/89efb1232e444194ac5dc472a9428c5a : SUCCESS in 3m 01s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/00a06c3f824c4136934b66ad06da9990 : SUCCESS in 8m 25s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/aa87dd2c56424229ad4e34254677387b : SUCCESS in 27m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b0eb26080447409ea58dd7b365d0e64a : SUCCESS in 7m 32s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/243e241e1d94412a921cb88cd2a5032d : SUCCESS in 13m 20s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c38b1497bd524fc6b7fa5b40daaba032 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/35def6ca5513450b8c1db81cc4f3b1b0 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/026eb6d70acf4d8c9ab4a95874a69c47 : SUCCESS in 26m 01s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/342ef1ed81104a4ca5eb4159188839ec : SUCCESS in 3m 39s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/42bef2510e7444f8a4a77a77fb48a506 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d7f28ce0add43bc8eb785cd216c5836 : SUCCESS in 7m 17s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/4e383b6095be4835bf5cec7e19df054a : SUCCESS in 9m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fb0c4415ccc94c16b121ecd2133d1afd : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f51d3cafbe9f4e438773921d3cae46cd : SUCCESS in 38m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25f3666427f649a28a522a8402d8e2ea : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"89eb2332a512217de4c02e649affa6b9ca98c3c1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 11:47:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd0925dc74ec1526dc6746ac6911a2a92805ecb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 12:54:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b40cce0e1a3433996842fa61ceeca85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a9b0277414c4b16a44e0b81b57ea61d : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/810f304c034041da8cc3bfc141d5a0a1 : SUCCESS in 1m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/339e8232b3aa4e3fb78ce49e438762d2 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5990c9417afb4ed0a2e2e24225298213 : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/43d9130bd4914fc4a529dea613dc6918 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74f4183bce1c48c2bbd68f293b5bdaaa : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16046819061f4c2c89dfc62a4aef8fd2 : SUCCESS in 3m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e52272f63bb2489b9c8c736afa6b4d8f : SUCCESS in 5m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c24f922e0ded44f7a0650a9ea27092a0 : SUCCESS in 4m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/70ab112648b741c2b3522aed1b749e58 : SUCCESS in 5m 14s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/00728e220e0746d7bf41b00eaa1662c7 : SUCCESS in 4m 26s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/f31e974e78bf4456b13957bf6a9aa39e : SUCCESS in 5m 25s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/598fe98b6bd64acc9359bf6087d12ac2 : SUCCESS in 5m 11s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/18eac54a47eb4937a23d001aa0321cdd : TIMED_OUT in 40m 58s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/ae88976846074f22a011f98057f53a58 : SUCCESS in 4m 59s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/53540def0ad4423d829cf1c678500135 : SUCCESS in 5m 04s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/6ff152ac8e084325be8d6471e3ab45c3 : SUCCESS in 3m 33s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b721dfcc5a0a46d7835ecca7914bf9b0 : SUCCESS in 15m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e738548e3c540168a72c8c1af417845 : SUCCESS in 7m 02s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/2164d1593899422a88a223f48b7171f2 : SUCCESS in 7m 59s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/cc3c6d0aa55b456da324a070bb9d543d : SUCCESS in 11m 10s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e0ed992be1bf43f5988915de81a3e3a3 : SUCCESS in 2m 35s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d5b9556191fb4936a42c890c04fcf061 : SUCCESS in 8m 09s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/7855651c4de5482b8af9c0f78c8bb073 : SUCCESS in 4m 25s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/4e550aa67dcf4902963632491db60c87 : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/fd284d6ee167406f8ffba67f147fff80 : SUCCESS in 10m 53s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/338bfdd333204240a82001aa4fc2c156 : SUCCESS in 25m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/96982046ced04bceb8ed29089da33ab7 : SUCCESS in 4m 57s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/cc6262c8878145f4b9c1918919ef3749 : SUCCESS in 14m 14s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c4179e68e71f441496f1706206acc963 : SUCCESS in 4m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/41e341776e58457ab1ea1debbf33915a : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3293cccc9794ddb9d1acacbbf5c3a22 : SUCCESS in 20m 53s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/2da90850bfa14ebc9cd79e56c32f8e55 : SUCCESS in 5m 31s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/47e9e20e604543cc84afdd41803adea0 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a8d0a609728c413eba2223d93fc21938 : SUCCESS in 6m 57s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/f3106e7630184b80936d932e45199a27 : SUCCESS in 10m 46s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8a6278142b504fc9b92c4b3dc76584a0 : SUCCESS in 6m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b88c97045ce3415caa5d3583264f5450 : SUCCESS in 1h 01m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91b3686f4647455098cc6c71dfcfa640 : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1b6e46a15ba96ac2d0fabb4c3d68c5304b0e49b7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-19 16:59:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f9220d4a0fa0d62e0415dfce09ce9fd008012bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 18:55:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07720b20af9e4209bf35ad59f8774161\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5cab2c2fcc1499db94ccefc62a937a3 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c03088e2dea14c0bb3a9b0686d0433f7 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d172891b2c2c44cc979e5dbaf12e799a : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb587ac740d3421eaf34fb0bfadcde22 : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/497fced2a9414a9fb5510b4f1e427866 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbdce7b5c88a46889f749545b1145544 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9f4ceea77384e89bef2015d43da9e24 : SUCCESS in 3m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/372880c658224755b81e631efb074cab : SUCCESS in 5m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dc49053c775f42c8bc241aedf1cc6935 : SUCCESS in 8m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a77b2d07a9564405aacee9ea0db80e5b : SUCCESS in 7m 26s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/280e5d0218ef48699ccaccffbe2e1880 : SUCCESS in 4m 29s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/6b4fdce8685a435c951c8477cbc0c730 : SUCCESS in 4m 51s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/7c8b1f79e0c24886b4e128c9d08fcc1f : SUCCESS in 5m 51s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/758f3ab940894cfc9ece1f2fa1c6cf83 : SUCCESS in 10m 13s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/00fd92cfbc8d45e6954c09683adf9621 : SUCCESS in 5m 37s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d7bd67058acb4dc5b3741a52edd071f8 : SUCCESS in 5m 23s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/fcf5596fec3045628474f7723a7a0c45 : SUCCESS in 6m 19s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/510528b90a5744c5b7693c9e9c16e1a9 : SUCCESS in 15m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d0fea00589b447fe89fb216a336cf25f : SUCCESS in 7m 24s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/57dffa03dc19451c9de3f300b884590f : SUCCESS in 8m 52s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/8ee42025abde4402a0cc6283e9a37ffb : SUCCESS in 13m 09s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5b25ac10fcae4c74b44d4f967fa77714 : SUCCESS in 2m 32s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/53f3ea26a6ad474b9d81a5f05b7b4350 : SUCCESS in 10m 31s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/f0efe623bddf48868606c0fbd80acbed : SUCCESS in 6m 43s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/7e700a36fff646278537f85a9961f777 : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/944acdd0ffa643e59c2db09712e2c852 : SUCCESS in 11m 29s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/abfc883f2bd84709a9f5aa2d16c716a1 : SUCCESS in 25m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4e948d16cfc3461aaaba586d2bc1f615 : SUCCESS in 4m 03s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/c8079839db654b868ff7cc5e89c0554f : SUCCESS in 13m 53s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/61e079766cf64c6085f80a202520932b : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8ad3f65f774f4b13bd4c1622eaf164df : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04d3f3a14f9d4493af83c6b2bfe117e7 : SUCCESS in 20m 29s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/7509a638002c49ae91a777c5b626bc95 : SUCCESS in 6m 28s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/28f627d147aa4e07bea2490a88c40126 : SUCCESS in 3m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/584270ccb681484188f59bbddc0a2563 : SUCCESS in 3m 37s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/9430eaa2207c4fcab20d4eb3b5c27edf : SUCCESS in 19m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/80bd9318942949a9abd6b882b15f79b5 : SUCCESS in 5m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6dc70cdb7abe4a04bb93867b7a1c3870 : SUCCESS in 1h 09m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b67e9f42df042bd90de7daca203bcf9 : SUCCESS in 1h 47m 12s","accounts_in_message":[],"_revision_number":1},{"id":"301f7d64e5a03b0449c377ce65b4e302bebc80b5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-20 07:43:26.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"d71fea059cf4eb9704669f63e8cfb1a299b4bdce","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-20 07:43:30.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"712c11e63e805ece65b52fa711e159992564b13b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 07:44:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7299f0dcf83de08b3a68c19bdc9eacbd4f0f5396","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:25:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a31a6126935f4939a98c973ddea94bc0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa113e30eae4f4597aa4a0c238d87c7 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db21849c41604a4795803991b3eb5523 : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb9e42425cc343db834f7219207d438a : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f59b7a74ce1d4e9dbfbf3bbd1461d46b : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/895265f963a84b8d9e7a97e1fe14817b : SUCCESS in 2m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/caff7fee75914faf941fcb616e951c42 : SUCCESS in 2m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9d7321c27cfd42df8d6a09491997d527 : SUCCESS in 3m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/846b7ab65375434fb2972b42a500e7c5 : SUCCESS in 5m 42s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/bc1fdfbc29ba44ceb828e25a4b45bc08 : SUCCESS in 5m 22s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/b2818bc81ee149dc8f31b3bcc35d6f9c : SUCCESS in 7m 30s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c605cff268b04da681269eb679ce88fc : SUCCESS in 3m 45s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/8a59db6b56534329a2f32bf1fc6e5afc : SUCCESS in 2m 58s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/439e0d2f8627422e86d596261964bef8 : SUCCESS in 6m 22s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a743bcefcf9243138a647c8ca795a255 : SUCCESS in 16m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95cfd0ed8cc746a191d55465666145c4 : SUCCESS in 8m 06s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/caff06d59d9d4ec491a0009bd619e29b : SUCCESS in 7m 34s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/52bdbd86aa38417bb1fc6b3869777134 : SUCCESS in 6m 28s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/514566a4f19641c88d315a6a7a41637b : SUCCESS in 2m 54s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/9855a264b39749bfac2f0043fd4aaf25 : SUCCESS in 6m 58s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/5b800df5c616423ebc49fa4ea63f3a70 : SUCCESS in 29m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/51dfa1a6f2e0400ba6d0ef2e791004b5 : SUCCESS in 7m 05s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/fac1d5e235c3442b8ed70817cd875892 : SUCCESS in 9m 36s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/8e096c091e514fd88941f96cdaab34c4 : SUCCESS in 3m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79c89c908386483780b2dcf843a8ca1c : SUCCESS in 2m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26b4e216d0ee4efc904a898ee59865dc : SUCCESS in 25m 47s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/26e7b1d967a64c9b8cbe1933b772362c : SUCCESS in 5m 41s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5da0fb22709d4eaba209db50b89e177a : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/178a1d0788ff4fc29cefa9944e4438b3 : SUCCESS in 6m 01s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/9692a37a241d43c7a4a57ee3f638166e : SUCCESS in 18m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1202f0bf2b5b47a4bb257519b8d8760d : SUCCESS in 6m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/629e3e7bca304502895cc0dcdb936773 : SUCCESS in 1h 05m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cdae688ff99475199f9de392b434866 : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":1},{"id":"848dd00ca07b9d2cbd2ce392c17f8b51a6be8e87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:25:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7bae9f4c4077a3f338bebcc04713920a270088fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:26:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc6caac80d42420ebc899d1975ed848e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57007d17d4df4fb4a4da2ff681a6d0b7 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03d414d9da22be32825a4d9617c52d31f46aaa0f","revisions":{"03d414d9da22be32825a4d9617c52d31f46aaa0f":{"kind":"REWORK","_number":1,"created":"2026-05-18 10:42:53.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/34/988934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/988934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/988934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/988934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/988934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/988934/1"}}},"commit":{"parents":[{"commit":"016f93e7ade7ccd65f23b34f720439a65fccb772","subject":"Merge \"update constraint for python-neutronclient to new release 12.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/016f93e7ade7ccd65f23b34f720439a65fccb772"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:42:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:42:48.000000000","tz":120},"subject":"update constraint for oslo.i18n to new release 6.8.0","message":"update constraint for oslo.i18n to new release 6.8.0\n\nmeta: version: 6.8.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I948d85a2818414f7972a5f9f0948e4e95ddac193\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I29a0566320b7012530b70184789dd54d7145b823\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03d414d9da22be32825a4d9617c52d31f46aaa0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03d414d9da22be32825a4d9617c52d31f46aaa0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
