)]}'
{"id":"openstack%2Frequirements~988940","triplet_id":"openstack%2Frequirements~master~I7c469b795e9a16f6aea1f48cc044f46d5339cc1d","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-05-19 10:18:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c469b795e9a16f6aea1f48cc044f46d5339cc1d","subject":"update constraint for oslo.context to new release 6.4.0","status":"MERGED","created":"2026-05-18 10:47:58.000000000","updated":"2026-05-19 10:21:14.000000000","submitted":"2026-05-19 10:18:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988940-new-release","meta_rev_id":"3651f8f76f8fb71849b3c54603ca4dd6e132b642","_number":988940,"virtual_id_number":988940,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 10:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-19 07:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-19 07:09:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 12:16:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 07:09:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"e78f0bfb117c996989bcb76ab4b2e9a73a3f81c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-18 10:47:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"656c5b2db9cf97284a5f36aba9d656a2adaeb677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 12:16:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddb6c46ed1654d4e9606dea4d84c2951\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658b98189c2c4a029da66e4e88eb8d99 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa932782ac644021ad06045aeb9ddc55 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10beb82431f14e9e8f5148b4b8a4fe6f : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd3988462034439eaf9ac90211fb92b8 : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/85e226fd75b34ad88e02776807312614 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dabc9c1b27ba4321883d053f5059e4c5 : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86dbbacef45945369e6b7287aa5e2ffd : SUCCESS in 2m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/84aa31d3ce4a4e14b26782930fca5ca3 : SUCCESS in 3m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3c27d12a993f43b5a7ccdc7cf713a4c4 : SUCCESS in 11m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e6e464d186bd44189b95ae4ad95ce5af : SUCCESS in 5m 55s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/97c81ff1b77543d09273bd7a29707dd3 : SUCCESS in 4m 03s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/d9771f0a733640139661a817f9782d74 : SUCCESS in 5m 06s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/f87657b4b7ac40e1b98ef480dbea35db : SUCCESS in 5m 25s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/2352ec1af0a546b09933265976121103 : SUCCESS in 11m 45s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/30e92d765f404ebf88cf466473fd7bb6 : SUCCESS in 5m 33s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/3e39b3b1c16a4ddb8683d6cba4c9611a : SUCCESS in 5m 37s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/5675563b05f6486eaca0afff097a9eab : SUCCESS in 6m 38s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b4f4fad58b384059894e531a1ea1de8b : SUCCESS in 15m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa7b69bb102b4677a9e43c21b2fda7de : SUCCESS in 8m 56s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/ad5699c39c88450199bb93b24e1cdbe0 : SUCCESS in 9m 25s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/11cfe2e631704ff9855816cc02af34e3 : SUCCESS in 11m 52s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/533fe5efbd124a9390c446c3e61e547b : SUCCESS in 2m 41s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/7c571f6bc4b640759de74900c2c7e7d5 : SUCCESS in 8m 45s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/8d491b5603d343dea30a833d1475ca76 : SUCCESS in 2m 32s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/147fef8efcc6463ea0d93ad901fb59d0 : SUCCESS in 5m 40s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/147b69df1b2d4d6cab2f037e92e32353 : SUCCESS in 9m 27s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/209d725e735d455e998fb3082f7a50fd : SUCCESS in 20m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/60d7250c28e84c7d9d475f22221d5a1f : SUCCESS in 6m 14s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/ae5469fd3d2d408e968871f5e724f49b : SUCCESS in 14m 28s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/114b3650003f423bb55c338dbb1e170a : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a78b9ce05d6c4d26bcd32b87d89f1ac3 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20501bd489ba46b9ae2a89c225985745 : SUCCESS in 26m 26s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/ed02eca2dd6347ef9b060dbbae5d0354 : SUCCESS in 4m 06s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/ee9e8c1d4bd5485689f9553c7e0d69ce : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c987242f41834e35af0ea4ae32669775 : SUCCESS in 7m 50s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/52eb8aa21ac84087a9020c1561ca6a80 : SUCCESS in 14m 06s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fdac53813b4745b7a70e679979516029 : SUCCESS in 6m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c903bb58fdee4a1e9daae91cff6ac863 : SUCCESS in 40m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1afb266753354395afc739d0a928a776 : SUCCESS in 1h 22m 35s","accounts_in_message":[],"_revision_number":1},{"id":"cd5ad7cc9218d130f9373d85f3b681aa48f55138","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 07:09:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"313baa0f8c4f3360b7a84fad2bc0411abe3bff11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:10:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c8aac27e9bf6f9e31cfb34a4592a48acec53dd86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 10:18:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eefd288e0314ba7970de46cf12b8ad7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5413a8d1a9154ae58825e0a239e1dacb : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dae918d68e148c8b8469316a1e5a582 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6088a83074184058bebc010ff4e898df : SUCCESS in 2m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28d375493cb643f9941f8dfa3c818204 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d3017f639e043fe9d08caa06fe1f4ff : SUCCESS in 2m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/355f5bacd6234c9b9919a1a0e4077caf : SUCCESS in 3m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0b40c12f798c4237af0d5a25d2620ce5 : SUCCESS in 6m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/43b70d5bcd9044b19292cf1916dbf124 : SUCCESS in 5m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9fadb182dffa426996528c307a6738ef : SUCCESS in 3m 11s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/b42c527dcbc4436cb5c61ebef09454d5 : SUCCESS in 8m 42s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/0e2f698d9a8e45a9825db30f3624ed0a : SUCCESS in 5m 17s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/11be7e337eab4b7298028d5995420135 : SUCCESS in 3m 04s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/bd8693724e814facb9f7fed68af0c260 : SUCCESS in 4m 33s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/dfd19b56c8544041ac6f17e29646ad48 : SUCCESS in 7m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3221e1fcf75418d9f18d478fc892d29 : SUCCESS in 8m 06s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/c2e882fa6dd744cc88f1094a2f8a8445 : SUCCESS in 4m 13s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/13dbce84089c43d49401c849330f73e1 : SUCCESS in 11m 36s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/f44f8f1b4b124f9ebdbf8cbe598bf6f2 : SUCCESS in 2m 33s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/8b83242871674e92b599296d117cbf54 : SUCCESS in 9m 03s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/6e7a9f8a8eac482a9dbe1af1c3f6e031 : SUCCESS in 21m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/03c78e6826e34415b1fcc3a41ef07bc0 : SUCCESS in 6m 54s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/dcf276f3b10744cda668dd0379f9fdcf : SUCCESS in 11m 54s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/87e7b1151023495e8ccc45678ed3b382 : SUCCESS in 3m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0efb923b219b4383960d50fadf58cdd1 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/28f30ffa6ead439990c7f2b3816ce302 : SUCCESS in 21m 56s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/8573220847d94531b595803921851f9a : SUCCESS in 6m 49s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5a12357042d24c99a0319c82d958cd22 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d426aea94d504c658e18b9768200ba3b : SUCCESS in 6m 45s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/5da0bccf89d24cbd9b6e8d17b01605b5 : SUCCESS in 18m 55s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dc21634203ba49cb82fbd2bb99f4281c : SUCCESS in 6m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4d8f9617ba343169766645708516a96 : SUCCESS in 1h 07m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34b88bda1314479db8e2a9238118a344 : SUCCESS in 1h 45m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8a9cc7bae7b215c2535c2c0268a0c2fdc3d7584f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 10:18:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3651f8f76f8fb71849b3c54603ca4dd6e132b642","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 10:21:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92c0fb3d028947cb8477d9f48e7d06bd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d26be3ca5e14243a227fc1d5e6eec09 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57f81a7334f6e71e21a7f8c4a5ab63b84ea61e58","revisions":{"57f81a7334f6e71e21a7f8c4a5ab63b84ea61e58":{"kind":"REWORK","_number":1,"created":"2026-05-18 10:47:58.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/40/988940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/988940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/988940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/988940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/988940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/988940/1"}}},"commit":{"parents":[{"commit":"016f93e7ade7ccd65f23b34f720439a65fccb772","subject":"Merge \"update constraint for python-neutronclient to new release 12.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/016f93e7ade7ccd65f23b34f720439a65fccb772"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:47:53.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:47:53.000000000","tz":120},"subject":"update constraint for oslo.context to new release 6.4.0","message":"update constraint for oslo.context to new release 6.4.0\n\nmeta: version: 6.4.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I4838f02eb35b77ad1457c349525cf81e5d9fad52\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7c469b795e9a16f6aea1f48cc044f46d5339cc1d\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57f81a7334f6e71e21a7f8c4a5ab63b84ea61e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57f81a7334f6e71e21a7f8c4a5ab63b84ea61e58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
