)]}'
{"id":"openstack%2Frequirements~988941","triplet_id":"openstack%2Frequirements~master~I7063949198001c02e13bce267b8a183544495080","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-05-19 12:22:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7063949198001c02e13bce267b8a183544495080","subject":"update constraint for oslo.middleware to new release 8.1.0","status":"MERGED","created":"2026-05-18 10:49:17.000000000","updated":"2026-05-19 12:23:50.000000000","submitted":"2026-05-19 12:22:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988941-new-release","meta_rev_id":"a7c220a135e72b9100f05dfdd518f068b73dcc10","_number":988941,"virtual_id_number":988941,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 12:22:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-19 07:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-19 07:12:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 12:47:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 07:12:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"8e54b41aabc3ca4608cda82519db63750ac38b11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-18 10:49:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96c55bb1e761b0ac3babb89367bce5e717ca9609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 12:47:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8df83fe2bf342afbf748f39c372046a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40296bedcbbb43e887064a7ff3ddceb9 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d3e49ba8824b98b884fa4d0ac98118 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6dcf0ce79e84450aae09ffbff0be339a : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90dc7c1a61154f6ab7b95bcd88f9ce28 : SUCCESS in 2m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5b93833823ef48cea32ece7f094274a5 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55c4ad92b74d4f639ac365ac1c62a585 : SUCCESS in 2m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d5e5b27466244599ada99a38bb82e0d5 : SUCCESS in 2m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d61d4c823ce745d196ae79d3c9601259 : SUCCESS in 3m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3bf25585e7144aa3ae69b599b3065cb5 : SUCCESS in 7m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/941c706801e1421388a851aebc2ddf5a : SUCCESS in 6m 28s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/5a182f21b0924a97ab3d9e5b5afd0de3 : SUCCESS in 5m 05s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/20aa6b7402b74601b9d5edd4c718c06d : SUCCESS in 4m 38s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/c5d9d51b4d824a76a4632918e79db46d : SUCCESS in 5m 13s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d8aa4e94237d4bf3a94a0bb388428b8b : SUCCESS in 12m 13s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/050f8036f21342c18732892642c34f43 : SUCCESS in 4m 39s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/18c5d3af4495428bb8dc15fcf41e5fb5 : SUCCESS in 5m 27s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/7344646200d84e2c976b05fbfaf7a8d2 : SUCCESS in 6m 02s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/6f386c05218b4f8aa55d10ab9149e36d : SUCCESS in 7m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a014a7d7a2b44caa1e47d42f875efea : SUCCESS in 8m 21s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/be54a0ded9284311bdb06703abca0103 : SUCCESS in 8m 19s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e74671a8d3a440b99b400ee0044dacca : SUCCESS in 12m 49s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/dbc0674ea31e4606b30a922cf70d70e7 : SUCCESS in 3m 38s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/93b81dc487024d288b7ed5d0b254439e : SUCCESS in 7m 32s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/e140a89ec70642bfad4eda61842820c6 : SUCCESS in 2m 37s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/350f1bf75d6d480fac6d01eae6eda85a : SUCCESS in 3m 05s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/65ac183322b2406baf336ef64a156fdd : SUCCESS in 10m 31s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/709db9c698b4484d9b5d7763b83f6531 : SUCCESS in 20m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e004dee5a1b34c0b8aa19cf6e20e175b : SUCCESS in 5m 31s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/bb31fa38ed6e4f1e903e828a0b2e1fb2 : SUCCESS in 16m 40s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/f0e661036ccf4d7384e3badac70e5cd1 : SUCCESS in 6m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9faaf83afbdc4970852845e6c279328c : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c8d5a596bce462a8b35bf5b7d1fb9be : SUCCESS in 27m 05s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/ac504cbe4874474d8464d2ac7ec9bcd3 : SUCCESS in 7m 48s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/aec34a0318a741c88bc3964f2c959987 : SUCCESS in 4m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05abec2ad3cc489fa06f2cbd6617d48d : SUCCESS in 3m 32s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/17ca3768d686465eb9a0ae2b7b9e3316 : SUCCESS in 9m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/08895af25ef74af7bec95b4a7d9c55da : SUCCESS in 5m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a86f1fac2e1d4acb9c8d1bb102cbb651 : SUCCESS in 1h 07m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04edb85a9ba941f28b86195c08560679 : SUCCESS in 1h 50m 33s","accounts_in_message":[],"_revision_number":1},{"id":"69197b8628700c0ab430d09ea02d53481fe866c6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 07:12:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e353eece502bc176b9113a12b85b5c8757878c46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:14:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8674bb64d07a47c76d90440d10d962359e966a60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 12:22:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27098cacf2d8463cbc07815c106cb10b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb21b72a3e764f5e95cf2105fe400eb0 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7349ea8d4902465484871e839650a114 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07c929794efe4559ada561cb8e3bc75e : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbf5ab667a1b49d6a5189c9e045f3b56 : SUCCESS in 2m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ec2bfb6f17a4f7992818f2e49955aa6 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1f5f19ebf5c4486e8064bed437ed48bb : SUCCESS in 3m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d65f37d7fea5432282cf14bc199c32b9 : SUCCESS in 6m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/08cc0e4eb68f4de7bcce86cbea92232a : SUCCESS in 6m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a4335132a83e457aac5525bb02551910 : SUCCESS in 7m 53s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/be81eb42b08f49e7aa208077b3d5a7a6 : SUCCESS in 11m 06s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/34c1fafd53bc46b29ef3860d63c633c8 : SUCCESS in 5m 27s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/4c08d3f7800d4122a61d54613df700f0 : SUCCESS in 3m 20s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/95b200fdee1f465d96f0270041259ef2 : SUCCESS in 7m 10s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/709a4b71a0004bf6a6d85498be81bdad : SUCCESS in 15m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a627df48662b4f5088eb5be38eb89640 : SUCCESS in 7m 55s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/552e5350ff19487d8b678139038e8830 : SUCCESS in 8m 33s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/dfd511157c4049ed87be090ed7573784 : SUCCESS in 10m 42s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/42dd62b59ff84e40b265440daae549f3 : SUCCESS in 2m 14s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/84d852076c64446ca2c34b240cc0b742 : SUCCESS in 7m 59s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/be09669845fd4cfaa7319b0827d89f75 : SUCCESS in 28m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/eb4647508c6d43dba0fb57790a7b39f0 : SUCCESS in 6m 54s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/cf2d62c734df45cf84cf644e6c0eec0d : SUCCESS in 10m 17s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/44ef59dabd6341158a908cad5d34b320 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eb545432e69b4a2fb23a3d95298b6309 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/85a47d43e91b48faa87040b010bb25f5 : SUCCESS in 24m 19s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4e612b4d77a344569dacc63438dbdf89 : SUCCESS in 5m 14s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/9ed208a30c61456db0c038b9a9be5411 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/338b25b2870c48de90ee71cf3e614c91 : SUCCESS in 3m 23s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a948b6cc2c074fc2b495580fbbbad0f4 : SUCCESS in 17m 33s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4f58deac5466443fbd6f7c720120dcd7 : SUCCESS in 6m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e263a1e1719848bcbe5bea7fbd12e462 : SUCCESS in 58m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6b33790ec6248bcb1cd1179e6e43d79 : SUCCESS in 2h 08m 00s","accounts_in_message":[],"_revision_number":1},{"id":"02cf6f01acb5a6bed5bc75a2a1dc38cebac46f07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 12:22:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a7c220a135e72b9100f05dfdd518f068b73dcc10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 12:23:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22e389dcf4b544268ab4ded94e75584b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb2b0bc99b444db89176db60e84a570 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83f2abb2817711f18941ce4b6129bbb7613416fc","revisions":{"83f2abb2817711f18941ce4b6129bbb7613416fc":{"kind":"REWORK","_number":1,"created":"2026-05-18 10:49:17.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/41/988941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/988941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/988941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/988941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/988941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/988941/1"}}},"commit":{"parents":[{"commit":"016f93e7ade7ccd65f23b34f720439a65fccb772","subject":"Merge \"update constraint for python-neutronclient to new release 12.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/016f93e7ade7ccd65f23b34f720439a65fccb772"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:49:12.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:49:12.000000000","tz":120},"subject":"update constraint for oslo.middleware to new release 8.1.0","message":"update constraint for oslo.middleware to new release 8.1.0\n\nmeta: version: 8.1.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id52202c01f3de11240ef1d8d95893871570fe0eb\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7063949198001c02e13bce267b8a183544495080\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83f2abb2817711f18941ce4b6129bbb7613416fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83f2abb2817711f18941ce4b6129bbb7613416fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
