)]}'
{"id":"openstack%2Frequirements~988942","triplet_id":"openstack%2Frequirements~master~Iacc4158ad672e814bc063a62f58deeb6ba387ed2","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-05-19 09:20:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iacc4158ad672e814bc063a62f58deeb6ba387ed2","subject":"update constraint for oslo.policy to new release 5.1.0","status":"MERGED","created":"2026-05-18 10:50:25.000000000","updated":"2026-05-19 09:22:50.000000000","submitted":"2026-05-19 09:20:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988942-new-release","meta_rev_id":"c035f048f9275749431812d802d6440fa96d65fc","_number":988942,"virtual_id_number":988942,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 09:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-19 07:08:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-19 07:08:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 12:36:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 07:08:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a184ad806d4e80c633d8cbbb8cadbe7e22d3bd83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-18 10:50:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b66f29ffb9bc9e3d261aef4f0f6c759a9e127b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 12:36:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cfbcd5f73f4421cb7c2a0e47418f650\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b2266fa5cec43b3a868c85645178367 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65828ec888014708b1d99d7f7b0814ca : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f25eba4e999d46a8adede9b77057babc : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be479483b9ec481bbe32b70c44f145e5 : SUCCESS in 2m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/06dc74dd4c3d413baea2f59ad880a53f : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33bf9ea6a95549b0955acf5693ddbe71 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d172f4f220644ae8b485fd70ab249c5 : SUCCESS in 3m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f5969c291db241c8b843fc0a107bceda : SUCCESS in 8m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3faadc0c0807467da8d11b61182170f2 : SUCCESS in 8m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cac7bfd3d0024dde959c0dae5482a6f2 : SUCCESS in 5m 31s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/77cfb2ae5d6943518bcf6f7bec8d54c2 : SUCCESS in 4m 47s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/007a3e5b093a4d01853f9a8b5d5437c5 : SUCCESS in 4m 54s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/e9ff66546aa84f5c83bec70e670dc836 : SUCCESS in 4m 12s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/55cac84c98ae4f92ab258ba20d655b59 : SUCCESS in 10m 06s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/687f4dde1f734d6a9cf79ba98eb9335e : SUCCESS in 4m 35s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/c0a74bcbc2ae4766acf6529f2c9416da : SUCCESS in 5m 48s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/2504992b26a14c898ac44733dcd7ba70 : SUCCESS in 6m 04s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a3d5b38d00684e38a66a4a1186bff7f6 : SUCCESS in 16m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cad8b14c69784c1bb84ee7da91df314f : SUCCESS in 6m 46s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/64bdaeef066945a68c12c16cab3b13aa : SUCCESS in 8m 46s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/4e9d6cff41b24cfebd41c0da506d5c17 : SUCCESS in 9m 30s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7ffc258d6988434fba61f0da834b9bca : SUCCESS in 2m 37s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/ef767a10968a4057850da040268913a4 : SUCCESS in 7m 55s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/c3f62fac9abc464c911af91045f509d3 : SUCCESS in 3m 54s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/ce579e0f96ec456e9f1df7a8149d371e : SUCCESS in 3m 45s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/2932651c7a6d4b6b8742027c320ec615 : SUCCESS in 9m 35s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/6faa230a692d44ad9ccf92fa75306aaa : SUCCESS in 19m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/16070f83c3754a7487bc58d1f1fd6c6e : SUCCESS in 7m 32s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/72a083b649b4458fa842861c02052531 : SUCCESS in 11m 36s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/bb039e81523d4e9dbdd4ce48a04583c4 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f7fe515c2114f9280a398350c49609c : SUCCESS in 2m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c5552698f9b49ef82e2a233758e18ce : SUCCESS in 27m 01s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/23e0db314a254bc885591768c2f2d73b : SUCCESS in 5m 36s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/4e00e74ee1b84fff9ae793388f98646c : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/015da101566246f7bd4be1070230992d : SUCCESS in 4m 10s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/518007131c65417a87393300d7ea4bfc : SUCCESS in 17m 55s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a35fc78a6d7043709d15b3860f6902f4 : SUCCESS in 5m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb6cea00e9d144d5bc378f01b1e1b724 : SUCCESS in 1h 04m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86457ac56d72453881d5dd121d3be218 : SUCCESS in 1h 40m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1dd61904589f0572094de650ca6a2f39a38dfb67","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 07:08:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70d72e8f46c4945f5ae03eb723bd7adf537c5c33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:09:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c06d2c9f7d764456eb664b78c9f5a79f191ac7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:20:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03b4c7d2bbfa439eaab3887f061d887a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1343ed53cf4f42489dd3659d1a3a49f1 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1866aa6d2c9c46cd8225278ac458a80c : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fa8fc73ef3b4d76b190a4f51ef78559 : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7182a07f3a24577888eb664d5d15a25 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe5a12f8ed3b49138fe7a52a6ca29a1e : SUCCESS in 2m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a971197e73646b2a4e8e10eb54ebba6 : SUCCESS in 4m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/72a8a878c8884799a744dbd795398c71 : SUCCESS in 9m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/669cb220e3ba4b4487f0be97980b77a3 : SUCCESS in 6m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/50c85b6bc0e94005ac9b51c699c15c61 : SUCCESS in 6m 36s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/37f03b1e25424285a00e903963287d91 : SUCCESS in 11m 13s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c760f89a0725449281cfec93fdc79867 : SUCCESS in 4m 25s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/599943c38b23405ea193f01bc5570bb0 : SUCCESS in 5m 45s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e85ea7297e0b4a96bbb8e8572be3a50a : SUCCESS in 3m 51s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/9979b9dfa4674ac69975d51f50e018a4 : SUCCESS in 15m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba45f90192db41edbc1ac4f1cb2fd2ee : SUCCESS in 5m 50s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/48f6aa0ec960464f9dd0c7fb0dc47900 : SUCCESS in 7m 54s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e06c66e42abe42fb862ed7fa47422cc4 : SUCCESS in 14m 19s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/3412ee01f0364819bc3bbda05117b345 : SUCCESS in 4m 13s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/5b05471f037148f2b4d34d29248a0009 : SUCCESS in 9m 45s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/f61bef23feb3402c86c1f4e3d28f212e : SUCCESS in 16m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4baae01c407248a49120340ed3c6d180 : SUCCESS in 8m 05s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/66c181c2e1644073a9c3a568c2979fd2 : SUCCESS in 13m 12s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c8b69e23d4784ef8b7d1e0c02145e0de : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e7eead9ea8c3447c9645e7bcde5c4f16 : SUCCESS in 2m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbe0df015b71454288789418eecf39f6 : SUCCESS in 29m 34s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/8304fc25033941b799fc5e7511f2894f : SUCCESS in 7m 03s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/9be700dadf564b6b8c0a7f23532818cb : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1fa62d60a3194981a1a646ef75d5d03c : SUCCESS in 4m 01s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/b493ce602d9849debc4066e1aeb9ede3 : SUCCESS in 15m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/00411c3246cf43c9af4f8f089686a8cf : SUCCESS in 5m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/872e90629ca8425c92cdf51f8564d8ad : SUCCESS in 56m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5ee38d237be47cd93d3382008d566e0 : SUCCESS in 57m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a40c67567036dfb8ad9a2c81f31527843f950a1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:20:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c035f048f9275749431812d802d6440fa96d65fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:22:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/842b282936a74520811b14ac9d3e65c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d294a291c74b119762454bb81761a6 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a10d5075293c160fb3abaf89b5715ade31c75aa2","revisions":{"a10d5075293c160fb3abaf89b5715ade31c75aa2":{"kind":"REWORK","_number":1,"created":"2026-05-18 10:50:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/42/988942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/988942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/988942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/988942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/988942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/988942/1"}}},"commit":{"parents":[{"commit":"016f93e7ade7ccd65f23b34f720439a65fccb772","subject":"Merge \"update constraint for python-neutronclient to new release 12.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/016f93e7ade7ccd65f23b34f720439a65fccb772"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:50:20.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:50:20.000000000","tz":120},"subject":"update constraint for oslo.policy to new release 5.1.0","message":"update constraint for oslo.policy to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I433ddeea7058785ac8bd93772d4350144f563d8b\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Iacc4158ad672e814bc063a62f58deeb6ba387ed2\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a10d5075293c160fb3abaf89b5715ade31c75aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a10d5075293c160fb3abaf89b5715ade31c75aa2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
