)]}'
{"id":"openstack%2Frequirements~988943","triplet_id":"openstack%2Frequirements~master~I6b853c69fc3c7408d45247a1ed12c59a5063ea0f","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-05-19 08:27:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b853c69fc3c7408d45247a1ed12c59a5063ea0f","subject":"update constraint for python-openstackclient to new release 10.0.0","status":"MERGED","created":"2026-05-18 10:51:52.000000000","updated":"2026-05-19 08:28:07.000000000","submitted":"2026-05-19 08:27:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988943-new-release","meta_rev_id":"554d7c688c85974551a2742cefb8b64708e84eee","_number":988943,"virtual_id_number":988943,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 08:27:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-19 06:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-19 06:58:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 12:56:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-18 17:20:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-05-19 06:58:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"970cf06ac95eb9933edcf900a32248242eae3177","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-18 10:51:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ad8706d45bae3daf843dbbe2ac92ae2789ea23a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 12:56:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8184512cb35e48b1bb1df6b7a2312c7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39a1b256de1948779b0d5d2ec4d05cd9 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de4346909df45cf972b1f3624879518 : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/047a14f8b2e249af9163ad15973ccea6 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca00e4794f2e468991034bc8cf969202 : SUCCESS in 2m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/42faf64241dd443897f87005f8e97e47 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/64dd2fcb496547929643b214a99ee8a1 : SUCCESS in 2m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/930af6e426c24bec9c9621ee230121fe : SUCCESS in 4m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7e74598168bb47388e807da2e64f6aa1 : SUCCESS in 6m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dc02276d9aa4450386ba9d525d0fe5e8 : SUCCESS in 6m 24s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1d68a34fe32e4465a55c55f2ee0b37c4 : SUCCESS in 5m 05s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/5124c7f716ef49bcb1f75c3515606822 : SUCCESS in 3m 49s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/c31c75e55afe419cb8c62d00df95bb89 : SUCCESS in 5m 10s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/be34d21370194bccbedad6d77babdd98 : SUCCESS in 6m 01s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/bf28df3df0c242efa003d6d88980bcb4 : SUCCESS in 10m 12s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/9d8af57fe8604f22aba98cfdca5cc26d : SUCCESS in 3m 22s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/3b2b59def4c34294af45b003944610c6 : SUCCESS in 5m 21s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/6f566a945b1940a0b2c010267acd1cf0 : SUCCESS in 6m 18s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5b119ae67ead401db66867a142a13523 : SUCCESS in 15m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3082ae4d6604be1990869631552f81b : SUCCESS in 6m 24s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/70b4a58919704e5fbd9c51cb13b01c5a : SUCCESS in 6m 29s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/c0e7d774b12b49ae8eb5ad2401df04c2 : SUCCESS in 7m 27s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/069cba2006bc498c8837c1b12e8ab423 : SUCCESS in 4m 25s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/37e6e69443eb4c71b9617543d911c726 : SUCCESS in 9m 16s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/30e6548838054d36b2cdd5e2869b594a : SUCCESS in 4m 30s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/4fa690f8f4fd4ca3ad552030f74e46db : SUCCESS in 6m 13s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/4a2c21a3e41541e7a0da74503fb0c3e6 : SUCCESS in 11m 45s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/ae2dfda89a0042598d6922d61da8a907 : SUCCESS in 20m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4d1a0155f0a3435a95bf0510bf899f51 : SUCCESS in 6m 10s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/fc639ee12f2146ecb497cce7c05c1f55 : SUCCESS in 15m 54s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/fc08f7a5d4ba4f799ccb62b643f4c754 : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dacba8e34e9e4ffab1b043a563e824f2 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8d3608f23fb431292592f0aa7abc63f : FAILURE in 23m 45s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/188635bf393b4722ae283f5885394f65 : SUCCESS in 6m 37s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/ea12fd3f8ec54b1fb3d8dfd6253e093e : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95d89f076b1f444088ebf73dc1d086fe : SUCCESS in 6m 17s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/2a78fc8ce92d4c68b4ec81b8079d1fda : SUCCESS in 18m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dbf78b4805f5428b9b15df65e2c73dd8 : SUCCESS in 7m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29b44014a374463aba7184f8d1219b91 : SUCCESS in 1h 03m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fafc3459e624eda9f42c10d9b61a7ee : SUCCESS in 1h 58m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a7864f7a660eb7ba0b3b917907451631768b6d80","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-18 17:20:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e5cdb4abd0e7a1e1da4431afba6b90dea913aca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 19:58:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e597e336ee7a4f7abbb188aade5cf624\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f21815abae47219cae224a28edbb41 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d6ce3af9af4e1fa2ab50482f85bbde : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b65cc94772a4188a38af316d27d4121 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0182da7db8eb410aaf394b907fef3985 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/57179a921edc4ced8177a02b492494e0 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c47283dbc81a4135ba6e53551cdcae69 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ec0834c9e3f74e9ab66fe90394c65301 : SUCCESS in 2m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/334644b8ccd14886a706bb8c941872dc : SUCCESS in 8m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b163bfb49a184318910a899ed0a71142 : SUCCESS in 5m 54s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/949d55da23554be184b3bf3e151e55f1 : SUCCESS in 5m 31s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/7994db6e894c4f0cbc877750901c02ca : SUCCESS in 4m 30s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/620f1f59acb84b6f90d7fc1a92da5da6 : SUCCESS in 4m 06s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/439c6567a12e4876bff282e32550ecbd : SUCCESS in 4m 42s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d219ac3269b245eb8db572b73b08ebf0 : SUCCESS in 8m 55s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c4a5522a00254aaaabacea99e4b02dd2 : SUCCESS in 5m 20s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/4bc597a7dc97448d956a150732c8bb1a : SUCCESS in 5m 18s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/eb340cb128f74d639ae99be12ca6546d : SUCCESS in 5m 56s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/2bf1cfa169b745ec9197b08fe0c376a1 : SUCCESS in 16m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d08bbf9830194689af1099807d8f36e4 : SUCCESS in 7m 16s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/02cf66447e754c2799d9fe52d99bc761 : SUCCESS in 6m 56s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0b96fcc3126140a8a0fe9e059f540053 : SUCCESS in 11m 31s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/3a8aa74acbeb462a847a401201241225 : SUCCESS in 2m 45s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/5ad2d566da0347d18583f7885956e616 : SUCCESS in 9m 58s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/5133ddfaa5624943aec9880ed21e26b1 : SUCCESS in 4m 39s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/44b2c39f9554435f9b618ef89ac83a77 : SUCCESS in 5m 18s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/a17dfa4b31f34502ba93ec949916f936 : SUCCESS in 11m 42s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/f5a852df88b845218a60f46a1f9adc56 : SUCCESS in 25m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0f7475e42c774e9cbe686747cc487859 : SUCCESS in 7m 25s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/9e6c533e73874478a6e48e9684e40d38 : SUCCESS in 13m 55s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/7e186cba7dc340449c2e9e75da957687 : SUCCESS in 4m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cff2123cacdf4435afa2c967350d5c7f : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a11e375c5f03404b88c20ce2d17f4450 : SUCCESS in 26m 18s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/b12059fe23944a21944c5074f5afe1ab : SUCCESS in 5m 40s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/13d81c75e9074590b2d294448fd86fba : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ca66e95e52542118f0cb4d6daeb3ae0 : SUCCESS in 5m 48s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/bc50ec42d8a7479f8a6f4e23633e74c4 : SUCCESS in 18m 00s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a8ad86e567d74992a0344e562f9c19a8 : SUCCESS in 5m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe9ba862ab814be2a6ae7a09ea524759 : SUCCESS in 1h 06m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b696635fc35b4de5934203993ef263ca : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f389d04920042163a6a05e0fe1cd800e085ccdb8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 06:58:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e07c90f3ec48fb8f98a1fdcbb898a93825a0a3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 06:59:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21c8c2ca03d42b2bb7de76414ee76667ec731e49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 08:27:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51bcd58df1724f6da42811ec0676941d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3636b9d65d648a0ae556e657f62cb2b : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44e43f39c9064374a45e6334b7d4f00a : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a0e0eba3461472483b9c10cef9ec61d : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8110e181cd74c00a90b7c37510b5f9e : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b932a333c3840aeb4d99fea02b922de : SUCCESS in 2m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf8192c876754da4bfb37768f2bdf58b : SUCCESS in 3m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/67cf275d4fbb4a6aa385d93387a3064b : SUCCESS in 9m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1fa6103b0d9448b19bffe926b807565c : SUCCESS in 3m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8a81c09186b5424cae117d78acc0ba53 : SUCCESS in 5m 16s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/ae0334dd17434581a7e72a0deb1be8c2 : SUCCESS in 10m 11s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c65ddf2dc7724dc7b269a2ed9b0ae428 : SUCCESS in 4m 51s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/1ead2347a5e044fa8b67edf1d2797dc2 : SUCCESS in 4m 56s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/8614c4c09d034dbfb30178f0644edff5 : SUCCESS in 6m 06s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/8e124ea824ee4807851bbb4659de1857 : SUCCESS in 7m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/290eef5306f24656a1f93a4955b93702 : SUCCESS in 4m 04s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/21c91ac629a24973a5ef7f5216f02e85 : SUCCESS in 5m 06s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/5e2f745ebf7746fab9b6edd047eba90e : SUCCESS in 7m 27s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/11a5877837d9467da629a43f47dda1e2 : SUCCESS in 2m 07s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/1873ef195d424d07a48bd0b339abc7c4 : SUCCESS in 8m 41s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/5ac6141aebd549c897e956b8ceb125cb : SUCCESS in 23m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/31eec3a71b5f475b99ef7e940363243d : SUCCESS in 3m 58s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a975edf3a3dd4ab99b225cc035275450 : SUCCESS in 12m 53s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/9000102fc24e42e6b91dead79b0461f1 : SUCCESS in 2m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a04fdf1fb6024c66b44ab99650e96bf8 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9419dfb864964debb421c6483c06f7cb : SUCCESS in 25m 29s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/9a0b3c1834764b6fb07feac15df6c341 : SUCCESS in 2m 53s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/0049315746ab44e6aa6306dae45c5ced : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ccf57e6ccff9450aae191f0553e5fdf8 : SUCCESS in 5m 08s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/5d6134ce2ad94624b3232e1f003d1bae : SUCCESS in 13m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/594a4602b65e4f88b804b3b8fe2a2313 : SUCCESS in 6m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4982f65c14f4329bcf5b699eb05d14c : SUCCESS in 1h 05m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ab5dea63cde40ba88b5aec26576f0f4 : SUCCESS in 1h 26m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0f97cb42501b06a846e9f1f53d1ad8700aade008","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 08:27:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"554d7c688c85974551a2742cefb8b64708e84eee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 08:28:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71318c242b244bedb9ee37f511e792c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2acfac792324021a56ded131e4cdc0c : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6db4e0ec6f7d406922e91831d7c52701cbf3d68d","revisions":{"6db4e0ec6f7d406922e91831d7c52701cbf3d68d":{"kind":"REWORK","_number":1,"created":"2026-05-18 10:51:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/43/988943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/988943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/988943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/988943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/988943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/988943/1"}}},"commit":{"parents":[{"commit":"016f93e7ade7ccd65f23b34f720439a65fccb772","subject":"Merge \"update constraint for python-neutronclient to new release 12.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/016f93e7ade7ccd65f23b34f720439a65fccb772"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:51:46.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:51:46.000000000","tz":120},"subject":"update constraint for python-openstackclient to new release 10.0.0","message":"update constraint for python-openstackclient to new release 10.0.0\n\nmeta: version: 10.0.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: If46eb7f56d076f07e1bfc714febb70f214d597f2\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6b853c69fc3c7408d45247a1ed12c59a5063ea0f\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6db4e0ec6f7d406922e91831d7c52701cbf3d68d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6db4e0ec6f7d406922e91831d7c52701cbf3d68d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
