)]}'
{"id":"openstack%2Frequirements~988944","triplet_id":"openstack%2Frequirements~master~I1db3eaa12ba5021d85f1988d7ee28f13261c7af8","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-05-19 09:20:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1db3eaa12ba5021d85f1988d7ee28f13261c7af8","subject":"update constraint for oslo.concurrency to new release 7.5.0","status":"MERGED","created":"2026-05-18 10:52:55.000000000","updated":"2026-05-19 09:21:49.000000000","submitted":"2026-05-19 09:20:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988944-new-release","meta_rev_id":"6c5e4c49d19392643d95f3161a9adb12bd46849b","_number":988944,"virtual_id_number":988944,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 09:20:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-19 07:00:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-19 07:00:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 12:23:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 07:00:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"aac907a80b7ce349fa53e0894d877f9dd63a5f95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-18 10:52:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba4a137f4b348ad7b190e414d980cdf94c7a6364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 12:23:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/943f20fdf3f147cfb198ed5d3936aabd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f457fbaffead4b3193af9b6ba6d9008b : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7596dd1ca5d7477295f7b3fc09aebf4d : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/104ad8ec6a544d3c8a8b02bdf130cf9d : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab7f48daf5044849aa71fe786308eeeb : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f24abb378434376b8de44d867281bf6 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d36c21ebd7db438cbe630cee26ed09de : SUCCESS in 2m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f8b0e7e12a54885b964bf142e40985a : SUCCESS in 3m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3a7faa46ca5a4b1a891863317d640dc9 : SUCCESS in 5m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d17f4d52ea3b49dcb7404891e4aa3859 : SUCCESS in 5m 38s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/50e1d825b63245749fcadb61ac28aa4b : SUCCESS in 2m 55s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/fae5f9099a734cabbc15a89b6f9f560b : SUCCESS in 2m 59s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/9eac4cf538e74d5980f3e99fe0bc5a7f : SUCCESS in 4m 44s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/436034be04be4f36b2e4bf4b4e8bdf05 : SUCCESS in 4m 29s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/2ea58b2b320843fb9b3a2f05042e21e8 : SUCCESS in 12m 02s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/d0963829230d405ca27cd70cf8fcba2d : SUCCESS in 4m 44s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/58da7d14cd7f4a48b20f0df923a75450 : SUCCESS in 5m 00s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ccbec89dd4fc45c1903b34f1c4534712 : SUCCESS in 6m 54s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b05d1730fd174132bdab05026083a904 : SUCCESS in 8m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3cd9a919563f431fb83e5633bc50882c : SUCCESS in 8m 49s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b99d282ae8714e58bf8c28caa356f674 : SUCCESS in 10m 50s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/100431d5734e4e4889a7e9aa14b5ab22 : SUCCESS in 11m 05s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/81bed0ff37394d949e23263c63f7095b : SUCCESS in 2m 56s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/b3912bb5797f493cbe94339d4f0eabe1 : SUCCESS in 8m 51s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/a4f2959f68744d7f922a394d83bb94ed : SUCCESS in 4m 27s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/42be3aeecd774fb4a11b7d15a1ca0119 : SUCCESS in 6m 12s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/21d550b0f69746b8b2142201f978e480 : SUCCESS in 10m 36s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/8a0cc307b52643999437f1be7ed609e3 : SUCCESS in 28m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5113ac7141ef411191708e583fb61e2b : SUCCESS in 4m 55s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0aa8290accfa48f28bb7b21cfa28df44 : SUCCESS in 9m 47s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1f36304b9eeb44b4b2415fb7ad6ab2ab : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8e31ebe137804af08691cf4c5066509b : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a49e082e2f14e989a410dfe170c41a0 : SUCCESS in 27m 08s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0f1039a9b5fa4726adb799842f2b93b3 : SUCCESS in 5m 29s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/2c6c0b92b5d2490a9c6000837985f1c4 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/412c63b2b7784a94bb99d60e32ba3505 : SUCCESS in 6m 12s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d5573717837a40288e187e4b1253de48 : SUCCESS in 19m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a38d4888e4f444a29146fb7e180583a6 : SUCCESS in 5m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec03c77f1dee47e2badf15e534c79a65 : SUCCESS in 1h 07m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/482ba2aa102740ada4caaffd8e66d8b0 : SUCCESS in 1h 25m 44s","accounts_in_message":[],"_revision_number":1},{"id":"05de264a8834952d2b6b064449bccf44af2f79b9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 07:00:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb6e3d7f1234c5b55e099278103cd29a6aa8b71e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:01:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8385c0972f0c6a22539090908b18dde9c15ff62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:20:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0ba14ecc6d94fb2941bad3643978d80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c32312f83194ac1a847608d149d4fa0 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83071e08c4a646b1bc3e728810084f6d : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9bcd9f27cb494585b7b7cad892cca3b3 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/955888e77ab5458cbc963e734b438d8c : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac543070a9174889b4b70744ec2e17b7 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8e681415baf48b9a0232d6eeffa2ab5 : SUCCESS in 3m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/31132e89bcd0406485793ac4e8ffe9fb : SUCCESS in 8m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b75ae27e92414ab9b8a9ef11ae3ab7bc : SUCCESS in 10m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/fdc7643ac52e45e48f4b7e61db715824 : SUCCESS in 3m 03s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/36a3d1bb3b7c4b589c8d5844bac65f1a : SUCCESS in 9m 46s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/209c991ce0834cffa3655775c1a00414 : SUCCESS in 5m 12s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/88a023666b2c46199ff01f55f15535ce : SUCCESS in 3m 31s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9cb39b75574e4d6bacab14156786f5e0 : SUCCESS in 5m 14s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/329a62dda8764dca81217f9891c3f6db : SUCCESS in 15m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c800531c4c94ac7baa51df9e1479288 : SUCCESS in 7m 14s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/2b7ad3bd1cc94e12a3da1a5a8d128cc4 : SUCCESS in 5m 40s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/6962c54c03614e11b056e6ec13650e93 : SUCCESS in 11m 40s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/ea1af95f912345ba84a05fc7c9e0a552 : SUCCESS in 2m 42s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/0db4a1c6c5324f55a1a34077129f927a : SUCCESS in 9m 43s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/b6a31f01e20746908453e3bcc8269eff : SUCCESS in 30m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/17a3174b36d14adcad1c9396d8233a42 : SUCCESS in 7m 15s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/c03c598381794121aa4373de10a21c42 : SUCCESS in 8m 16s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/3dd1ba31677f4837b18d182f2a6daff2 : SUCCESS in 3m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/34d610871e8043798358fd3b7d92fccb : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c1f50b0364314c8abbdf132741985baf : SUCCESS in 21m 47s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/de83fda223424108a2e9ae1fa60b694c : SUCCESS in 4m 15s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/891a3a0bbe174844b16b96a541c68399 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9bdeffca021d47dc807b8ee45ee76844 : SUCCESS in 6m 25s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/efc4d897f24949198c1c1b433051ea90 : SUCCESS in 8m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3183b11dcf044cf9a62c158471be9f35 : SUCCESS in 5m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14a0b38ade3e41c88a95014a1296b3b2 : SUCCESS in 38m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bca6e87d7db045c0b18ff51c543bba14 : SUCCESS in 1h 46m 02s","accounts_in_message":[],"_revision_number":1},{"id":"156a996d9be428639846372d0ee3d63b82cb820d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:20:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c5e4c49d19392643d95f3161a9adb12bd46849b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:21:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d99e6048c79c424c9b63cddf709d9995\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/644b336b4f8e408e9190c5b9d1badb31 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc8ea7091764980182b85776b578f88195ab6b6b","revisions":{"cc8ea7091764980182b85776b578f88195ab6b6b":{"kind":"REWORK","_number":1,"created":"2026-05-18 10:52:55.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/44/988944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/988944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/988944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/988944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/988944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/988944/1"}}},"commit":{"parents":[{"commit":"016f93e7ade7ccd65f23b34f720439a65fccb772","subject":"Merge \"update constraint for python-neutronclient to new release 12.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/016f93e7ade7ccd65f23b34f720439a65fccb772"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:52:49.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:52:49.000000000","tz":120},"subject":"update constraint for oslo.concurrency to new release 7.5.0","message":"update constraint for oslo.concurrency to new release 7.5.0\n\nmeta: version: 7.5.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ic56828b2ca80962736c62f8296037d1dffac2f06\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I1db3eaa12ba5021d85f1988d7ee28f13261c7af8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc8ea7091764980182b85776b578f88195ab6b6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc8ea7091764980182b85776b578f88195ab6b6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
