)]}'
{"id":"openstack%2Frequirements~988945","triplet_id":"openstack%2Frequirements~master~I9467ae291968a8c0d68c3ccbb5de49e14c8d9f33","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-05-19 16:31:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9467ae291968a8c0d68c3ccbb5de49e14c8d9f33","subject":"update constraint for oslo.config to new release 10.4.0","status":"MERGED","created":"2026-05-18 10:53:54.000000000","updated":"2026-05-19 16:34:53.000000000","submitted":"2026-05-19 16:31:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988945-new-release","meta_rev_id":"16b59000dd60bf8797433e6324e6ea687dff762d","_number":988945,"virtual_id_number":988945,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 16:31:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-19 07:11:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-19 07:11:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 12:23:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 07:11:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"4c9b756e5466a2e1671da8ad1ae3a667ae3cd3d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-18 10:53:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e64c189f3676ff124860430b68ab53727a3e673f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 12:23:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5205b9600dc4c739e442e8c0c07b1a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c60b19d1954023b8362b4d3dadb25b : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7171280c8bd54f8f8fe3712a809c7b9f : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a649d07a0474688adac7f71da60ef47 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70029aa128e04ad78abe29e10477f873 : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6971d5093c15450f9c65f12cafae96d5 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/659d108b76a0457b9d70a03782fe9717 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b67d658a1ac24d6dabbd28222793361a : SUCCESS in 3m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/eed8933357444bd2aacdb4e5a1d35455 : SUCCESS in 3m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bebc150ee1d44d5187f7972fa3d91a0a : SUCCESS in 6m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/eb59b2498d8a4e8e875ac81f0378941d : SUCCESS in 5m 03s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/14bdcc935d49421a8aac8f60f079c765 : SUCCESS in 4m 18s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/c90f5fc7af7842b380567df07ed22d45 : SUCCESS in 4m 50s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/0350ad4aec044aee94af5957b4e8757c : SUCCESS in 4m 41s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/691da15ea2b049ddb2ceafc4d974aa4b : SUCCESS in 7m 19s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/5bc3912ab0c046bb94c9c23216635590 : SUCCESS in 3m 21s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/82ec28254eb14099a6675bc555e110f8 : SUCCESS in 3m 05s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/b9d37880e71344c789852dcb9b691f24 : SUCCESS in 5m 35s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/8ea5e6ecbed94c41b7dd840588807c54 : SUCCESS in 15m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f44d94bcdecb4e50ab9b70b79e173388 : SUCCESS in 4m 17s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/55fc490808f24c3a89aec5240261088b : SUCCESS in 4m 34s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/37c467d93ecf455db637e6f6ded6a5dd : SUCCESS in 11m 40s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/dd900eacbe1f4e94bd2daa5794c44948 : SUCCESS in 3m 38s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/58c1854fca294587bb30be147bab0998 : SUCCESS in 8m 15s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/796902fcaedb4cee90f17ffc2a58f554 : SUCCESS in 7m 37s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/92dd4155fee340d7b3d446d857cfcc7b : SUCCESS in 4m 03s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/f88cfe0c065b4959b70d584118d79e10 : SUCCESS in 10m 32s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/213c2fb102c94b7c978cdcfb197b1df0 : SUCCESS in 21m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/664dc30ad8a94e7e9c3d0f78028bde46 : SUCCESS in 3m 57s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/9277bf0d6ac14bf1b1874c0240d6c74c : SUCCESS in 8m 54s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/7900363b6a9c4d6bb1bada9c418e21c9 : SUCCESS in 2m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b81b2adc99704e1f82de1c76b65106e0 : SUCCESS in 2m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/12eceede941046789169420098dd35e1 : SUCCESS in 27m 45s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4d8d74e6086f45c2933f0d205dc9b3c5 : SUCCESS in 5m 12s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/9dfcee2b47f041b79da5904a56c22473 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b295a18c7ff4231b15a2f5445bae623 : SUCCESS in 3m 57s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d4951095322245969bba361a5a4ea009 : SUCCESS in 11m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/39188c4914be498fbafd1bb4a7c99d9a : SUCCESS in 6m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/356ca9456c144c4e82756b10bb69e384 : SUCCESS in 1h 08m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b32b6360d014d2b80efe1a0263af5d8 : SUCCESS in 1h 23m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e5c47014200f93b5e42acb688ef0469f50e2f44c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 07:11:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"05482545afb13292fece6643ae87d80880e4fafc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:12:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6bab50e00f4981509b056133f64de5497784fffd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 10:52:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94b8318be9aa4528af4507f97f018f9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de8c2e017c2e4428bfc2aef6906370b1 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead68afec3414331aba4dfd4567872ee : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c97b928e2f74318ab5ee902fd7cf58f : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/891c164993f94d059389c073db5bb588 : SUCCESS in 2m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc8fa37f105742b281e4e95db19c70b1 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/35092936a81c42bd990a0addea743ecd : SUCCESS in 2m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7ef1ef907de540b5add6161ef7b5249f : SUCCESS in 3m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d881bff342e74e2aacc3b7c8f69d36f8 : SUCCESS in 4m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0756a188ea7f4124b8c6645e3363dd27 : SUCCESS in 3m 50s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/dae4bb8da1b54da99e50c5723933630e : SUCCESS in 10m 55s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/9095e2a64e004d99934c6c2513064a4c : SUCCESS in 4m 49s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/8fd54b309a3a4986ac4de75bd03d9087 : SUCCESS in 7m 01s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/001ac6cdbaa94350b15bc4a266c589f8 : SUCCESS in 6m 31s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/3fc89dc63c3543699fc0aec30fa04cf6 : SUCCESS in 17m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d063ebfcb3d477ead6eb2b1a7c30868 : SUCCESS in 7m 03s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/8d90bed4cbbe46e9af017d06a520faa5 : SUCCESS in 9m 19s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3eaeb1dd8bd542b6ad1ffacbc685eb8b : SUCCESS in 10m 29s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/4410291f69e940acb4ea6b1a4c2b4923 : SUCCESS in 2m 53s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/8161a0449f904913b2d192702d0fa238 : SUCCESS in 7m 53s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/9938e72fa0534dad9ca24c1e7cd6bf7b : SUCCESS in 19m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d4dbb30f6518490fa81bdecb9e5744f4 : SUCCESS in 8m 43s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/563e5c06b4a542d5ade22f2e67cb5069 : SUCCESS in 13m 42s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/342731a73e2943a88c31ab1f862f228f : SUCCESS in 4m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/830a32ccb51a48f8ae0d4098e0dc5d93 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea275d9647bd452da4344b5de375f745 : SUCCESS in 25m 08s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/a47867010f954c1688b403a44e6c313a : SUCCESS in 5m 10s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/091cb12e06814c7a911e02623fef1a7f : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd1c617448a44f749b5f245e428e247b : SUCCESS in 6m 16s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/186bcc7caf3f48e093ed095373fde561 : SUCCESS in 12m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e88b608612754bc8965894e3d50e813d : SUCCESS in 5m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e525f6b0b25545309051c727601a1276 : SUCCESS in 1h 04m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6b38441f4bf455c810615058531421d : FAILURE in 1h 24m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1be8396ed7edcb1017f722a3a297c129a7374e42","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 11:45:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc3992ccd0e1860027b60cb81c4ab19e2c137af1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 13:34:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5ddc75338164198976a6c91b272e6d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/932635227f8c43019620d06035adb1fb : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b9df8400fc64394a7163a7904058b2c : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a61e6c2016f4226bc76b00206e0de85 : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b387d2c7a5be4a5395c2e2d57b3b7df7 : SUCCESS in 3m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d045dc003b264c27a3f45aa479301331 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/056823bb6b20400495daf524f60b3496 : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5236a191b0f849f3a58fa2188c2eddc2 : SUCCESS in 3m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f0e06bf5440843678193eb779293421e : SUCCESS in 2m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f40165b66f284913bf2f83de83565ba7 : SUCCESS in 5m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/47c85a688bb24170bca03fea8951d21f : SUCCESS in 3m 20s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/592e33f3ed5b4e3ab4ed5d2975c0d20b : SUCCESS in 4m 30s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/f0068e2d3d5645569ec01a6e34da8df5 : SUCCESS in 2m 47s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/1e81a621dc2c45aa85671a98b36530d2 : SUCCESS in 5m 59s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/b58ebd1b7b8f4ed68638ef54dfaf0bd1 : SUCCESS in 9m 46s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c43d7a012de7401a9e38c420f92bad68 : SUCCESS in 4m 58s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/770b4db089bd486589ca97faa1b4a468 : SUCCESS in 5m 31s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/23ec3170eda243ce8754ec41e112c0f0 : SUCCESS in 6m 39s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/27d90e49e48545229c3977945ac89156 : SUCCESS in 15m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e8b6d3d79ea94f3c9b7283de3a7b61b4 : SUCCESS in 6m 17s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/7af9d23d21234423a50865f48e704839 : SUCCESS in 7m 37s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/442cd7e9c6374c79aa226fe5467d535a : SUCCESS in 10m 41s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/86b0d785925d4a23a42aa3a81db8d754 : SUCCESS in 1m 41s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/29539544efb24686b214ab366b469da4 : SUCCESS in 4m 16s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/85405117b4bf4566abf580a687469607 : SUCCESS in 2m 25s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/e90825e8aac44d1388774da46bf1617c : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/46c58adeb81c4307925a43e7f2a833bc : SUCCESS in 11m 06s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/67bf9cc1e3ed46a99e69990be643c10a : SUCCESS in 25m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ba3559c8daff42449f4d88b1ed361da4 : SUCCESS in 3m 28s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/86683afb9afe4ac18eec757e9e3dd332 : SUCCESS in 11m 06s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/ae5b22942c2946aa9ace5e5668ca38df : SUCCESS in 3m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa8450e73594409582dd2386fdded4f3 : SUCCESS in 5m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/428fcdea34064fce90d3a985dfedf7e5 : SUCCESS in 22m 57s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/9529c15ddc4d460dabdde58d1e6bff1c : SUCCESS in 4m 45s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/adc47fbac3a746819bd6ee757f79c23f : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c018f682df354d758a1f9bcb81827f69 : SUCCESS in 6m 20s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/1d169608ffc648a1914a8c0b193b48e0 : SUCCESS in 14m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f07975bca6624c669e8024db59669e6c : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d3199e93d4248a0aaac29117a3ed1cb : SUCCESS in 1h 03m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71e07c044a3e46b7a711f1362e51df37 : SUCCESS in 1h 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"722a3984fcc606402a3bee4664612f58f9abe9f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 13:35:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47758d05514a4be29812061ee3c41ac93e57579c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 16:31:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87cc1c3888a2451e983ab8c7d989ec60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8e228a09714e2d8d17e05e9069a842 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/494aeacfc63749dfaae434b1ad6b7386 : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b671dbd510b413193f247f7bda964c6 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e97b280ffc35430ea6b1c40a141b16a4 : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/264e09173258421f97438449dd387c66 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3d0f74627c741b5affff15c2ddf088b : SUCCESS in 3m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/47cf1b8d7b1c466aac3d993d1d55ea24 : SUCCESS in 7m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bd2fe59e21bd4ecd8274b98d30b8832a : SUCCESS in 4m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ddba6020f1ed426f8edb93fb0209c34a : SUCCESS in 6m 32s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/248bea12bc6748ff9f286ef51895453a : SUCCESS in 11m 30s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/972cea8b21a7420194745176b5652214 : SUCCESS in 4m 28s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/f9d929d8d3214f14a3d0664911b027d8 : SUCCESS in 4m 56s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e67a505565d642c38aff40b0433aa34b : SUCCESS in 4m 37s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/1739f80112984853bbb193956c3c73bc : SUCCESS in 16m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac6088f58c8a4ef4aececf23f6a1149e : SUCCESS in 6m 52s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/907f07156ae74b03ab1f247c13606c88 : SUCCESS in 6m 59s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/816acf24797a4ccfa020a8445959cee2 : SUCCESS in 9m 13s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/f1ea878a0f8b4266a9d2f8006df68d9c : SUCCESS in 1m 51s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/228fefab82a548bb87d82ed87261eff8 : SUCCESS in 8m 56s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c91129cd602b4bec82d632dd2ec2ac93 : SUCCESS in 25m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d31c3af8ba56429d94af517edffe2a6d : SUCCESS in 7m 30s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/d63ff39873ec436ba7abdc0f1c79c2c3 : SUCCESS in 13m 16s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/ca23b03cf8ea4424bcbfebf8485a714c : SUCCESS in 4m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3747c6aef1db4d12aaba136eccf5f349 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4df1ea78be44dbaa0ccce6e88155c33 : SUCCESS in 23m 55s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/6b0ad7ede4b4480b801d8d3180baca54 : SUCCESS in 5m 24s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/3193bd3808754f11945c6e5e794e7ba1 : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/11fe923f10ec4b979a99ef2e6bd8d78e : SUCCESS in 3m 58s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/b0218aff048449b5957c4387fa3238be : SUCCESS in 19m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b992a080c6104fa6a76aae3d27ebe894 : SUCCESS in 6m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77946f03894c42beac1aa73fb5fe9b52 : SUCCESS in 1h 10m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/076161e4ff1543d88ab6ee7bfab99aa0 : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e7fe8aec0c8f923194b0de9a196add161d02005d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 16:31:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16b59000dd60bf8797433e6324e6ea687dff762d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 16:34:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d275cb10a724fd6832c7cc982215d9d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f3d7f1616549098bb491c2ea336026 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b25eebfbf1c43e438c2e8e99d255fc8ba9a86d46","revisions":{"b25eebfbf1c43e438c2e8e99d255fc8ba9a86d46":{"kind":"REWORK","_number":1,"created":"2026-05-18 10:53:54.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/45/988945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/988945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/988945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/988945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/988945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/988945/1"}}},"commit":{"parents":[{"commit":"016f93e7ade7ccd65f23b34f720439a65fccb772","subject":"Merge \"update constraint for python-neutronclient to new release 12.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/016f93e7ade7ccd65f23b34f720439a65fccb772"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:53:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-05-18 10:53:48.000000000","tz":120},"subject":"update constraint for oslo.config to new release 10.4.0","message":"update constraint for oslo.config to new release 10.4.0\n\nmeta: version: 10.4.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia5e911c4eb897a52a4dd5fea885cc51bb88b8483\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I9467ae291968a8c0d68c3ccbb5de49e14c8d9f33\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b25eebfbf1c43e438c2e8e99d255fc8ba9a86d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b25eebfbf1c43e438c2e8e99d255fc8ba9a86d46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
