)]}'
{"id":"openstack%2Frequirements~989093","triplet_id":"openstack%2Frequirements~master~I687dbf311bfd778cc2f30415c1cfad9eeceb23c8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I687dbf311bfd778cc2f30415c1cfad9eeceb23c8","subject":"update constraint for python-barbicanclient to new release 7.4.0","status":"MERGED","created":"2026-05-19 10:05:41.000000000","updated":"2026-05-19 22:16:05.000000000","submitted":"2026-05-19 22:15:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989093-new-release","meta_rev_id":"aec93665384b49e2a7874f39984c508ffc2d406c","_number":989093,"virtual_id_number":989093,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 22:15:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-19 11:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-19 11:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-19 11:52:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-19 12:14:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 16:57:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"85656d25476b183de992bb9ce24727809d8d0771","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-19 10:05:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76a00a5b417921325a478c58e86a115ba399d2c9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 11:52:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1532ed14258aaae0c7a93016340ff224ac4cb57d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 12:14:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f9ad199ff694b23bb975af0da5e9661\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f92105223e7d45b5952989256b069fb3 : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b6e010474642588c2f5c92ea1bc994 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d300d588adff4474a188cf1f1a6c9f71 : SUCCESS in 2m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/801515f54d78486ab87e0bcb47fabdef : SUCCESS in 2m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1415063af3174ef797747968af44e369 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b843502e3e64485a89a4670b5ad22e8 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8cc5d69886b4407292405ebe72aae71d : SUCCESS in 2m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/194e674050314602ac8a544fd7307ee4 : SUCCESS in 8m 42s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bb652d370f9443488d96e6ca31b53870 : SUCCESS in 4m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4f29bb2ac37c456ab9679d51adf67d0d : SUCCESS in 2m 43s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/73acaf1d14f742569e7352687e2414e5 : SUCCESS in 3m 58s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/1fa206fa4c754c51b18775f7db2aab25 : SUCCESS in 4m 03s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/f2a5753d360348fcb56030e11c6b066c : SUCCESS in 7m 16s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/aa1ec00965d54dd4ad6e978342b04f20 : SUCCESS in 6m 58s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/ca2bc47280484929b7a068ddd5a53e38 : SUCCESS in 4m 51s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/e4561f3188e34fe79822159f3039afd4 : SUCCESS in 3m 13s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/7701ac3a16df4acfa29f1ca921cff209 : SUCCESS in 6m 24s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/de190d7a93394dc18ab7ac0ed8183a07 : SUCCESS in 16m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c561dce2a2f4dcc816581df4a3b4b02 : SUCCESS in 3m 34s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/417ef6b82c5a402b9bba30c6d57f8edb : SUCCESS in 4m 46s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/4363e8c51ab9406986e4ddeefe073fef : SUCCESS in 11m 09s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/1b74ad4b6a4146169d812badcad0fe7d : SUCCESS in 2m 18s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d8d517b4a93843e4abeda0a19ce0194a : SUCCESS in 6m 42s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/f25edc71d29c47879500706dc44a7c1f : SUCCESS in 2m 43s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/87979bc60a0d40adb56e70b9ce329263 : SUCCESS in 4m 30s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/dea811dac3594b7a8c4bf3fbab7d226b : SUCCESS in 10m 55s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/f24365ef2bed42eebdc885d250691af8 : SUCCESS in 27m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6c77d75d35e64e5aa5805297df2d42ec : SUCCESS in 6m 06s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/49ca4104d437479d8fcee3cbba97e698 : SUCCESS in 15m 08s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/50df6f0d78814c6881375537f9f01807 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3dac07f9b53b4121a636cbc7d4565d7b : SUCCESS in 5m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bbec9da154ca4b54a113e026f02d23b6 : SUCCESS in 25m 29s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/b16c77bef8304ce0b626c4d6e94f99ea : SUCCESS in 4m 00s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/efde12eeacf942d6af2b9880aa1f8dbd : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/25d326020a184573a22d810d57676ca0 : SUCCESS in 8m 20s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/814e935271f9486ca8c20934d52f4412 : SUCCESS in 18m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/28cbffd748704e2e8aeb3fee92e0315c : SUCCESS in 5m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f6d5d3dee5541f989d31ab32ed81b34 : SUCCESS in 1h 03m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/701a15831c97475b8353f994376b0ce0 : SUCCESS in 2h 02m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8d8cf9e244f06ec0c47268366df5f8f89c8f0207","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 12:15:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"966e40a8e3757d99bbf9beb155e94f8de61160f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 14:57:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/947afb294f69448684c9fea52e32b0af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a3f29a976bd4af398a742b4baf632ad : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a288d82de61d47649eee191ec88e37a7 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d5c31e9aa624008a5881b45f59a3fe2 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4415c689e85649e59bc8a407e6f8cbce : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e2b2d5a23bd4c9e90f93e468c089eff : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe151791ea2e427fb69f6331944f83af : SUCCESS in 4m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dfcf84c2e1ba4a51bd5e868aaa43d5b6 : SUCCESS in 5m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bdf609dc709042c89268597a85487b46 : SUCCESS in 7m 34s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c2af4095c9874d0582f4b8ee6e71ce0e : SUCCESS in 7m 49s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/625a1485cef1452e9a4ad2d93c675075 : FAILURE in 10m 59s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/7c18ce7bf6a84d1483df44e7a291276e : SUCCESS in 5m 01s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/6912a20de032433989d17968115bb60a : SUCCESS in 5m 56s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/d1c19fe2b72f4e678bff2a97291711d3 : SUCCESS in 6m 46s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/940cac7a464e4d2191ce4d6f593ad9ac : SUCCESS in 15m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d2a176697ed4dad98105e94b2ffbe26 : SUCCESS in 7m 53s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/7e7f8989c97a4ce48ad85b300db4f0a8 : SUCCESS in 7m 27s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/01291e618c5941c1ab2f5487c1d899b8 : SUCCESS in 9m 27s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/837ce46cb3cd4f73995977afe19fc5ce : SUCCESS in 3m 27s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/64e07775c37b4d8cb44147145eec4711 : SUCCESS in 9m 24s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/90d4a643a0f344afb83aa124c833266a : SUCCESS in 25m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b559b44777a442d38beeea38d3a7731d : SUCCESS in 7m 18s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b087a0836e634e29b4c0a672dd561b2e : SUCCESS in 14m 21s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/4792c6e622d0483093ece1ad3101baa7 : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d2b5eb34716343cba3363557ab5f1947 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4744bf3c08847249470b108e8a73391 : SUCCESS in 25m 33s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/7bce769f789d49009c79764695c92448 : SUCCESS in 5m 34s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/29cfb4ac2941404c8033c6e5e79132c9 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a3b0431d91f148dd997d8e49d480953a : SUCCESS in 6m 55s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/02be53646e08434f9862dd40d9dda0f8 : SUCCESS in 21m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/afca85490f594f74abffc3d39987e252 : SUCCESS in 5m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdf5992d459748a182ea811deeaae399 : SUCCESS in 1h 04m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0bf85a28e08e4a699657cee850c07bc6 : SUCCESS in 1h 57m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a8a927249c03b9542b35255315a526c440146b94","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-19 16:57:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37680c2ce45dc794294686a0648630dcf9cadcd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 19:07:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/107a259e56f948bbb82359cc322b2188\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27304b2cfc364d5b9888b3856a4240f0 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb22a9815928438a97db5acc392ba078 : SUCCESS in 2m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bcfbe5e4dc2747f28f3440ad00558fa5 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2eedbb16035d4cc287490cfbcefb3389 : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/91a4b7bf9e89494f81c9c758a5b65bc7 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df8a37d73f294adea65de36215edc0e5 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd23681f9bad41788b6b8ef5b1c976be : SUCCESS in 4m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/886accb2dc5b49eb8ec69e09d9746d7b : SUCCESS in 6m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bfd0a59dca624b35ba9929200ae99ab3 : SUCCESS in 6m 04s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/34add1f2ec2f476994504d73b04e8351 : SUCCESS in 9m 36s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/94bbe7a2642848d0865d912b2b056782 : SUCCESS in 6m 14s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/e6fc4e2a5fc94181a636236a1a6c1fd7 : SUCCESS in 4m 57s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/490ecef5ce754f1c89e9ea9e76056d69 : SUCCESS in 4m 01s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/88db15901edc4c879d524976049f47be : SUCCESS in 11m 15s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/95fef1d0d07c4551b2ada53dc4488f3b : SUCCESS in 3m 47s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/a3685f638339451e9090c37fed0a8fc6 : SUCCESS in 4m 32s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/30a0695b968b4b339814676ace9cc81e : SUCCESS in 7m 24s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/16366d6bb45f474fa697c88e89289fe9 : SUCCESS in 16m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a647ce758b4e431bb3020dc21ed1d952 : SUCCESS in 5m 58s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/44dad98c603142deb2dad7532b5899d0 : SUCCESS in 7m 05s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/1f8e87b4e0b54921b49efa06ece0d46e : SUCCESS in 12m 40s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a920cc6eb21e45eba75d667e0105f9ef : SUCCESS in 2m 48s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d6805ec8184c4ec492318b5dc27a4689 : SUCCESS in 11m 03s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/d4385fd9d03447b59b4e7722af3294ef : SUCCESS in 3m 41s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/36fb1ed158804c9e9fa1d5be8655ea8b : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/bd7d9954d7854e949092530a1843dd4a : SUCCESS in 12m 10s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/54b1da1718434e598189c5e9a1dbe976 : SUCCESS in 23m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ebedfe88c95742f5af4a677a4a7fb048 : SUCCESS in 6m 53s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/e8862bf5826d4d1780c36ea62cd6d4c1 : SUCCESS in 13m 00s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/cac711af813b42cbb6050d2d1cd7989b : SUCCESS in 2m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/635f440092ee4d47a5cddd2ada0ebe8d : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ecfe8b8066b84caaa23497a6b475703d : SUCCESS in 28m 36s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/98c68a568bc34c0299c27d0fc1bd6917 : SUCCESS in 5m 30s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/32ee2de328f5421fa262da4676088dd7 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/29fd62e597704ef3a42889f1fdda27a9 : SUCCESS in 6m 53s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/6f3b5905b07e400288acb6256e631ce5 : SUCCESS in 21m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4712c89486514011b94deadc4a3ba82e : SUCCESS in 7m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32cffbb629a64e86b3ea8aa68cc31e55 : SUCCESS in 1h 05m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18e57d7226df496cbf15c56505b23c70 : SUCCESS in 1h 59m 20s","accounts_in_message":[],"_revision_number":1},{"id":"48372889f1e959c50c2fe4980700ad5c3d6623aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 19:09:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab1e159927c6bb4533a8ff0b91bf8df3b6e13173","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 22:15:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e8d99d6387f428f85d2ff9357479bfb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/131f5e744cff44b0b156b6d0d330d6cc : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab65c206beb4841b615dcced8518d21 : SUCCESS in 2m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09e83713c6e248e49099abb4af149294 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f20ac3e0739437991951238d94a8140 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58f7f7b0171846f69074d80f3b953888 : SUCCESS in 2m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/466f5c8ee4b946f39c7d443c38d5331c : SUCCESS in 1m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1027e584f3624caebeeb1a7614b4d348 : SUCCESS in 6m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4723ea02c31c417d808ba0b90eea8230 : SUCCESS in 3m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/253418a09c9245d9b631a5cc4aae2347 : SUCCESS in 5m 44s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/fe7c0519e8094947bd37fee7686e6dd5 : SUCCESS in 10m 38s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/99a34122d9dc44d3856acd3e93411b96 : SUCCESS in 3m 17s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/3bdb4c0f74314cbea8d03ada011e639f : SUCCESS in 4m 18s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/30f973431a0c43ee92bb36b6c9100dbb : SUCCESS in 3m 42s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/9a0a40b1e5b04aa99d9154d5ebae6af1 : SUCCESS in 7m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e7cef439576d46889011b63163a8186b : SUCCESS in 8m 00s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/1da0e56d519d4b5197c3464a2086bb77 : SUCCESS in 8m 13s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/63842a9ad5a542e09a7581edacc76ab3 : SUCCESS in 11m 53s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/f9a59de6b8574bde899b9c5dae44cf27 : SUCCESS in 1m 40s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/f80fc94dc5b24c0ca095febb51c7880e : SUCCESS in 8m 35s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/52d71811166a418382672c26f0d7c8d5 : SUCCESS in 24m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7f923eb2a01147ceac79a906aa709add : SUCCESS in 3m 36s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a9a57d61585e4b1c9e504ab2679cf5fb : SUCCESS in 13m 09s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/f6de6fb6f4d046af89c18825a70b4548 : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5785f97fcc6c4c4ca34e11f0c71fc45e : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b394fc80daf47219a8dd4d62b200d9a : SUCCESS in 24m 14s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/10f10773346f44d6a71a366f8713e213 : SUCCESS in 5m 38s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/8739b552c6bd4b4f8347b70ca1215cbb : SUCCESS in 3m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c43dca220254edc938be1be171012e7 : SUCCESS in 3m 22s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/c1cf36d93c8d428ca432482af23c1da6 : SUCCESS in 18m 05s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/10616ab6a761426b81d4808e2fedab26 : SUCCESS in 3m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7e6bd2cd9b54d64bfc8143c09f7ea49 : SUCCESS in 1h 03m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56a064859d994585b8877503eda63e4f : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a22bebc1a2c31374fe9ced62f0c7017a84816016","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 22:15:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aec93665384b49e2a7874f39984c508ffc2d406c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 22:16:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcf58cd507c44b83b094b1a71b0c5147\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e89b9ca1c3449994735d76c2e85ec7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"243d5b7765347a9e60374d2252361df5f1d99840","revisions":{"243d5b7765347a9e60374d2252361df5f1d99840":{"kind":"REWORK","_number":1,"created":"2026-05-19 10:05:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/989093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/989093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/989093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/989093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/989093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/989093/1"}}},"commit":{"parents":[{"commit":"3edf0209c9d4af934c305027c4c5ca2accb73960","subject":"Merge \"update constraint for oslo.metrics to new release 0.16.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3edf0209c9d4af934c305027c4c5ca2accb73960"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-19 09:55:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-19 09:55:39.000000000","tz":0},"subject":"update constraint for python-barbicanclient to new release 7.4.0","message":"update constraint for python-barbicanclient to new release 7.4.0\n\nmeta: version: 7.4.0\nmeta: team: barbican\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I0232f755cbc23a221c05ae924fc72c74908f2e8b\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I687dbf311bfd778cc2f30415c1cfad9eeceb23c8\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/243d5b7765347a9e60374d2252361df5f1d99840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/243d5b7765347a9e60374d2252361df5f1d99840"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
