)]}'
{"id":"openstack%2Frequirements~989133","triplet_id":"openstack%2Frequirements~stable%2F2025.1~I575c54ea977578880010dcdfb4fbb14a9135b776","project":"openstack/requirements","branch":"stable/2025.1","topic":"new-release","hashtags":[],"change_id":"I575c54ea977578880010dcdfb4fbb14a9135b776","subject":"update constraint for oslo.messaging to new release 16.1.1","status":"MERGED","created":"2026-05-19 13:43:42.000000000","updated":"2026-05-20 09:25:26.000000000","submitted":"2026-05-20 09:25:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989133-new-release","meta_rev_id":"88bf8ba94429c0acd10fbea07d337a122d99eb34","_number":989133,"virtual_id_number":989133,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-20 09:25:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-20 07:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-20 07:42:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-19 15:33:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 22:14:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-05-20 07:42:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"e804fa20dab94b5b891465650872f5c1e464c5e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-19 13:43:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65ce0f1dc241e029227019af54e7fde6f8c76686","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 15:33:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6242e26fe68042599b9fc257e6c04914\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a29dfeefa8a24f4fb27edc1416aa26fb : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d4e550ca05f46a8bb68033ee0c5e86b : SUCCESS in 2m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83cdbfef715a42aeae1c9a8a9e54d0c1 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ff2d238e2a74a4eabf844bb1c52e270 : SUCCESS in 2m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/261e2d7f77484ca2a7e708608eeaa994 : SUCCESS in 3m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/14df089c830a47c98f58478c55685058 : SUCCESS in 18m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/84fb4613a22243f4a5c0843893e2da15 : SUCCESS in 11m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1fb479aab7254347940baaf42d433e25 : SUCCESS in 7m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cb5b82098dfe4c9f805f44e19eeb0db7 : SUCCESS in 23m 52s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fa563cf712c14df98f63bfadbef3ad01 : SUCCESS in 3m 28s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/36f4ac5421c842d8af9d7454f5e0b727 : SUCCESS in 5m 04s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ac98599a33a542dca2dee924075c369b : SUCCESS in 4m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c0c6fdf832784fdcb1c03bc13406ef3d : SUCCESS in 12m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a53fe8b83567428b970a98d1fb1185d1 : SUCCESS in 5m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d0d424c652214e51a1bc62d317bd8667 : SUCCESS in 6m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7e4a444837c2408682663b71606078c1 : SUCCESS in 8m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8199623544144172b34e8fb2c0a3916e : SUCCESS in 15m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dc368028ff3b4a07aa20a46fe0bde063 : SUCCESS in 6m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/be8d82de551c4cfb8e53d361b72846e3 : SUCCESS in 8m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a75f7e8247b94704a38c88edc31a6177 : SUCCESS in 10m 57s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6895f31fb7a54dd9885b2af39152b3af : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4f567c897e0e437aa848c1d161353037 : SUCCESS in 9m 25s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/83905805a47b4a09b2c2c15673ad50bd : SUCCESS in 2m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d7117542e19d4b73878092cc778a8525 : FAILURE in 4m 21s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/800d87a1194045d39d505468748ea1cd : FAILURE in 4m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e82ccf62a10e4a6db22f75dd416204b5 : SUCCESS in 26m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0ade0c30997f482595e947a1dbdc1d00 : SUCCESS in 7m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2374ad03732b406d9646656be57402f9 : POST_FAILURE in 26m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/18a539e04e98426e81e94a2f85d22385 : SUCCESS in 6m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/84a4419b9a2647caaceb0632fafe4204 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82d61dccc2594c3691498f8c285a63f3 : SUCCESS in 28m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6f7765fb52d643a8922bc7c2d355aaa9 : SUCCESS in 7m 05s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2de55ea1d5ae42cda00b368e55f9491d : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f3ded646f6ce40f1bdd7b67d35a4cad5 : SUCCESS in 6m 02s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/42c5e696df384f05bc17940e3a46f6db : SUCCESS in 20m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/318aa18721b4494b81f37f7dd4f07d49 : SUCCESS in 54m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60b7edb15f0044cb956ac0678f8200be : SUCCESS in 1h 44m 03s","accounts_in_message":[],"_revision_number":1},{"id":"123d2bfe3e6c05901ae456527930c06a3a76c0d0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-19 22:14:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac903236cfe74eca546d5011270cfc74feaecd4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 00:06:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f50bcadb263743e0831b5d433e60cf9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a89f3b3172404b95e066ab6dd56d68 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5913f0d98a874e6aad33d3012f9557d7 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c71b65afea3d4460ab1ebe3f1d83327c : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bed107d974d483c9a2b95e2cb97c2e6 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a4b8fb2adb3470eac864c52b39200ce : SUCCESS in 5m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/87d47faeb1d34659a9fe83a38b370284 : SUCCESS in 18m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/16fe30eeff1e49aeac68bf6c58289d19 : SUCCESS in 17m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2af1faf2c7674d71aa827f096dcc2421 : SUCCESS in 13m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/70a5f5cfeed142e481ac510965eb7798 : SUCCESS in 20m 54s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5e91e039c7f049d0accca7bcb1b6254e : SUCCESS in 4m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7ea6a068e539468c8a829cd8e476f0f5 : SUCCESS in 5m 17s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/573451e8939642378f636eac062a8dce : SUCCESS in 5m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/53ae78e9a7854371a8224c01b03ff64a : SUCCESS in 14m 05s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f3968cd2f5af4f39975af384bc506126 : SUCCESS in 3m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/20a0294d9f644f819cc68d2ee0c80d84 : SUCCESS in 3m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c279d5816531416985ac4d71cdd137cf : SUCCESS in 6m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/843a04e3afac4ba6966c71a153fc2bb7 : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/618c9f3e79ed4eab9110f83c3b54d6a7 : SUCCESS in 9m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3ac0cd1bdb1e4588830b98a095264f1d : SUCCESS in 5m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3122776afc714c688cecb5c091011f24 : SUCCESS in 13m 14s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/85288b219b0c40afbfb612a597812a2c : SUCCESS in 2m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f8f00158196b4061abef742bcbd83da4 : SUCCESS in 6m 44s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/189d45bedd9541a288395920978f2ce7 : SUCCESS in 4m 32s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f92e6a827522498ca61266a9c75254d8 : FAILURE in 3m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d8a70d67f3574bf980612cff60b3dce7 : FAILURE in 4m 43s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1550130e930c43ea84616f02a725b916 : SUCCESS in 19m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6cfaf7f8cf104a0f9bc423d62deb6842 : SUCCESS in 3m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a8b25be3eb8e446e9d395f603e3c89f6 : SUCCESS in 13m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/778fc56f248c4c538f14dbcafdfbc584 : SUCCESS in 5m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/365d66f979a74a44816e5cd054d3547f : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4033218007ce41e787f17769ce8c87a0 : SUCCESS in 26m 48s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5443757223d8453abc970452d0a5bc6b : SUCCESS in 3m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b7034c6c6a75436f895cc74d7953943f : SUCCESS in 3m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ea7f35f2818476292ad2c5ad86c45aa : SUCCESS in 6m 41s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/448f05a0b71441b29fa12c3cb32b77cb : SUCCESS in 20m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b932dc9faa2048b4815e4f81ee11ae96 : SUCCESS in 1h 04m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4516359eddd40d1b0ba3965abee207c : SUCCESS in 1h 48m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8e27ae59881e133b049fe84ad23644beedf8a060","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-20 07:42:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51736334d663746701f38ce8e0e788410599cf71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 07:43:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64a3c0ad64baed8ab9ed7d912edb1ab8cf9a8e2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:25:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13fb1f12a92d4691956ad17c0a431c23\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4bff9fa33a441bad14d012debf41a5 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a424e962d3047978ba6b1aeab243dc0 : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5cfb5c029c7434cb9f2d1a6dd13b16d : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d22f5ace8a0d4366aee89b1eed41f93a : SUCCESS in 2m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ceb791967dd47088c2017956b36e645 : SUCCESS in 2m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ca36c3825dff4766a5cfa677d57a50d1 : SUCCESS in 9m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d9848a96a7dd4cea9aced7c310e9d886 : SUCCESS in 11m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/53623bc8dc5b480b9653e5dacd58a255 : SUCCESS in 17m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ecceca405be841e5acb71f5b4249fe3a : SUCCESS in 8m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ccb271cd6bdf41c2ae235d7db05fc58f : SUCCESS in 12m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e9189325f7c84517bd01b93dc7544e48 : SUCCESS in 5m 23s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3f3c1c54a9174b0ebc96cdd2151975f6 : SUCCESS in 3m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a005bda56479422c95df7a8ed4f9d71b : SUCCESS in 8m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3ddef95dd4d44355963aee3c6c255328 : SUCCESS in 5m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c0b1dc76d834ecdba19942279859d48 : SUCCESS in 6m 47s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9e0e3c7964d04caca0cdd558c576a3b3 : SUCCESS in 8m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1f832bebf78b4777840d39aa554d6dc4 : SUCCESS in 2m 36s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d992dc89cabf43a5a6376ade6447e229 : SUCCESS in 9m 24s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fe47654e1d034deab9f482b8c90d3328 : SUCCESS in 14m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/62978590ffa84482910f6981457aea14 : SUCCESS in 7m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cde7f5de5b004e05a652cbf280779ec8 : SUCCESS in 8m 27s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c9620ebed0344e568dd8fcdd30f431ac : SUCCESS in 2m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f58c02534a7b4234a06db2cc4c8a0387 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f792487b6b9f41ca9507dff90a404de4 : SUCCESS in 20m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/932f9b50401c44c8b1c56362f1a78362 : SUCCESS in 4m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a9294ad45bab48d5b6d7902fb8fa23a1 : SUCCESS in 3m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a6daee537398413594d1ee9b4c1a43e0 : SUCCESS in 20m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1fb7853fb12429f96ea4a0d48a19d86 : SUCCESS in 1h 03m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be21330b758f48cbb04467fcbc0b88c7 : SUCCESS in 1h 40m 26s","accounts_in_message":[],"_revision_number":1},{"id":"88bf8ba94429c0acd10fbea07d337a122d99eb34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:25:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29314ba6f30cb991027a3dcfabbed8d800827ad7","revisions":{"29314ba6f30cb991027a3dcfabbed8d800827ad7":{"kind":"REWORK","_number":1,"created":"2026-05-19 13:43:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/989133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/989133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/989133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/989133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/989133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/989133/1"}}},"commit":{"parents":[{"commit":"c65c03328454f762ef0e173f77b1ed05e1d80937","subject":"update constraint for sushy to new release 5.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c65c03328454f762ef0e173f77b1ed05e1d80937"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-19 13:33:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-19 13:33:41.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 16.1.1","message":"update constraint for oslo.messaging to new release 16.1.1\n\nmeta: version: 16.1.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\nmeta: release:Commit: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\nmeta: release:Change-Id: Ia36a288a8f29ddfd098f708a5aecf5aa219bda80\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I575c54ea977578880010dcdfb4fbb14a9135b776\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29314ba6f30cb991027a3dcfabbed8d800827ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29314ba6f30cb991027a3dcfabbed8d800827ad7"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
