)]}'
{"id":"openstack%2Frequirements~989345","triplet_id":"openstack%2Frequirements~master~Ic0c7a20ca677e857b44f0f168c3c32bd8b44b576","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic0c7a20ca677e857b44f0f168c3c32bd8b44b576","subject":"update constraint for oslo.cache to new release 4.2.0","status":"MERGED","created":"2026-05-20 09:19:39.000000000","updated":"2026-05-22 15:48:11.000000000","submitted":"2026-05-22 15:47:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989345-new-release","meta_rev_id":"f15ce22b7d416bae1bbc79d28524f1d89d953026","_number":989345,"virtual_id_number":989345,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-22 15:47:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-20 09:50:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-20 09:50:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-20 09:50:31.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-20 11:23:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-22 11:53:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"dac1aa448c198ab438f03262e472f482d161c064","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-20 09:19:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4f1616eaccc7dbd3b0073534728deda9e7dd5f8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-20 09:50:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30835f86c3149c3f85f24f8b5b0a1fac8a6433cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 11:23:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b730013033a84f8e9a5e3b1e856729b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0961b3682f8640ee9d53b089dd8f3f48 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6737ce86a34a47898502f2cab9adbf87 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d083c960b894111b8d144fd0c3b0eb2 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68782ad3bd284b4abf664d4d13ff4d54 : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/53266cfc98c94dbbbacfa3348add5495 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d61aa47ca27f4d7fa33ae9f7166ce765 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93753464b8ae4816869f33d021248cf9 : SUCCESS in 3m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bcbc3d5fb80048ab89ba90ae55d0d10b : SUCCESS in 6m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5ce2c8ad17d24970b10fe637d1d9e8db : SUCCESS in 5m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2a948ae043f84873a8ff25a1e7768a3e : SUCCESS in 5m 40s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/4e1309e0da1a4ed38aaafa60ef341035 : SUCCESS in 4m 29s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/26e00d9c4e41462093b6d6783e4baea7 : SUCCESS in 7m 04s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/9ea6be78f85941bdaeabd4be523b9cd1 : SUCCESS in 6m 10s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/82fd62fc12474c8c964dbcbf97bc8f76 : SUCCESS in 11m 23s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/479752567a384c7fa0fbfd6f040eba43 : SUCCESS in 3m 19s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/0a3bdbaef88e4efcbf741884f48b1b03 : SUCCESS in 7m 42s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ed39cc84e9e84ed18fe7e05fa035d075 : SUCCESS in 7m 11s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/e47e70ba05684382a2426150d7fb396c : SUCCESS in 15m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/824916a091df42f88ddc7e0194dd810f : SUCCESS in 6m 53s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/76f4f13d12124218889c897091829fd4 : SUCCESS in 4m 45s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/938068d5a0e24b828aaeb23e958ac33e : SUCCESS in 11m 10s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e122062f29f34ada86234cd13ce93bd0 : SUCCESS in 2m 53s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/327de082306749888edb3ab4573a41d6 : SUCCESS in 8m 00s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/b1425b089c494458b8c11936da5afb1f : SUCCESS in 4m 14s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/26bf456fb69141d58dfbf379fdc1b794 : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/4a0001c14db64ad5b327d415af905081 : SUCCESS in 10m 18s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/ac5ad0d5bb6c43c0b9a97fef34aad655 : SUCCESS in 25m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/75f022b8d8e0440d969a47ae5af35c6f : SUCCESS in 6m 17s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0031731b348f4085a924360e2c0270a1 : SUCCESS in 14m 02s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/5af3c7eb68514c9e9af8c60d6877b3ab : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d51d91f2d9a94a1c82905ecc22344214 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e327044701d241d08fdb611caeefbfe6 : SUCCESS in 25m 57s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/636e4c928f9348b69d0bd52a85c6ec39 : SUCCESS in 3m 13s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/27807a408c8647c1ab6a5eb7f36de092 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d7a6a9dd39b481dad21a987ab69a1ad : SUCCESS in 6m 14s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/2992d13563594e4dbf162b84b813921a : SUCCESS in 18m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/db4e3e2327d8497aa187af1f848b8fea : SUCCESS in 3m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab926d8c1b4e47ba934164a58b595619 : SUCCESS in 37m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8db5332671b244e2be22ac8d0ca68733 : SUCCESS in 1h 55m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6ac31768374649cc0fb541d0f5af6a92cf80bbe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 11:24:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e4d7dc79d51f2d283a36398ebede0f8138875b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 13:22:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/465dada84bef40a9b5e945d583c837b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004ac136657640bb90bb9e7f7e0c0ade : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9127b68be9534dcaaa8d8ab6930b2014 : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0688c3819b724c3183175c62f2769e06 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69bdb9460e074373b6c2bef3b6c38086 : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6add439a3aad416784f2a7d2c4748d79 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7168cd2452234b75a708d322b1ce3807 : SUCCESS in 3m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/656099ea28a445238d08d75f52ad25b0 : SUCCESS in 4m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5dc091c9341a4b0eafd90bd0b4b60ccb : SUCCESS in 3m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2f3279bfd2e04c759ed3338ff8d1d662 : SUCCESS in 9m 57s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/5b730cda59d34c819704fa7d6edbf026 : SUCCESS in 11m 29s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/62a9f38e25584033835a5959f863fe0b : SUCCESS in 5m 03s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/5a08b61750434e98bee616b9730ff3f1 : SUCCESS in 5m 06s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e1fcc1d636364ce8a137dcaca505a5a5 : SUCCESS in 6m 18s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/4b62d3c599d84540851d62566f8d4bb2 : SUCCESS in 17m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a67a41d678b141a5a9dad9ad2abb2418 : SUCCESS in 6m 40s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/aa5dd37dc9ac427285b2c0ea77afaad0 : SUCCESS in 8m 05s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/2e652bdb5d1940a6b64900d0120bcd7b : SUCCESS in 11m 49s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/ac74f24f62b34312888a4c28caf43f42 : SUCCESS in 2m 36s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/a6a0f60b786e44a9a9c0b4cb0b72a3d0 : SUCCESS in 4m 52s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/3ee8cbb9c28e478eac131d8eafb7fc9a : SUCCESS in 16m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dc92afa274d4452b8b18c943e361deaf : SUCCESS in 3m 57s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/e1fb78076eb74a3abb9315ba9d842696 : SUCCESS in 13m 42s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/d4e272e606dc48de8037a7590aa3f89a : SUCCESS in 2m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b9c81930be584254b06db392a41ebeae : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/010834bfaee343df8502f3be9df53e6c : SUCCESS in 23m 59s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/49c223f9a0d340bcbe5dd3e14e96982d : SUCCESS in 6m 25s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/fb056f8c705746698d36342616551b03 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f4e509064124f25a7686d165ad3547b : SUCCESS in 8m 06s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/aabb69c91e5b47e480d497949f6af256 : SUCCESS in 14m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1f78ce74eb244390aa4bc1d5b0a4a5b6 : FAILURE in 5m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2832eed605ec4c4890bdec05cce8c1d9 : SUCCESS in 1h 10m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58f27c72a98d4e16a3e2d07e939229f5 : SUCCESS in 1h 53m 40s","accounts_in_message":[],"_revision_number":1},{"id":"825ffe7b9ecf47510f59757066fd976f6e490e72","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-22 11:53:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f030930ecf6327377b4351e02b9f842d756e8b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 13:45:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a2aad2dee7447d794340decb048bc30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a287363d1e40ebb878b2e13a6f3a01 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93ca24d12627417c8a8943eb446c7597 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6603645e8e3f4db9ac4779a6e08cb70b : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/504502a9d99f4e5ab2983991208493f4 : SUCCESS in 2m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b8017cbaf5f84df38dc1412b60cab815 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe70af6e0bca4e8d9cc012560cf8541a : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d3e28a2757f54a67b84645cfcdbfc51b : SUCCESS in 3m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6109b53f75494bd5831f73d3b082031c : SUCCESS in 2m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1942dabb94c84619bd58ade7490b3044 : SUCCESS in 3m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c521882ab7e74309aaa43dcb67fa0580 : SUCCESS in 2m 37s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/9948161421cc44a2aa152f4ffe521d84 : SUCCESS in 3m 01s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/2e525aede8094ef7ac79912bd563f244 : SUCCESS in 3m 24s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/a7d94645b21b41e7bf6b6dee14b78810 : SUCCESS in 4m 12s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/160d532b2ab7482493ba366f255ef872 : SUCCESS in 7m 15s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/5650f7801c3e48fc996d02551e3b39a0 : SUCCESS in 4m 26s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/497393fe6f374c26ba3fb2b236a2528d : SUCCESS in 3m 33s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/5869a42982cd4fd3a04860513e2f77df : SUCCESS in 6m 28s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/05210782450f4c59852b603d935a4c8c : SUCCESS in 16m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/93ad5ad73db44a01ae6fdd9f27aa3438 : SUCCESS in 8m 06s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/2b08e61914b345a4841901ae5c58c30f : SUCCESS in 8m 32s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/cb5b2536d19242d1881e44c392845a54 : SUCCESS in 11m 01s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/48af3b391d664a9b912c8dacf4e244c6 : SUCCESS in 2m 15s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/7f8f34f01ff54381ba8c8d8467f0d8b8 : SUCCESS in 7m 54s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/fa36dfe7471b417cafb71967628d647e : SUCCESS in 3m 52s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/f5a589a1ba704cf6b658567110d30ee5 : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/29b6d5faff03410ea4ec79f9e961f308 : SUCCESS in 10m 15s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/490c0f0633704ce3951839f92096efd0 : SUCCESS in 25m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/305b4b0765e441768e5d90491a8c0f2b : SUCCESS in 6m 04s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/41869ed74519487580490296317cb522 : SUCCESS in 13m 33s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/ed9ccb5e2b1f4c7681bab038537293d8 : SUCCESS in 4m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9ee743ace8224ef698a80f1c6240ace2 : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/39343f974daa48a68f6bb25d74109b38 : SUCCESS in 23m 59s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/df102a19a0c34625ae62327c281097bb : SUCCESS in 5m 48s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/bb92cad9a2e048bbb36c347d2edb4826 : SUCCESS in 4m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b0b9c25d7ea84de4832624d504a5fa4b : SUCCESS in 6m 24s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d5fd5f74131b4ad3accab4938c0e4f22 : SUCCESS in 18m 38s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/86d552712eda4b1babfe1ebf51e4d16f : SUCCESS in 3m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ceed860f1f54d399ef5a6ee45b95e3f : SUCCESS in 1h 13m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5eefc385c100494cb6095071aa86a96b : SUCCESS in 1h 45m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1241358fc8d91e5979ce97c68138d998bf21b989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 13:45:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6206f44ba788d820fbd58d46dda810b4a69cb3aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 15:47:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad057d07a82d4ea4bd5f157def329f16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42199b2f53f24780837c0a31ddb716df : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c33d039b6a65445693f7c848f45d873b : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff7b0ec80f9247b285ec7e5834d21457 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b87eb47816247aba370af060ffd2281 : SUCCESS in 2m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/921c9b8c8d1946b4818c59f336052096 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8cedc302a2444774a0e31e9b582dcafc : SUCCESS in 2m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5ecdf965757d4d7180ff720e44c5e78b : SUCCESS in 3m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9e790f6f98c24f959879578cd90ff8ba : SUCCESS in 5m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/21c04abce36c4d2cbfe98561f5f57980 : SUCCESS in 3m 55s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/c05321bad0444237b84aa19a3ea59a4f : SUCCESS in 7m 10s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/f1dec28420f6476084d9cadbe2f0603a : SUCCESS in 4m 45s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/1f74a7f05ba94166b3bf4487552f345a : SUCCESS in 4m 34s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/75aff64b35f94dadb10ac4552a99d3ec : SUCCESS in 6m 04s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/cefef54d514243a7bee90dd497a8e744 : SUCCESS in 8m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/665e0aa391494824ac9601ee576a0508 : SUCCESS in 6m 06s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/e8927c2719c74212a9ae0456cee7861c : SUCCESS in 6m 42s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/f9034032856b4efd8f14f4db20b7cb98 : SUCCESS in 10m 45s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/659d503599eb42a8af54f2310cc96a34 : SUCCESS in 2m 03s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/45126f3e0b4c428aaf1981b2ddb67362 : SUCCESS in 4m 28s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/cbf8c561507d4397864ab7c4431f5e8d : SUCCESS in 22m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cf239dcbaecb4b45bf9f21ae797e0c41 : SUCCESS in 3m 32s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/5fc123530f174007a650e3ab97314c84 : SUCCESS in 8m 13s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c309bc0a927a4152b0701da57071add1 : SUCCESS in 3m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fefd064287e44f18b613cccb57a266aa : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd13beefc69f424fa1cfe15c3ff75916 : SUCCESS in 23m 54s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/97c4fd5e3e9c4ab487777ab057310bee : SUCCESS in 4m 33s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/32d75764b08548e5a358717451314137 : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39fd6aa87e17458d9501d7d2d2bd5772 : SUCCESS in 6m 02s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/df437d0768054551a397e387d308182f : SUCCESS in 16m 26s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0c4e18065c184faf9f2f6112ba414e03 : SUCCESS in 5m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c989a09909f045428be80044f030d0c8 : SUCCESS in 36m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6af17243fce49caadb626a48044d14f : SUCCESS in 1h 55m 31s","accounts_in_message":[],"_revision_number":1},{"id":"11a9829d5d265d2c1699f67e300ca3aae4d436fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 15:47:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f15ce22b7d416bae1bbc79d28524f1d89d953026","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 15:48:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b2f82a027d499694d08f1f6dbd268d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac66e6d982d40b2a58811abe63b9f2c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"926e10e80b62d51e9d83106ef23d6a0fa2fadcb3","revisions":{"926e10e80b62d51e9d83106ef23d6a0fa2fadcb3":{"kind":"REWORK","_number":1,"created":"2026-05-20 09:19:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/989345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/989345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/989345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/989345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/989345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/989345/1"}}},"commit":{"parents":[{"commit":"f8a07e962c7b0b2f1839761042d3a35ae3c51c43","subject":"Merge \"update constraint for oslo.limit to new release 2.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8a07e962c7b0b2f1839761042d3a35ae3c51c43"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-20 09:09:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-20 09:09:37.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 4.2.0","message":"update constraint for oslo.cache to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I638372414d9beb17262b160ad32a81cd5cb6976f\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: Ic0c7a20ca677e857b44f0f168c3c32bd8b44b576\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/926e10e80b62d51e9d83106ef23d6a0fa2fadcb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/926e10e80b62d51e9d83106ef23d6a0fa2fadcb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
