)]}'
{"id":"openstack%2Frequirements~989489","triplet_id":"openstack%2Frequirements~master~Id2bd0f8c40da43126c881d03cca72c667480ccd4","project":"openstack/requirements","branch":"master","attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2026-05-21 10:40:51.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-05-25 17:37:58.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-05-21 10:40:51.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-06-05 06:26:02.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-05-25 17:37:58.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Id2bd0f8c40da43126c881d03cca72c667480ccd4","subject":"Pin keystonemiddleware to \u003c13.0.0","status":"NEW","created":"2026-05-21 08:33:00.000000000","updated":"2026-06-05 06:26:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c2603201ab66a288b90c5c25210ab4c568644661","_number":989489,"virtual_id_number":989489,"owner":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-21 14:47:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":-1,"date":"2026-05-25 17:37:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-21 10:05:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-21 10:39:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-21 10:40:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2026-05-21 10:40:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-05-21 12:50:09.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"8deec10ea16c089b54bf8c1abe9d70705f922cf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-21 08:33:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db8ccdc20d629b94b29e4edbd54618e4c0befa42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 10:05:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e06d0ee197c46e596d2ef635b218164\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d68cf87f0614b809435fc73df9f16b8 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d3364f5ac534bd0a8c17eca277f1e54 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3fe8ecfdb9fb47f4843ace9b2a570694 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0216e5e6ec540be8efe408b827fb2d5 : SUCCESS in 5m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f68904be0a544310b7ac7e13278fe260 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c9669ad551b4812a92f98cf57c81946 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b863fdb2a4b442284bbc9c0fccd7b94 : SUCCESS in 2m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3a0470211d40407baeaf7841e992cb34 : SUCCESS in 10m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/210d29144ea245efb2088a2dd666232c : SUCCESS in 10m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/84fbb54daad64d95b183e4cb490e0001 : SUCCESS in 10m 20s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/f8c90ab3eaa54ce49cdd1205d4f3e511 : SUCCESS in 4m 17s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/f2e3487cbe3443d683b227088f8339f8 : SUCCESS in 4m 55s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/e0b250ae140b40ceb9a6c04391a07b2e : SUCCESS in 4m 55s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/6de4e4d01b9046f582762b7bc8bd9418 : SUCCESS in 10m 55s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/0bd9784f4e7d4ec9a4f7852dc9be3cc2 : SUCCESS in 5m 04s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/b68c496c73364ae88615ca3c8e74db18 : SUCCESS in 6m 03s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/a654be0080e04055b60b67b15f14cf35 : SUCCESS in 6m 00s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/7ef96033a2f34163b1b8555c1679b944 : SUCCESS in 19m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2d53be5ca2c468f8920bccdff1fcc40 : SUCCESS in 6m 43s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/cbc1209c40ac49779247472c7f2c328e : SUCCESS in 7m 20s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/91189867f0a34ca8ae0280f4f7f7d5d6 : SUCCESS in 11m 36s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/b71a990208b54f6bb1547fc3ce8673f7 : SUCCESS in 2m 50s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/255320dcafcd4a96ba23915fb6267e1e : SUCCESS in 9m 35s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/8bffada1399d40908141964eed7d1059 : SUCCESS in 4m 18s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/3b0d996da84141f4934b0e5647196bcc : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/ab579cc77ebf4bd1bb04981bd206ea0f : SUCCESS in 11m 22s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/f3c3772dc0414d01ae9c2efee61f53ae : SUCCESS in 23m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f0310e43e0544f389ad3f0f3b723ea7d : SUCCESS in 8m 51s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/f238f36a8da74825a4d439e90668da6a : SUCCESS in 12m 28s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/d1992df77da74e09b068f773878af5b4 : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99dbe86b0fab4b31917ad5863045353a : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9d0ee41605146638e6387aa4e62a381 : SUCCESS in 26m 19s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/c847b9da01494ff982698fba7d172893 : SUCCESS in 5m 21s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/6da76293d17e4360833febb935f69a20 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b099b06dddc47678e764ecac38bb2b3 : SUCCESS in 6m 05s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/ccb5771f36944de2949abaf9e6666c50 : SUCCESS in 18m 26s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c71b4cfdf15b4309a42a452f7ebd12b3 : SUCCESS in 5m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cc219186c5d4c7d9b078220da1f51fb : SUCCESS in 1h 02m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e28d988a7dda4c319756f74c0c60d541 : SUCCESS in 1h 29m 18s","accounts_in_message":[],"_revision_number":1},{"id":"9d0db7ebec49d93f416e6c484e35a810f5642848","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-21 10:39:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9719b9a735110f0ec46fbdcb8d9e033e6bd6ffeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-21 12:45:09.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7905639663e63d5b812d219b0cc574787c1e3b70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-21 12:46:27.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"43404daf7cefb964cbbb9f7a5aff36ca416e44ca","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-21 12:50:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8c8d97e2ab1c371a1978525eae5592c302040db7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 14:47:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b4124f1baad45e88e8c70645eaed81c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a79c45f1ed11491fb41304ad7c6b2c8d : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cbe5d72ada24f6a9e542605304410e3 : SUCCESS in 2m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/067df33fd9904d8d93929da23dbcc63c : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55ba8ce908624870b7e6becc14cca38d : SUCCESS in 2m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/12cac57490fd45a7ad43371f0b1e9a24 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85ea615806f2431c80334e5d8fcc78cc : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ac0c51b98694ffcbaa0810d6447d8d5 : SUCCESS in 2m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8ebe1844e7794a758698eed0e96ae843 : SUCCESS in 11m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cba0686bad1647f1b1cb4884b9e47f75 : SUCCESS in 3m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/62b1bcc69e214566833fce5766027e47 : SUCCESS in 6m 02s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/317beebab6b24ffba167a923dfceff1e : SUCCESS in 3m 24s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/158382f5074b43faa36e8c090be47fce : SUCCESS in 4m 44s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8b25972a344c4e3d96c855fabb4c8767 : SUCCESS in 4m 00s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d8d894112ca94628b59cfe160541b6b2 : SUCCESS in 6m 57s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/8627b6e180c0454db0a64e73ab0bb36c : SUCCESS in 3m 01s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d5d013da643e477e9b3c8beb15fbbed4 : SUCCESS in 2m 50s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ada99ac6e8164977adc72af1f2d84f32 : SUCCESS in 6m 17s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/89be2cb128d84d029a009d7f40ad5b46 : SUCCESS in 17m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a88137f1d77646cca24bee5105632016 : SUCCESS in 6m 28s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/fd559fd4a5294c7ba4fd436f30f6586b : SUCCESS in 4m 50s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3a96148e469b47eea443ba09945dd98e : SUCCESS in 11m 34s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/575fb7d84973473697f3690787c43983 : SUCCESS in 3m 58s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/fb9fc1eca8e044c0bdb613c9b1ac543a : SUCCESS in 8m 17s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/17caf736ad1f4e649a67d0e181283b5a : SUCCESS in 3m 59s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/1aa7e0db9b1947b69214dc9c3166d840 : SUCCESS in 4m 08s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/b55d67106cb545c3a442ab1b6fb02ef1 : SUCCESS in 9m 27s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4f8c539ff9eb4bafac079d0bd1ae9c56 : SUCCESS in 22m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/36b1fd9832dd473fbcdacb1c21212d49 : SUCCESS in 6m 08s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/2a493c36e2d4402bb9b71f10d2364fa5 : SUCCESS in 9m 34s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/aa4116763d1d4a01bfb7ce8096d5d110 : SUCCESS in 2m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e8f9548d76fe46fa9e545b59757a01b0 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f53ec15500b547fb8def00992c7aff91 : SUCCESS in 23m 58s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4f38737f864a4495abbbef0416914316 : SUCCESS in 3m 52s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/dddd9032c548472abad8a74802048031 : SUCCESS in 3m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0705aecd38b14853bc13763f26ce4f6c : SUCCESS in 7m 33s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a15d7fe89445452b87467923b0bcb7f8 : SUCCESS in 13m 59s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/235166a98ef547439c817f88ba96b0ba : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71dd66f958e64664989220d26f9a5f84 : SUCCESS in 1h 03m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99c1f4b7f87a468badb3b1e4ad8dbe8e : SUCCESS in 1h 54m 47s","accounts_in_message":[],"_revision_number":3},{"id":"47c2a2bc2b7bacca735e4c00d192e649939ee8ac","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-25 17:37:58.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"acfbf9b971ee8279774eac71630fc0b05eb914e5","revisions":{"36b17a31935290a850d531ce4338a62fac18dd80":{"kind":"REWORK","_number":1,"created":"2026-05-21 08:33:00.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/89/989489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/989489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/989489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/989489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/989489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/989489/1"}}},"commit":{"parents":[{"commit":"2bd07c0208ec11b29bc2df2412b211cd2cb3df88","subject":"Merge \"update constraint for oslo.messaging to new release 18.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bd07c0208ec11b29bc2df2412b211cd2cb3df88"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-21 08:24:18.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-21 08:32:56.000000000","tz":330},"subject":"Pin keystonemiddleware to \u003c13.0.0","message":"Pin keystonemiddleware to \u003c13.0.0\n\nkeystonemiddleware release 13 changed service_token_roles_required\u003dTrue\nas default, exposed a bug where valid tokens lacking service roles\nreturn 401 instead of 403.\nthis cause regression and breaks openstack projects, noticed in tempest\ntest_old_versions_reject which expects 403/409 status for now.\n\nproper fix should be made in keystonemiddleware itself to return the\ncorrect HTTP status code:\n401 - authentication failure (invalid token)\n403 - authorization failure (valid token, lacks permission)\n\nthe bug is in keystonemiddleware where all service token return 401\nwithout checking correct auth.\nsince 13 is just released and next release with fix may take time this\ntemp pins all to use keystonemiddleware\u003d\u003d12 to avoid this CI bug.\n\nthis should be removed/fixed once keystonemiddleware releases a version\nwith proper fix.\n\nthere is also a workaround, accepting 401 in tempest test [1], but that\nmasks the http bug, this pin maintins correct behaviour until\nkeystonemiddleware is fixed.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/938766\n\nRelated-Bug: #2153382\nChange-Id: Id2bd0f8c40da43126c881d03cca72c667480ccd4\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36b17a31935290a850d531ce4338a62fac18dd80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36b17a31935290a850d531ce4338a62fac18dd80"}]},"branch":"refs/heads/master"},"5bb9bff6d28b417949b0b5b5f21ea0fcaf90b6f4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-21 12:45:09.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/89/989489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/989489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/989489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/989489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/989489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/989489/2"}}},"commit":{"parents":[{"commit":"2bd07c0208ec11b29bc2df2412b211cd2cb3df88","subject":"Merge \"update constraint for oslo.messaging to new release 18.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bd07c0208ec11b29bc2df2412b211cd2cb3df88"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-21 08:24:18.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-21 12:43:45.000000000","tz":330},"subject":"Pin keystonemiddleware to \u003c13.0.0","message":"Pin keystonemiddleware to \u003c13.0.0\n\nkeystonemiddleware release 13 changed service_token_roles_required\u003dTrue\nas default, exposed a bug where valid tokens lacking service roles\nreturn 401 instead of 403.\nthis cause regression and breaks openstack projects, noticed in tempest\ntest_old_versions_reject which expects 403/409 status for now.\n\nproper fix should be made in keystonemiddleware itself to return the\ncorrect HTTP status code:\n401 - authentication failure (invalid token)\n403 - authorization failure (valid token, lacks permission)\n\nthe bug[2] is in keystonemiddleware where all service token return 401\nwithout checking correct auth.\nsince 13 is just released and next release with fix may take time this\ntemp pins all to use keystonemiddleware\u003d\u003d12 to avoid this CI bug.\n\nthis should be removed/fixed once keystonemiddleware releases a version\nwith proper fix.\n\nthere is also a workaround, accepting 401 in tempest test [1], but that\nmasks the http bug, this pin maintins correct behaviour until\nkeystonemiddleware is fixed.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/938766\n[2] https://bugs.launchpad.net/keystonemiddleware/+bug/2153561\n\nRelated-Bug: #2153382\nChange-Id: Id2bd0f8c40da43126c881d03cca72c667480ccd4\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bb9bff6d28b417949b0b5b5f21ea0fcaf90b6f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bb9bff6d28b417949b0b5b5f21ea0fcaf90b6f4"}]},"branch":"refs/heads/master"},"acfbf9b971ee8279774eac71630fc0b05eb914e5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-05-21 12:46:27.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/89/989489/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/989489/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/989489/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/989489/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/989489/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/989489/3"}}},"commit":{"parents":[{"commit":"2bd07c0208ec11b29bc2df2412b211cd2cb3df88","subject":"Merge \"update constraint for oslo.messaging to new release 18.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bd07c0208ec11b29bc2df2412b211cd2cb3df88"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-21 08:24:18.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-21 12:45:29.000000000","tz":330},"subject":"Pin keystonemiddleware to \u003c13.0.0","message":"Pin keystonemiddleware to \u003c13.0.0\n\nkeystonemiddleware release 13 changed service_token_roles_required\u003dTrue\nas default, exposed a bug where valid tokens lacking service roles\nreturn 401 instead of 403.\nthis cause regression and breaks openstack projects, noticed in tempest\ntest_old_versions_reject which expects 403/409 status for now.\n\nproper fix should be made in keystonemiddleware itself to return the\ncorrect HTTP status code:\n401 - authentication failure (invalid token)\n403 - authorization failure (valid token, lacks permission)\n\nthe bug[2] is in keystonemiddleware where all service token return 401\nwithout checking correct auth.\nsince 13 is just released and next release with fix may take time this\ntemp pins all to use keystonemiddleware\u003d\u003d12 to avoid this CI bug.\n\nthis should be removed/fixed once keystonemiddleware releases a version\nwith proper fix.\n\nthere is also a workaround, accepting 401 in tempest test [1], but that\nmasks the http bug, this pin maintins correct behaviour until\nkeystonemiddleware is fixed.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/938766\n[2] https://bugs.launchpad.net/keystonemiddleware/+bug/2153561\n\nRelated-Bug: #2153382\nRelated-Bug: #2153561\nChange-Id: Id2bd0f8c40da43126c881d03cca72c667480ccd4\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acfbf9b971ee8279774eac71630fc0b05eb914e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acfbf9b971ee8279774eac71630fc0b05eb914e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
