)]}'
{"id":"openstack%2Frequirements~989765","triplet_id":"openstack%2Frequirements~master~Ieb19c71a2f1461e6925a1d6de889b2a18f6c93a6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ieb19c71a2f1461e6925a1d6de889b2a18f6c93a6","subject":"update constraint for glance_store to new release 5.5.0","status":"MERGED","created":"2026-05-22 13:00:49.000000000","updated":"2026-05-23 09:23:55.000000000","submitted":"2026-05-23 09:22:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989765-new-release","meta_rev_id":"9fe71cdcaa3476956cae7404292853fb70d1f95e","_number":989765,"virtual_id_number":989765,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-23 09:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-23 08:03:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-05-23 08:03:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-22 14:46:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-23 08:03:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"46672a7d3cd9c9b25e5fd12cc576ae0356917acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-22 13:00:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9144f1fcfd91f4871503ec1398959f56c503f089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 14:46:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc49b2045794415d8a75821e20d2cdcc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2692cfa8111849f9be88f96f31d936e1 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70f10798ecf64413a6d9b0b2307d95bc : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a26490fa350c436d8b8bad8d44f1ec4a : SUCCESS in 2m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/068a2a2d1efc4d39ad95bf1ee4dedc71 : SUCCESS in 5m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4f1ad259116c4646a83646b1c14c9630 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/452e95e397a848958e8d6fd05b89a0de : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa447d25912a40fc9973a57420189108 : SUCCESS in 3m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ff509d1732be463aa2fa5c1f83062fea : SUCCESS in 9m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/85f400b7bab549369c86b016defee0f0 : SUCCESS in 5m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/090daded5ed54e6ca1c4c2eb6f0a0ce0 : SUCCESS in 5m 07s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/580dc4d8656e4c3b9623ce36da1507f9 : SUCCESS in 3m 23s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/d201dff4897a49568d6282c888572bb0 : SUCCESS in 4m 41s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/0a16764bd1494615a7ee0c68ac744e9d : SUCCESS in 4m 27s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/374948c2681142c08d9b6bdcf306e80f : SUCCESS in 10m 12s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/263348a0208146fc99b18ab4370d14cd : SUCCESS in 5m 25s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d9bab90525cf42eda6bc4e63fc2eb2dc : SUCCESS in 3m 38s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/b90cdbef110f431785a8c697ca0c0fc1 : SUCCESS in 5m 56s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/8467fa613fc749099c8f5d96d9a911f5 : SUCCESS in 17m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e009e26361d40c89ab1bd30eb4e5e0b : SUCCESS in 6m 09s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/34155f50bbf845afbfc555d96143b53e : SUCCESS in 7m 16s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3ddec4ddf0d641ffa11eadb6f5c6e0d9 : SUCCESS in 8m 54s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/93664324ed94427f9695e85765f6ea90 : SUCCESS in 2m 16s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/4788f40818284c4990467130e429761b : SUCCESS in 8m 13s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/d93cf751032c495ea78c5018cf42cbd4 : SUCCESS in 4m 40s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/d191dea9d4d7481583e2ef2d848d361c : SUCCESS in 2m 49s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/759761b918f94d3f84e7f5c1247addf6 : SUCCESS in 10m 22s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/a3268aa2a17d4e21813b93d26a892990 : SUCCESS in 17m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5a8bde20f98d4e4f93480be56ff843a7 : SUCCESS in 3m 33s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/e29f578f117f4f50a1a3370b8748860b : SUCCESS in 13m 07s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/f306a1b32ee949bc96fda8c00f2b6bc3 : SUCCESS in 6m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d31acd5b6b924a41be87c16a1bdf6d27 : SUCCESS in 2m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a0003470aadc4a7784162e2dbc2c7e51 : SUCCESS in 26m 32s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/3c10ed1553c647bf8decdee6790ff896 : SUCCESS in 4m 34s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e8c580b00cd14af3ac2762660fd95c50 : SUCCESS in 3m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3df33176c084921831eac1d7f64862e : SUCCESS in 6m 54s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/c961d978a4fb4c43a9d064c7dc8c0f03 : SUCCESS in 8m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4093fa95619a4a73b7b7df74cf6ce4a3 : SUCCESS in 3m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cb5425cceaf4709bbf54f190a16486d : SUCCESS in 54m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28346a9704c740b9beb5b39a6d0210f7 : SUCCESS in 1h 42m 37s","accounts_in_message":[],"_revision_number":1},{"id":"44cc48905cbc84374dc2a29fd7992686a0a4c110","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-23 08:03:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"874989cb2cde2e559e91684a35c75c00b4c146c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 08:04:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a010046e91ae4f736fae7dd13139a7afcf0cc939","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 09:22:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db00d29085c44db5a806788dc53dc076\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9709f2e566f4c968147bc06546e0f37 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19c8f89c1e7f46adb2914962bbeca7f2 : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44461e5d62bd4f4ca6360a150ab6d6cc : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/faf453a80e844ac5acbdedde89e1dcb2 : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4dffcb6bb8694b55817e8c306cb36200 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf0ab82137d74cb1a5adaca27a0c602d : SUCCESS in 3m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2999d204e2e24cb1907b58300cb34487 : SUCCESS in 3m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fdfa78425f0c47c48a6198bf49c15f9e : SUCCESS in 6m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d33654d70dd245468d354dd965fcde3c : SUCCESS in 3m 34s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/65826ef55e51479a959c66df7c8e8742 : SUCCESS in 6m 46s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/931eb018434d4abebfcf2c095a2eef6c : SUCCESS in 3m 21s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/b1a5d06535e746e7b0ffaca3b4c4d790 : SUCCESS in 4m 43s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/7b98f2499d7d42aeb4af1f70076d840e : SUCCESS in 6m 37s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5aec5c8015b643a99f372c1821518c12 : SUCCESS in 8m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/257d5500bbf24093a0f57bbfe98ff23d : SUCCESS in 7m 04s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/004b912f434e41118148fdb611e75ced : SUCCESS in 5m 37s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/057745bcabdc403f9cba4debe71a0bc8 : SUCCESS in 6m 59s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/113284af692c4b448c2bd09c21fa3b59 : SUCCESS in 3m 32s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d668d760e3bd42b6be877044fa5b82c0 : SUCCESS in 7m 58s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/7abdc0ec7eff4d609c44bd4a9d519f9e : SUCCESS in 24m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e235e32bed1e4defb3a93cf925d7b8ad : SUCCESS in 4m 49s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/8f302d59370d40d084e7f776e1ba29ac : SUCCESS in 11m 16s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/afcf10dbc9ee4abf903677eee8f4ed5d : SUCCESS in 3m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/14fd5b86bd4c40cfb0a268da15add9b7 : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e12e53c931a143d38284229314e33da9 : SUCCESS in 25m 51s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/d34da200c49f4b9d97515c4751ca7b0c : SUCCESS in 3m 41s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/4c8dabd9e6ff428e9790634e63fa11d3 : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d58eb647b6914c59b94bb7b0e6592c6b : SUCCESS in 6m 52s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/7b60175ccd98400a95c2b9fb564c1160 : SUCCESS in 9m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e3e7f571220a43cd83fad0a33f9c55b1 : SUCCESS in 3m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a42094099b44f1bbfa9071d3c24c780 : SUCCESS in 51m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72160580f46642fcba65b31c7e97ef04 : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":1},{"id":"79b302b3fe1336d9fbab0a774a9dec71faceea59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 09:22:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9fe71cdcaa3476956cae7404292853fb70d1f95e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 09:23:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e916bf14b804819bd40dded551381af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad09afb1c9042eeafcb4b82163ba2a9 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c8af58125d5ae718480551bd4e8efdc5b561ef9","revisions":{"3c8af58125d5ae718480551bd4e8efdc5b561ef9":{"kind":"REWORK","_number":1,"created":"2026-05-22 13:00:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/989765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/989765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/989765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/989765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/989765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/989765/1"}}},"commit":{"parents":[{"commit":"2bd07c0208ec11b29bc2df2412b211cd2cb3df88","subject":"Merge \"update constraint for oslo.messaging to new release 18.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bd07c0208ec11b29bc2df2412b211cd2cb3df88"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-22 12:50:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-22 12:50:48.000000000","tz":0},"subject":"update constraint for glance_store to new release 5.5.0","message":"update constraint for glance_store to new release 5.5.0\n\nmeta: version: 5.5.0\nmeta: team: glance\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iaaf3bdf5208f5314656493358271f7d893425a80\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ieb19c71a2f1461e6925a1d6de889b2a18f6c93a6\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c8af58125d5ae718480551bd4e8efdc5b561ef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c8af58125d5ae718480551bd4e8efdc5b561ef9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
