)]}'
{"id":"openstack%2Frequirements~990790","triplet_id":"openstack%2Frequirements~master~I3d592e78023764d11b8656b10d231c81077399e6","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-06-01 18:33:56.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I3d592e78023764d11b8656b10d231c81077399e6","subject":"Updated from generate-constraints","status":"MERGED","created":"2026-05-30 08:10:03.000000000","updated":"2026-06-01 20:35:24.000000000","submitted":"2026-06-01 20:34:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990790-openstack/requirements/constraints","meta_rev_id":"7b28e15fd384c32787aa41f1f968df9a553f8d00","_number":990790,"virtual_id_number":990790,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-01 20:34:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-06-01 18:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-06-01 08:17:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-06-01 18:33:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-30 09:20:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 08:17:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-06-01 18:33:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3ad3c791e00d94c90df3c9e5591f4ac9a1efcff0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-30 08:10:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38401ce403c897c3622fc252b42d63ff98da221d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 09:20:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ba73d24061a4bf598e66bf5601e5dab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32099875c5594d0c91f94a37b234ef5c : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5237ed8677ea42859806b1d989a28975 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f73e6a5a0f4346a7b390c63fc0940995 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb16c73729d34ed9a73dc8083099d6e1 : SUCCESS in 2m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9287959474fd4c528a60fda9343b15d3 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d82e5413c4eb40528edb01ae1b4de0a2 : SUCCESS in 2m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/168d405270fd41ec8b15382b74368254 : SUCCESS in 2m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/77d24aea82a641d1bc7bc29138ae555c : SUCCESS in 3m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5bffe6e28fce437c8103a45d81237f6d : SUCCESS in 3m 24s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b537cadee93b4f948114810c63f22a9b : SUCCESS in 3m 42s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/29f3156c39004184a026904194317176 : SUCCESS in 3m 00s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/3bd9f3511ccf4ed7a1d0776a002c9206 : SUCCESS in 4m 28s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8db51e4c43ed468bafd27cffffe6a0e4 : SUCCESS in 5m 40s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/4e2d7a6504fe4c8e9563c8d4dd67c44c : SUCCESS in 10m 47s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/1a8881cd87a64a88b14d22a65fa19c18 : SUCCESS in 4m 20s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/0d6174a803644d50b41e9e025674c217 : SUCCESS in 6m 34s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e249f1ceb7544dcb8a6eafa006c569f4 : SUCCESS in 6m 36s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/e1b97730ffbd49db833be6a9ae6d64bd : SUCCESS in 15m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74da8253245d4d578dfc498d6cfdc436 : SUCCESS in 6m 39s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/4e77ea328cca439ba60bc26384568700 : SUCCESS in 4m 32s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/ab2d48c5311f49549c0bd7bab45f9d62 : SUCCESS in 11m 06s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c3576dbca7bd412196a1d0ed537f5c19 : SUCCESS in 2m 41s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/a29156f1ad8c4747836a5d26a94a7e95 : SUCCESS in 8m 15s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/56d7d2d1a4c44ba896bc05b32af63ac6 : SUCCESS in 2m 51s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/d5bdc4e7589c440d8f8aab816ae447a0 : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/6be30172bef7432a9375498d898b9aa2 : SUCCESS in 13m 18s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/141853856a0e495fa211169d827c890c : SUCCESS in 17m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ac8734b6197040188983ec651c596317 : SUCCESS in 7m 32s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/4f05ad492c044bfe99df2014effd1510 : SUCCESS in 12m 48s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/763726da85914457a70f332fde989d40 : SUCCESS in 3m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/edc6c6ddfebe4ab6aae87c265bc935bd : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e40c856285448b7bd9abf3a12879c2d : SUCCESS in 20m 38s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/82853ac0b7e14379af933710b3a59470 : SUCCESS in 3m 17s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/71dd0e1ee887423da7bd18715f06f4ec : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c3320a8d03742e38fd0999da0c1e893 : SUCCESS in 5m 50s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/b2765f9532574eb0bbe76bc9135c80b7 : SUCCESS in 15m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/29b848a4914b4da8a055178d34c245e6 : SUCCESS in 5m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d166329bfca14df5828cd4617395d1d4 : SUCCESS in 50m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1f47c0093f54a1a9c9efb61b93ea266 : SUCCESS in 1h 07m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ce6ac2461bcaac83c64cce6782a689abd7679b7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-31 17:46:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ee1b3b204f9d63b73f8215067c3b60ea02d5d594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 19:14:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4f28ce2869b4d1aa9c6d7bfb13cc498\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09b1bb144330479b947836fee342bb6d : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a8ec4330cf449b8966a8046f543d9c4 : SUCCESS in 2m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/313e6aa6b7e84e6eb8ad3386fcbf243a : SUCCESS in 2m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2f36ef419b34533a01307cffac41661 : SUCCESS in 3m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3e5970fd586f4200bf51ff93713d3d09 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c355e0712c84a9b98a952b47aae80b7 : SUCCESS in 2m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e883e04f58b44f198dc8c5aa88b7284 : SUCCESS in 2m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d41ff9b4b0de42db85222dbb598c9570 : SUCCESS in 7m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e142c044b9304b689acbe7d9debffb29 : SUCCESS in 8m 24s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c7da174b6a8040318580e2b69d892d43 : SUCCESS in 2m 51s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/2d0385fbf51f4aea9c193c640e7d1090 : SUCCESS in 4m 27s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/53e769edc8824fab934d867385276999 : SUCCESS in 4m 13s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8e15e5370b44476086f8e7c0ee81e6f5 : SUCCESS in 4m 56s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/ad2d532e5a7a4796aeee88cf54ffdc18 : SUCCESS in 7m 47s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/2691cf27e5a841e98ed1f224c9234002 : SUCCESS in 4m 34s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/931f90da220247358f6001ac57deb0a9 : SUCCESS in 4m 27s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/2d578a6cf81d4cc18f9b81111e2d2245 : SUCCESS in 6m 03s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/82b48f64ab624453bfa716e081dd0101 : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fe34e50fe20d461ea089ee921a4b4cc9 : SUCCESS in 6m 15s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/bdb8f52a41f7421f952c92ea4e6dcd3f : SUCCESS in 4m 50s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/2617429b031840e0b490584082b0975e : SUCCESS in 10m 24s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/d6261f073ab748e19d570013339f747d : SUCCESS in 1m 59s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/4d32203230654e3e82c2c7c807f85ddc : SUCCESS in 8m 04s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/f10a95daafb44c2092cd9e21bb0dfd29 : SUCCESS in 3m 37s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/09952614175544cda656c5c7a1c5b993 : SUCCESS in 4m 21s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/846a3fd72fd4483c8c017b4cec63c8cc : SUCCESS in 10m 06s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/ead0ca328e1c48d1a36b481b82a555fb : SUCCESS in 24m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5f120b2a46b242708dae0f6a88255b0a : SUCCESS in 6m 20s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/7fb63f24795d475e8ed60985bc2fe5bf : SUCCESS in 10m 35s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/e1307f8d4b2f4b8eb243f8fd3b602b4d : SUCCESS in 2m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/810fd972ba4f4939bdd75ea9fa969095 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/acf359cc58bc407785dc711fd34bea22 : SUCCESS in 26m 19s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/c82cdcd89594418bb75f7bae89b6065d : SUCCESS in 5m 11s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/19452e2e3f1e479a939de8808922965c : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba5e2b03c0f8407f935bc68a70283f0f : SUCCESS in 5m 07s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/7d094313babd45429ad77731c5c58c32 : SUCCESS in 18m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3b9d05d62d91485c89fd690367844de1 : SUCCESS in 5m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59a5482b8545476da68dd4054f966825 : SUCCESS in 47m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0183eb2bba0423db317dec262a02168 : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2bd354e47a1f2309450d6ec39b887a54ac4276c9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-01 08:17:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"57fbf5cab108e34ad398488ab6b9bb3479f4d0cd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-01 18:33:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"649704f6c766a2a1d75fadd26270f3378901c359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 18:35:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9cdad7142679849295c0b58595896cb74f747fe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 20:34:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f22de39c8a904f91a972ea8be1a78352\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f53bff9ce3a4929a0e4c3f031d21fd6 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b50594b594042b0851c261de878d90f : SUCCESS in 2m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ce3273b0cc144309744e610f168b8eb : SUCCESS in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d1dc7fef7134fdea14b6b173717beb7 : SUCCESS in 3m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a7202f799b0423e857378f7c1ca1054 : SUCCESS in 2m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/960ae582da2545eca8d9cef539b8f455 : SUCCESS in 2m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/986b03a3d6a147eaa1b4c2ff52410eb8 : SUCCESS in 9m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/90408d3225d144a4b38646b5e777143b : SUCCESS in 4m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6807111c21684dca8ef39193b8141a7e : SUCCESS in 6m 05s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/c3eda4cd30104f7d97e9020ccb12b273 : SUCCESS in 12m 34s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/7425807cedf042e388c8e406688c75b3 : SUCCESS in 3m 07s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/1b6d56ee4c72447987dab80d699042be : SUCCESS in 3m 45s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/b438a50ed61442f9b46efb5845dd5596 : SUCCESS in 7m 21s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/31ade54e8d8a4da58ce6b3130a319d61 : SUCCESS in 17m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d9c329b417cb4783a3dba2f7c228ca43 : SUCCESS in 9m 43s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/6028d60c3d1d4b27be1b6a90b7b77258 : SUCCESS in 10m 45s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/f338434b47f945ad9bfe0259fda60771 : SUCCESS in 11m 30s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/96a78e34f6b2490bb986a3b8923ff6ab : SUCCESS in 3m 09s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/01daa84541a74588a3bc4846ab1861db : SUCCESS in 8m 11s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/02e5afe66d604771bd27e33e578d3f6d : SUCCESS in 20m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cf8922ae74564e7187415a7281c6c19e : SUCCESS in 4m 11s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/1c8c9b5478a04ae897059d6fac7e812a : SUCCESS in 10m 19s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1f54fe50ad154696ab65fa7ec4ceacfc : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8b2fbde4cbf24bc8b511117fbf84e080 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/27749e7f7ce74e57968f22be12fcf656 : SUCCESS in 21m 41s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/414c4bc9c2d94c0f8f9d84fd85021305 : SUCCESS in 5m 27s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/6e0fc6ecab604fb1ac1145a40b4b94ab : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a3ea6682e8c4f2f8028cef4109ba0e9 : SUCCESS in 3m 38s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/08399f67e6d64d309e37a5c63db97da7 : SUCCESS in 18m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bc06bc176a4747399b380ec83a6cdb99 : SUCCESS in 6m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa3449d5f9164c12963b9402ccb7cd04 : SUCCESS in 1h 22m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28ee76938cec4fdaa074b459b03216ee : SUCCESS in 1h 52m 31s","accounts_in_message":[],"_revision_number":2},{"id":"bfb6c6c73c91410c9ef8df15788cba81294558eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 20:34:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7b28e15fd384c32787aa41f1f968df9a553f8d00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 20:35:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fef2590956e2417f8c01788aa524c6d0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f658406ab0bc4d34ba1442b52f764f86 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f76e6442a25143e43bbd71cea2332f2e61e2187","revisions":{"9aa84ee59b2c670da31a58ad3e06dc61dfd4c5fe":{"kind":"REWORK","_number":1,"created":"2026-05-30 08:10:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/990790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/990790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/990790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/990790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/990790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/990790/1"}}},"commit":{"parents":[{"commit":"296bdecc4f0a60b537f3554115a063e6957fd5b6","subject":"Merge \"update constraint for openstacksdk to new release 4.14.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/296bdecc4f0a60b537f3554115a063e6957fd5b6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-30 08:10:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-30 08:10:01.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I3d592e78023764d11b8656b10d231c81077399e6\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9aa84ee59b2c670da31a58ad3e06dc61dfd4c5fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9aa84ee59b2c670da31a58ad3e06dc61dfd4c5fe"}]},"branch":"refs/heads/master"},"6f76e6442a25143e43bbd71cea2332f2e61e2187":{"kind":"REWORK","_number":2,"created":"2026-05-31 17:46:20.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/90/990790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/990790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/990790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/990790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/990790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/990790/2"}}},"commit":{"parents":[{"commit":"296bdecc4f0a60b537f3554115a063e6957fd5b6","subject":"Merge \"update constraint for openstacksdk to new release 4.14.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/296bdecc4f0a60b537f3554115a063e6957fd5b6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-30 08:10:01.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-31 17:45:24.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I3d592e78023764d11b8656b10d231c81077399e6\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f76e6442a25143e43bbd71cea2332f2e61e2187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f76e6442a25143e43bbd71cea2332f2e61e2187"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
