)]}'
{"id":"openstack%2Frequirements~991960","triplet_id":"openstack%2Frequirements~master~Id69680af36e575448855ee9b093d3339a0d5f1ed","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Id69680af36e575448855ee9b093d3339a0d5f1ed","subject":"Updated from generate-constraints","status":"MERGED","created":"2026-06-06 08:20:53.000000000","updated":"2026-06-06 15:25:55.000000000","submitted":"2026-06-06 15:24:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991960-openstack/requirements/constraints","meta_rev_id":"879c7930b846d0d4699fbda1b490c7e1a0e44bba","_number":991960,"virtual_id_number":991960,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-06 15:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-06-06 10:53:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-06-06 10:53:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-06 10:04:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-06 10:53:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"434754280bf0e5df889f6d522e5f0d0bbc5c4aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-06 08:20:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c05c09636614cd9f8972c98fa18dc14a0f56c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 10:04:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/828a68b910714ab587f20f91c0a975eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0f09f3008f44609bfbe01a71f5feef : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/248d275df1e7443a972f112e957ae2dc : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/309350922b0148f6b2be47c35f2ba14d : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70ab892d13e04ae78c3f9cfd3d1faedd : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f061da1f25cc4c4ebba33b4e5c99d6f2 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4238d4913974944a1c3f5ac2433bc37 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/051e9cdf461e4d9ba71001abbe380bb5 : SUCCESS in 3m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8bdddbd0661244b3b2c3fabb2bcac6c6 : SUCCESS in 5m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d578e84ea27f4f22b434763b1db11ecb : SUCCESS in 5m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b52b812b8ab843d0bd6238dc4b5eeacc : SUCCESS in 6m 57s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/75d110873c924a2990230f6b5fa371c7 : SUCCESS in 7m 38s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/b0f2ba0802d84f61a5c3120604b65b18 : SUCCESS in 4m 57s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8b80bdb284a54f4f851665489d558cc1 : SUCCESS in 5m 24s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/4a3eadff81b9416d837c7dad38c771c3 : SUCCESS in 10m 35s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/a51265b5a49a40f3b110d1ecad8744a1 : SUCCESS in 4m 47s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/18266afcbaab499285f5d7335eb8ffef : SUCCESS in 5m 42s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/54a856816d38414b8a8b1ebdafa9ad80 : SUCCESS in 4m 21s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/ba1f679f9a414f4c8e0812a308bb1751 : SUCCESS in 18m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b6eaebff77c4153bc749b48e531cbce : SUCCESS in 6m 17s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/bae67a64b8e64712b52aad87ca531420 : SUCCESS in 7m 11s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/80d547fdeeac4c10919f3fdc93328646 : SUCCESS in 11m 27s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5f251839d0fa475a82fd7a86dde173fc : SUCCESS in 3m 42s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/3232bc5a52aa4aa7adfb3e8318b80bdc : SUCCESS in 7m 00s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/28b81829f4184b7a9e4722908903cc14 : SUCCESS in 4m 23s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/8b94526318f44cd7a487a0cd7fcc0657 : SUCCESS in 6m 05s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/e7cba4404a7a4b098f7c93299a009f49 : SUCCESS in 9m 55s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/e17aeac385a14b9983147c7dbaaf6d84 : SUCCESS in 17m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/361a5da6ee304a9caaf67c85a63d6513 : SUCCESS in 4m 01s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/c57a8af8d3bc4410ba4167a6a76bd9a6 : SUCCESS in 10m 16s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/8226328f98a745d6bca1833681e23e90 : SUCCESS in 3m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1753fbf2104548bbb971a79a8eb7eeca : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd1c667727e74724af4cd86b326c0cb5 : SUCCESS in 23m 40s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/6b5f6ad4e20f4c24a3207569497e3b23 : SUCCESS in 6m 23s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/a41c61c55727438eb3ddaceb09f5b101 : SUCCESS in 3m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f911d05bc75a4d82bfcc31774b77846b : SUCCESS in 7m 15s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/294e2eb4ff444e958b165a77656d571d : SUCCESS in 8m 46s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4968beb5c5c2439d8c1c9313e2b71091 : SUCCESS in 4m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31c4fd85a111407f9dd4af754d088dd3 : SUCCESS in 1h 03m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6d2dcf1150f46e7a807bdadb5038fc5 : SUCCESS in 1h 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9a7fa875bbffb76a226abddaa07e5cf453bb2a4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-06 10:47:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5f8b8b89a43b78723a52326ac1f8a84277033d03","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-06 10:53:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"79ec05115d6748ce78bd48d2a4b29625363212b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 12:46:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/535669bd75404f5b8505bebbd9ff118d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473dd065e085434bb0e68ad9ce62ecd8 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb032403554849c88e79ffc675e5b10e : SUCCESS in 2m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da7c70725add4239811a2e75771e91da : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e2af544733d4a8a9f7578ec41644121 : SUCCESS in 2m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ee35840a0f3142b38f1472878ff8b8e7 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cfccfb22846f4f8b80a9885612963ea6 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bea4694595d44a6b850a41f187d03d11 : SUCCESS in 2m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/23037693930c493a84a4562e06aadef6 : SUCCESS in 8m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/61bc3707f57040eb8b2e192e2d45432d : SUCCESS in 5m 38s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c6a2687da0114026b03176dc79ce28a1 : SUCCESS in 5m 14s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/f04cfad5e78e467b9877c67129b710fb : SUCCESS in 4m 39s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/0be51b8bca01493e98ad9349dd401253 : SUCCESS in 4m 49s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/d9682b02831941989b447ec56e17ed2b : SUCCESS in 4m 41s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/17906b7919fc46d48833b125fa94f147 : SUCCESS in 10m 00s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/d0c8a22b73e64012bc42c9e37231923f : SUCCESS in 6m 07s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/efb518cbc33b4085b63b58a2e5fc8381 : SUCCESS in 4m 57s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/bbd171b8699648dfb95d839cfc54198d : SUCCESS in 5m 33s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/f0c69919578b4e6980ec361fba1963cb : SUCCESS in 17m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2713c06520a04abea51cc80eec5a513a : SUCCESS in 6m 12s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/37bfb6256a76414580edb22d21d32c3e : SUCCESS in 7m 32s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/b17dc8b7ddd94e93b2b19fffbfe46f13 : SUCCESS in 10m 39s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/84086ae44e4a47ac83d8ff2d3da5accd : SUCCESS in 2m 26s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/114237b1d8d7421a885676e9013409ee : SUCCESS in 7m 41s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/ba941a165e2644a8819e93d31da9829d : SUCCESS in 3m 45s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/5e101739eb1c424ca6e6416cefb22826 : SUCCESS in 3m 58s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/9bad8d6973714f2a9d50901d7f3b4b3a : SUCCESS in 9m 46s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4bae4eb64e79403e9f1e914d765a7307 : SUCCESS in 28m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/69119152fbc74ed6a27f7503d1b88fd4 : SUCCESS in 6m 15s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/51148530af794abaaa0b8d367983f087 : SUCCESS in 12m 55s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/f8c2efa5f11c44a0b7f220964f5f9bab : SUCCESS in 3m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eb28043aad364b669853db0246e132e0 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92b80439e9cd46d487b86751cd1aad87 : SUCCESS in 24m 45s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/245d891fe8db4a80a5aa27e590270141 : SUCCESS in 4m 59s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/19c49bf0bd4346b893534b6496650319 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1612bac5fbe5424aaaf9eff4fe8eacb6 : SUCCESS in 6m 10s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/9ed90c52d3894784b91c5a74ff41095e : SUCCESS in 18m 46s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/aada91950e7b47c5ace7c71fe1425fbe : SUCCESS in 5m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/260749f0a2114fd9af6312c8a59d1dca : SUCCESS in 1h 08m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/304827da216d46359e292d6b784c4eef : SUCCESS in 1h 52m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9baaa1a79578920de548c6e3ce736d099501d462","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 12:47:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"66b364c90b25b0613e07881557c21aa7e8e00fe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 15:24:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23a8b04c259446d2b97cb944d435f90c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06221f42c2964356a1189e6b4b39298b : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d8543db3ad54dcab2ceb1afd6f942e2 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3e6b4bc7112400f8b1716216bb32bd8 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5694a914fa1a4fd59c682eb32209be04 : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62c99177f92a4341a3da0fb6e7333236 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c571e199eafe44df8f26aae852b4b80f : SUCCESS in 2m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/010786e196974e7ab9448e5407e487a0 : SUCCESS in 3m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/99ea5a5df8944cfca550c0dc7074731d : SUCCESS in 9m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3aeab5796f8148548599fc25e91eda07 : SUCCESS in 2m 58s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/34d80613b10d4b8594b6982a8df08953 : SUCCESS in 7m 08s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/4427b1b5a5dc4181bdb412cd0c9f82f8 : SUCCESS in 4m 07s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/9f62b4cee6e04232bd2655e927da0888 : SUCCESS in 6m 32s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/5ee3a11bdecf461fa8ba67a027072472 : SUCCESS in 4m 03s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/e92df23fabf6414ea2ef022d22f3454b : SUCCESS in 16m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1728ea449692450cb2fef188d093e37e : SUCCESS in 6m 04s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/42b79378b10c47ec85bd34a7a2937929 : SUCCESS in 4m 26s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/92d3c952138145a6a0b8acbe91aa980a : SUCCESS in 8m 09s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c0c2c2673be04783bd5562aac5bae8cd : SUCCESS in 2m 09s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/6566f28f4a3846bda61015c0f655d107 : SUCCESS in 9m 46s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/fd7d831399294a119397a1dc4d81d666 : SUCCESS in 21m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4cbdfc4c6dfa45db90b8f55c379be630 : SUCCESS in 5m 58s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/df5480aef7504e61bd4b044d319ceea0 : SUCCESS in 8m 03s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/143345938ebe418d8478f2057caa2dbe : SUCCESS in 3m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6d57aa3c0e3e458390b46c59477c57ba : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1914d54997bf4557858ecbebddd4f7b1 : SUCCESS in 25m 46s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/842a3f8ca19a4dfb92888f07c59965b3 : SUCCESS in 5m 04s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/7133ebafabfa437d91e078019b97de46 : SUCCESS in 2m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d8a3c0a18fb4752bf7d71650e2edbfa : SUCCESS in 5m 58s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/9bdfd1d12a1b4cdca62bc9a1e31378ad : SUCCESS in 7m 31s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2e3e3b46f3424f03a2ced51488fe508e : SUCCESS in 3m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d9bfe388f114a599b7aca87cb74015c : SUCCESS in 1h 03m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/183841a82650495187eb624344f8a400 : SUCCESS in 1h 37m 57s","accounts_in_message":[],"_revision_number":2},{"id":"30e9d819c883c8c463a4d81599a765752dcf5ba8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 15:24:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"879c7930b846d0d4699fbda1b490c7e1a0e44bba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 15:25:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfe07360edfc41bf859cd51710064c46\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884ac8d3180c4ec88d67815394a3a706 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4a4f7d3d0eb2a8db4602c027c45a70b3a991775","revisions":{"a75e5686ec232952d1656e4ccc2bce371a5e5676":{"kind":"REWORK","_number":1,"created":"2026-06-06 08:20:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/991960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/991960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/991960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/991960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/991960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/991960/1"}}},"commit":{"parents":[{"commit":"bdc4e18ff354825af258288e11491b4e2ce17f35","subject":"Add playwright","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdc4e18ff354825af258288e11491b4e2ce17f35"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-06 08:20:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-06 08:20:51.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id69680af36e575448855ee9b093d3339a0d5f1ed\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a75e5686ec232952d1656e4ccc2bce371a5e5676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a75e5686ec232952d1656e4ccc2bce371a5e5676"}]},"branch":"refs/heads/master"},"e4a4f7d3d0eb2a8db4602c027c45a70b3a991775":{"kind":"REWORK","_number":2,"created":"2026-06-06 10:47:28.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/60/991960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/991960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/991960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/991960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/991960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/991960/2"}}},"commit":{"parents":[{"commit":"bdc4e18ff354825af258288e11491b4e2ce17f35","subject":"Add playwright","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdc4e18ff354825af258288e11491b4e2ce17f35"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-06 08:20:51.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-06 10:46:11.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Id69680af36e575448855ee9b093d3339a0d5f1ed\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4a4f7d3d0eb2a8db4602c027c45a70b3a991775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4a4f7d3d0eb2a8db4602c027c45a70b3a991775"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
