)]}'
{"id":"openstack%2Frequirements~992233","triplet_id":"openstack%2Frequirements~master~I4642ba4acc0150df75fdfdfad8129d8e00d83db1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4642ba4acc0150df75fdfdfad8129d8e00d83db1","subject":"update constraint for os-traits to new release 3.7.0","status":"MERGED","created":"2026-06-08 12:58:23.000000000","updated":"2026-06-08 18:35:31.000000000","submitted":"2026-06-08 18:34:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992233-new-release","meta_rev_id":"a6b13e78b186a88658d2a44a26a1cc5cb9e01605","_number":992233,"virtual_id_number":992233,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-08 18:34:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-06-08 15:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-06-08 15:56:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-08 14:03:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 15:56:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"b99cf90e9898dded9286b842414388589f2d9485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-08 12:58:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bafc367320fdd57d5f2999baa1cc25dd70a4bc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:03:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07654b0ad0df4fe7922f523be46a12f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9857a14e915d4c5a9d6533d7a115f3bf : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f443419acdf4c318ab3c31f2802b6d8 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3334e02a7db041ec8b45f1e6a348bfe2 : SUCCESS in 2m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/74cd592424004908b96a38da718f4ad0 : SUCCESS in 2m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8e6957e187fa42e4806fddcafa1fda04 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a117fc6e69ea451b9b36df59cec17604 : SUCCESS in 2m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6585b8033c6d40ed90d4350c1532848b : SUCCESS in 2m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/42318ec047964c32812e04cffe302812 : SUCCESS in 3m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7a8b34e4c97649e590939965d27856e4 : SUCCESS in 6m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8f5d7f5335a94adb8680a9ccefe4ac46 : SUCCESS in 5m 39s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/40be2ec201c340a4896a37e117124d77 : SUCCESS in 4m 46s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/e7daaee82e404d6db4194943236e33f4 : SUCCESS in 2m 59s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/97fc08bcb0204d71bc6876b352e2d93d : SUCCESS in 4m 50s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/86d3a24b9f0c489f89313c0800ce5b39 : SUCCESS in 9m 16s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/97bd672dad7740088f1565d98ba554d7 : SUCCESS in 3m 11s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/a5bea7044c344671b7e495780d7043b4 : SUCCESS in 5m 02s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/16af82dc61d748d88cd461c69c66e599 : SUCCESS in 6m 37s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b2d44fd8154348a496e6ac3cd7a9222d : SUCCESS in 7m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/983d8f33ba6c44039388870726f4bb16 : SUCCESS in 4m 54s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/5c23605fe22145218c3a6313e6135445 : SUCCESS in 9m 55s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/5cf042863bc340b89b8025590f278350 : SUCCESS in 11m 47s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/eed88d4a1d604457a7470b9d8e878868 : SUCCESS in 1m 45s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/184139486e234cf78f9751df1a12f358 : SUCCESS in 6m 40s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/8864928bfe3b4cfeb074ea48f68621ef : SUCCESS in 3m 23s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/2247bb4b6c674d97a8783c6dd0a65730 : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/5dd110c3af834d369724e535f5df9ec9 : SUCCESS in 8m 36s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/77c4bd1fdd7f491986efaaaf3224f55f : SUCCESS in 22m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6682603804b048f29bc35815ca3e34bc : SUCCESS in 3m 27s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/6675f0d660f4495c80836a35ab08ea91 : SUCCESS in 15m 55s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/22fb78e0e9654c9f861c2fadbeefce0c : SUCCESS in 3m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a3b4cafafe70499cbffcd1df6ca2615d : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9838b0f61e548ec824dd49f80a64cce : SUCCESS in 30m 11s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/75912236d9e94fe58c6181e5ae8cd3c3 : SUCCESS in 5m 16s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/a9e628e653444bad81c0cb737209a749 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fa42de305bbb4eda8ced62f9454fb82c : SUCCESS in 4m 00s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/202bba2a5e74476d8c43108b6b5a8780 : SUCCESS in 17m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/10661a05af07435ba670556163b05283 : SUCCESS in 6m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c354256b3f414f4789cddaeaae31a172 : SUCCESS in 1h 03m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06f86a2b41de4d2eb956dcb1d5d80256 : SUCCESS in 56m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f86335065c0341a13f276c2293aa7e5094f320d0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-08 15:56:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc4e296e1d2fc7edf206bd8ade93b7423577ddd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 15:58:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de7d2246b6b2d8bda04b3bf02d7beaf05b14e3e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 18:34:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abcedcbcea2b44248cd46c32c96ab836\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3d53cfd0a24fd5909d099a3b93cac5 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb41dc4ab954a52a758f2f1955ecea8 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6a8226699774bff9ffde647c0845ec1 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d2eef74fa9e462e8d3ed96a8d9f6168 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c57896b383734607a014e1929677734e : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2f909723eb6541fca4501b70cd4772a3 : SUCCESS in 3m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8c545374131340669375c30fdad0e928 : SUCCESS in 6m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1d430595a9a34e57aa54e79b8604e0ce : SUCCESS in 3m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8deb4f36be2646efbe3fb6d6cee384e3 : SUCCESS in 6m 25s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/f0820733d2e94b678863f263b22ab18f : SUCCESS in 10m 59s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/a1c3dc9b8fbf4f9ea1b84803441dd68e : SUCCESS in 4m 45s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/44c29cf8c7a4453498b589b961f12fc0 : SUCCESS in 6m 42s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/d39cc5c54bfb4044b73ca7464cb50ffb : SUCCESS in 8m 15s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/06de3b86ff5746e3a58511707b3f5659 : SUCCESS in 20m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b74c5de1ef3f49ba9fe788b7d5adbbdd : SUCCESS in 8m 10s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/bd09bb45ee3b464aa978e96c0d31fbed : SUCCESS in 8m 28s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/6f33848571ac4b85a03dc7b83ec3fce8 : SUCCESS in 11m 00s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c6b295fd576e44eba68b0ffd66b8a2c0 : SUCCESS in 3m 19s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/28e33767165f4bb68a12216afe401a9f : SUCCESS in 10m 06s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/ec1b0b1ba3a14954bb3e9b50d7bc5483 : SUCCESS in 22m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f429f21992824db7a52b018c87f12e94 : SUCCESS in 6m 09s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/71cabaeb51424fc9be62fb4c90b66a16 : SUCCESS in 16m 30s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/2bd4e82ee7564966aef608c25c43e6d6 : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/728375dc465043cc8fbabba35d5ee142 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38c26e753f474fa0ada8a931b4d5dfd6 : SUCCESS in 24m 33s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/751caf536450429b9e7dfaca9f531856 : SUCCESS in 6m 07s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/8801b9199ae74c788a2e13d20f8ca2df : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d812819993394614810513f5d1ced162 : SUCCESS in 7m 15s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/8a9f6c9ed1aa46bab52836633592e66c : SUCCESS in 21m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ed0bedb3ad804d70ba31352065b0e667 : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c857098a5c74fa2ae47ad19ebb68099 : SUCCESS in 36m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5a4395ace1746f0804c79cf3bd057ae : SUCCESS in 1h 43m 01s","accounts_in_message":[],"_revision_number":1},{"id":"81e03b7265f807837dd464eae2b0594ce841f2f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 18:34:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a6b13e78b186a88658d2a44a26a1cc5cb9e01605","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 18:35:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd5025fc4dc648ba84c5a65735ec81d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5c1311fa0c048558168653ecdc4d363 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df8c1d8697a531140de152d548ae02b731e1a235","revisions":{"df8c1d8697a531140de152d548ae02b731e1a235":{"kind":"REWORK","_number":1,"created":"2026-06-08 12:58:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/992233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/992233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/992233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/992233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/992233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/992233/1"}}},"commit":{"parents":[{"commit":"cc073afc19a64a11fc7ffaa51df8f449e1c27644","subject":"Merge \"tests: Add missing coverage for project parsing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc073afc19a64a11fc7ffaa51df8f449e1c27644"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-08 12:48:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-08 12:48:22.000000000","tz":0},"subject":"update constraint for os-traits to new release 3.7.0","message":"update constraint for os-traits to new release 3.7.0\n\nmeta: version: 3.7.0\nmeta: team: nova\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Balazs Gibizer \u003cgibi@redhat.com\u003e\nmeta: release:Commit: Balazs Gibizer \u003cgibi@redhat.com\u003e\nmeta: release:Change-Id: I2291b42c9e31244e51b6ddffa1cbff59fa66e174\nmeta: release:Code-Review+1: ribaudr \u003crene.ribaud@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I4642ba4acc0150df75fdfdfad8129d8e00d83db1\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8c1d8697a531140de152d548ae02b731e1a235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8c1d8697a531140de152d548ae02b731e1a235"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
