)]}'
{"id":"openstack%2Frequirements~992491","triplet_id":"openstack%2Frequirements~master~Ibb8383c48896894dfd6afe3d6fa87df49a71c4b2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibb8383c48896894dfd6afe3d6fa87df49a71c4b2","subject":"update constraint for os-traits to new release 3.8.0","status":"MERGED","created":"2026-06-09 17:04:32.000000000","updated":"2026-06-10 05:00:59.000000000","submitted":"2026-06-10 05:00:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992491-new-release","meta_rev_id":"1fff835d8dd4527d6aeaef10183fe8e1dd2b2fed","_number":992491,"virtual_id_number":992491,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-10 05:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-06-10 02:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-06-10 02:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-09 19:10:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-10 02:51:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"53ff2ffbc34a6b7cfc129b52343e5570cd976d38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-09 17:04:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"009b794360dea547816eb268ad6432208ccc98d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 19:10:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c990a364d4f4be9b584ce3d80f190d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a7f6974caac4446b2f033fa280c1ea1 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d6cf84ad3b94e65ba40f6790dc44f95 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/949ee712713e4a5b89fc7530f4979d21 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67d1d036c63145d5ad3df83f0f1608bd : SUCCESS in 4m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8110fdd51c36424ba61e7f1c2c8ac725 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf681bca6e7449e4ad1b944cfc86419e : SUCCESS in 5m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82f79f20e00547db94ac5ef45bfc6fe8 : SUCCESS in 2m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8d6ce74bcd54488cb7a750809a5ab453 : SUCCESS in 7m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c144a856d41c42b2a26bfe1a800da744 : SUCCESS in 6m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5a98497df6e54bd2a80a6bffd9b7e271 : SUCCESS in 6m 17s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/305499c9b3ca4456ad2151923bb2e268 : SUCCESS in 3m 10s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/39fc66fbcf43450ba485d0bc4094fe61 : SUCCESS in 4m 41s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/60823e665cff476992c029333dcead7b : SUCCESS in 4m 44s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/5766f6a1bb654619b46e649971808b25 : SUCCESS in 11m 18s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/214dbe7ef7e244959db66ddaeb76895d : SUCCESS in 4m 50s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/7fca4b9d7ae1413aaee7701de5b35ac4 : SUCCESS in 9m 45s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/5bf1ae15bc9d4dd396503869add156d6 : SUCCESS in 5m 32s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/94be9bac09c8458fb0e2d1c18da8cb00 : SUCCESS in 7m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8bb5df0933544d7796355c571f17b28b : SUCCESS in 5m 23s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/a99717fadb5745eea9dd181b558802e9 : SUCCESS in 8m 21s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/8d5222b114284d0e9dd8a8556cdcfb8e : SUCCESS in 12m 14s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/2b455159f3014316bc98813a0c4833ae : SUCCESS in 2m 37s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/cf14a702fb894617910d2c6848b76638 : SUCCESS in 8m 17s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/199403eee3af4978b8730cab6cdf7903 : SUCCESS in 4m 40s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/61caa26987e543e8b8155aab61f2617a : SUCCESS in 3m 38s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/da1d74c4776d4d788294a07b3d0e2119 : SUCCESS in 10m 24s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/012f7512088e4780bab798149e7fd455 : SUCCESS in 22m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4b4c5923ad364828ad825a0d018501e0 : SUCCESS in 6m 47s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/8d9dfee67e324e2aba5c2edbcc05401c : SUCCESS in 13m 35s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/264fb90a5709405587e1266504fc3f44 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2de6c5c0b1324d86adb62fa53d42b654 : SUCCESS in 2m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/540d45c83a894123a13b47ce95439de0 : SUCCESS in 27m 37s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/b4a92ff498904412a17483f60289d8c0 : SUCCESS in 4m 14s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/887d5eea98c84f5992b8faf9143dabcd : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f906a827ee644a3bea6336f5179aafd : SUCCESS in 7m 19s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/faee1fa43b734fa58bde8eb4c1747f99 : SUCCESS in 21m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6307f20c9f0441029ccc4fb978d1ec72 : SUCCESS in 5m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1385df3a5a2249b89371bca63255fd0a : SUCCESS in 1h 07m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07b52a8936d94f699a36ac76c30ed966 : SUCCESS in 1h 58m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d8fdb4c030c78b6a9c61011ca5851c753e08939c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-10 02:51:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ef06f17ee93501098911bf658c2951bc2ef75bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 02:52:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6397eafe3a094d55d57b874c6cb217fe5810d52e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 05:00:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/595991f9ab2249d1acf1d5a9c0023580\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31da688621e5466ba9a7d8d1aaaa3158 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1db64ce362c64302b295d48cac96a62c : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5635f6bee2d049cc806138480755801a : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91172474b21a4de280fbf818f4c7fdf4 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b6ad0b015cd48afac64be68f18f04cc : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a498d46e1e4468493a16fd705612ecc : SUCCESS in 3m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0b129591494649448b3c4afd2d1d5b77 : SUCCESS in 6m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7c7995607d1141ca92bc0b378c405041 : SUCCESS in 10m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0485662832ba445c97c82304a25103a9 : SUCCESS in 8m 15s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/914c6a4af30c4266b0aeffb10208f2c2 : SUCCESS in 11m 16s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/78098140e5964a9cbadac5ddada1f6a5 : SUCCESS in 5m 24s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/f3408acef6f742b0ad7713494c6e2220 : SUCCESS in 4m 08s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9f15be9bded6455e93dd78712f0322e4 : SUCCESS in 4m 21s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/57fb7c2a5c5541eda14d374e86d99146 : SUCCESS in 14m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f8ce1c8fea9b4386b93dbae65d228f39 : SUCCESS in 6m 36s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/c1c8477796ce4a09bb9e2f577a77d2e6 : SUCCESS in 9m 02s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e8a48f24ec2f4e61b3f48eed4f09e0e2 : SUCCESS in 11m 46s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/688f5a14a2324d49b91e2a346aceb943 : SUCCESS in 4m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/7ae679bcd45447e7a381a24919911702 : SUCCESS in 5m 58s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/a74eb0c2afe34f92b75d65b064dda7ae : SUCCESS in 28m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6c95b0b0b8ae45caa523bef1e8c4af1b : SUCCESS in 8m 58s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/c7c7d7e78ff749509c5a735d81bd0543 : SUCCESS in 12m 22s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/abada2a9eb6344d0ab40e104f946eb7e : SUCCESS in 6m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d2efa9d400cf418da378644e127c6734 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c243bd91b014725af4d577eb2029506 : SUCCESS in 21m 33s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/792a944b59cd418aa38cc9b5f9e723c5 : SUCCESS in 7m 08s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f7d9003b67224c6195fd3fbe6f44887a : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d7b3b3c8ed294922aa5a953d9bb98bd6 : SUCCESS in 6m 27s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/9abfaf9908604b08b73e7c903c261ee1 : SUCCESS in 19m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d6f5f5229fe5492bb80ab60fef0dd06e : SUCCESS in 7m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49823102c0834e68839c2b136f4ea466 : SUCCESS in 1h 01m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47d767bea88c441b8288ea0d7cc9adfb : SUCCESS in 2h 02m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f5d6614933f70f64a53909d49bb2a63901684538","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 05:00:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1fff835d8dd4527d6aeaef10183fe8e1dd2b2fed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 05:00:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c07ef054d8047cf9f4319c8067d48ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9651cf19f3d54be59f5c8ac899006627 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9faec191e5cc2c60ede2a200fd083aa4cd8c3b6d","revisions":{"9faec191e5cc2c60ede2a200fd083aa4cd8c3b6d":{"kind":"REWORK","_number":1,"created":"2026-06-09 17:04:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/992491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/992491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/992491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/992491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/992491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/992491/1"}}},"commit":{"parents":[{"commit":"df8c1d8697a531140de152d548ae02b731e1a235","subject":"update constraint for os-traits to new release 3.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8c1d8697a531140de152d548ae02b731e1a235"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-09 16:54:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-09 16:54:31.000000000","tz":0},"subject":"update constraint for os-traits to new release 3.8.0","message":"update constraint for os-traits to new release 3.8.0\n\nmeta: version: 3.8.0\nmeta: team: nova\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Balazs Gibizer \u003cgibi@redhat.com\u003e\nmeta: release:Commit: Balazs Gibizer \u003cgibi@redhat.com\u003e\nmeta: release:Change-Id: I276d06d3026d2fa658b2fd7cfcdb1a14693d4b50\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: ribaudr \u003crene.ribaud@gmail.com\u003e\nChange-Id: Ibb8383c48896894dfd6afe3d6fa87df49a71c4b2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9faec191e5cc2c60ede2a200fd083aa4cd8c3b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9faec191e5cc2c60ede2a200fd083aa4cd8c3b6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
