)]}'
{"id":"openstack%2Frequirements~992897","triplet_id":"openstack%2Frequirements~master~I730ec0449325834122f2ae12e5851665e0251bbf","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I730ec0449325834122f2ae12e5851665e0251bbf","subject":"update constraint for oslo.config to new release 10.5.0","status":"MERGED","created":"2026-06-11 13:42:55.000000000","updated":"2026-06-11 17:37:23.000000000","submitted":"2026-06-11 17:36:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992897-new-release","meta_rev_id":"484d470d4e5176f171586ebb11ced74a045c0404","_number":992897,"virtual_id_number":992897,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-11 17:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-06-11 15:42:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-06-11 14:21:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-06-11 15:42:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-11 14:21:26.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-06-11 15:39:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-11 15:42:57.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"81b2e9e87e2f260369565b06b6b65b52885bc18a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-11 13:42:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2e2b7e8606ab321f897e6d2d9006095efc9c8bb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-11 14:21:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b0103ea8a8d4bf590f24144e506f3e17926f1982","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 15:39:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f2a734563b4b6fb2b3b1034fe8190e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f2d9890c004b0581eff9a5178ce043 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e486171520c24da3b7bcf5499c5f7649 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb027559978545debf378de9ecef4f6e : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c7a3c600efd448eb1bfa566e3f081bc : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ee08e5392d95437a815b99ab04d8c5a4 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d7be369e0314f9d9cd99295844648e1 : SUCCESS in 2m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0611098f8d12432c9fc1e1f9f643c300 : SUCCESS in 4m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/96b790e5385e4ed38c6f651ec2ff7a67 : SUCCESS in 6m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5c5a01a85d3e44ebab36eb44a3de4978 : SUCCESS in 5m 57s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2a29ab373ddc4b55ab1b6322cfdd1606 : SUCCESS in 3m 59s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/13763b7294dc4fdc8cbce7e37db47e50 : SUCCESS in 5m 07s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/066cd3e192bc4cb2845167357ba70d57 : SUCCESS in 5m 14s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/067e7279006b4de79ea658218fedde97 : SUCCESS in 4m 01s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/c76d77b7d4f14d0aa26e976d34666396 : SUCCESS in 10m 56s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/db6a48787a5e4a50a91755697de9805b : SUCCESS in 4m 52s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/7f0e249e904d41da9f582bd6f6c05f87 : SUCCESS in 7m 02s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/31153a7b0b5f4062a58759cab3144eda : SUCCESS in 4m 06s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a719d8315606414a86dad21a2b6cc3de : SUCCESS in 17m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bcbe7335998c42f39756da58d3e21489 : SUCCESS in 6m 16s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/754c8a21474e4f85b2b41bcebc2aac18 : SUCCESS in 7m 31s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/a188a54e0763493bbe254860bd1eb3ce : SUCCESS in 11m 07s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/abc94c26a2c244e7ae40edf6309de0aa : SUCCESS in 4m 05s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/952a0833e3b24452a1781f4fc26e4df1 : SUCCESS in 9m 49s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/5ab0cca40b8f4750a1fcea6b01e826e2 : SUCCESS in 4m 53s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/2af0eb045a744edfa2eb44d9dcb1917c : SUCCESS in 4m 48s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/fd6bcb04ade24c73948c6f3b27e30cc5 : SUCCESS in 11m 32s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/7f1aa5e7a37846f1afb67ce258ee3c23 : SUCCESS in 32m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9d579906fe80406680d8e264cfdee4ad : SUCCESS in 8m 15s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/1be5f83724a04912bd35afa9ec5a68ae : SUCCESS in 14m 33s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1f1a86206c7b4bbc95dd0a2db956947a : SUCCESS in 3m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1fbd136a98484ecbb2f7bb4ae338c2fc : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b73a96a67a643d5bb75e58250402577 : SUCCESS in 25m 47s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/713e5671311a4201b28754f08312d2b5 : SUCCESS in 5m 38s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/348168faebd24838a73e86aa6fae25d6 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3042474961aa43a78e2a20f2e9a76412 : SUCCESS in 7m 09s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/99c453108cf44cf299b8340485875641 : SUCCESS in 12m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a8ac9d6569bf4bc68948c531345dedae : SUCCESS in 6m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f61925275dc4528a3cb640391afdd29 : SUCCESS in 1h 03m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76f7b48bbaf74246bb8bd45329998ed0 : SUCCESS in 1h 41m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4669041dc100262a371dcd9f010e0593316152ea","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-11 15:42:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4443ab6aee5209763099cf5e6078c451b542742","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 15:43:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f55be1ee8a8721471d8331ee6b5f8fd1cd5dd12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 17:36:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be59f1cd57c1416380e51855441c231a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e5f4241198e4d28a978f81a3169eddb : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e337326e5c7a4cc3bd135f0d417c3ee2 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e1090776eaa464cb656718f14b0fd87 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/762d34a17bff497cbffae656a02aa60a : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17a53b4c69474add8a2c1271102ce322 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/91721c2d3ff04d1c968c0664104d32b1 : SUCCESS in 3m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/da22adcda28d456d99351033da248066 : SUCCESS in 4m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/710a32a6a4f647c088b84381c189bd7f : SUCCESS in 6m 32s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/619f0edd934f424691593c89b4f72655 : SUCCESS in 9m 52s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/25cecf5287d643838d323c1b820cdc63 : SUCCESS in 10m 51s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/b2fe402a7ece484a965c92284fc9cb61 : SUCCESS in 7m 23s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/14143a9d1e874628a84a2b7315f6e675 : SUCCESS in 5m 43s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/2523cf60b35e4f40902c67b20d2078fb : SUCCESS in 4m 26s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/ba4161d732e04029a3daebf62fc672b6 : SUCCESS in 16m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ff5cb66917c448a819db8fc06c1d824 : SUCCESS in 8m 58s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/00d9ca2373794bb2b250c12716167661 : SUCCESS in 6m 03s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/4f1355402f1c4eedba3e89c7b23bb0d5 : SUCCESS in 7m 42s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/fb2e18779dcf451ca2bb5bb555a2a855 : SUCCESS in 3m 06s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/aac8e32b36fd4c0cb5032911ea5635d8 : SUCCESS in 8m 15s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/aa6a161f174d4bfd99fd15c21db27701 : SUCCESS in 25m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2f3960a3e7e84d20acc7ce8c87be1b9b : SUCCESS in 6m 10s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/71ec9a7de7da4099b187476e9d82c568 : SUCCESS in 14m 32s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/cd7d14d2f3374819916f1f7be65ac58a : SUCCESS in 7m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/696d7f2672c44eff8a5616a927190ca7 : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90d52a29eb7e4717b427a9dc57127c3a : SUCCESS in 25m 26s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/75d74ef0e42c4a2d8bc5d6db30888243 : SUCCESS in 6m 05s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5b1d514fb543425598e825b83b3326a3 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ef4dcd43d7db4946acba5bb0f76113ce : SUCCESS in 4m 39s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/b2da59329d5947eaa70aef1da1351174 : SUCCESS in 24m 28s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3ca96776ec2547a1b5bd4bae7f751906 : SUCCESS in 7m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb36effed40f49fc90e2876d59f6c431 : SUCCESS in 1h 03m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/096b0e3315464bb484429b1df888644d : SUCCESS in 1h 46m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d75fa61b9db7fb10f8fa50d118daa05c8b95727f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 17:36:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"484d470d4e5176f171586ebb11ced74a045c0404","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 17:37:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a35fd76bd1f4de0b714742193104bc2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d5a543bd72496ca4205ba446287c89 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d4eef13ba933bfce9d4ab6e934ebdb2e1abf4a6","revisions":{"3d4eef13ba933bfce9d4ab6e934ebdb2e1abf4a6":{"kind":"REWORK","_number":1,"created":"2026-06-11 13:42:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/992897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/992897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/992897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/992897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/992897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/992897/1"}}},"commit":{"parents":[{"commit":"bc78431533f2e05fc2c68d44c245e3f21d7a4550","subject":"Merge \"update constraint for os-traits to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc78431533f2e05fc2c68d44c245e3f21d7a4550"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-11 13:32:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-11 13:32:54.000000000","tz":0},"subject":"update constraint for oslo.config to new release 10.5.0","message":"update constraint for oslo.config to new release 10.5.0\n\nmeta: version: 10.5.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I96a0aa930e8b5db2cca8fc4fcd2ccc55595dfa63\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I730ec0449325834122f2ae12e5851665e0251bbf\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d4eef13ba933bfce9d4ab6e934ebdb2e1abf4a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d4eef13ba933bfce9d4ab6e934ebdb2e1abf4a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
