)]}'
{"id":"openstack%2Frequirements~993074","triplet_id":"openstack%2Frequirements~master~I9e39ba8a71a2414a3028862308b41ba5a12b1c15","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9e39ba8a71a2414a3028862308b41ba5a12b1c15","subject":"update constraint for os-ken to new release 4.2.1","status":"MERGED","created":"2026-06-12 06:23:54.000000000","updated":"2026-06-12 10:46:11.000000000","submitted":"2026-06-12 10:45:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993074-new-release","meta_rev_id":"4e2ac44f452cb8cb92772559472c87c61f121697","_number":993074,"virtual_id_number":993074,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-12 10:45:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"all":[{"value":1,"date":"2026-06-12 08:42:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":2,"date":"2026-06-12 08:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-06-12 08:34:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":1,"date":"2026-06-12 08:41:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-12 08:21:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-12 08:34:19.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-06-12 08:37:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-06-12 08:37:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"CC"},{"updated":"2026-06-12 08:41:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-06-12 08:42:49.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"}],"messages":[{"id":"7fdc0d89423a4f9c46c6ccb45a07f512a7d9e005","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-12 06:23:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22118c04a883b0219a24122afdfa311012f97f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 08:21:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2de8b57429d74e3da92f57cf295af720\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34587dadc142464e84419fb3c595c82e : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adbf9f11350547cea0572e95bcaeb1ea : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ba545feeb4b460a933096393f58d3f0 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04b5c65b23f5479c912d92e0512c1648 : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aa3b5258f751403cb72a2483ab245544 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54aea22a52dc4593bbc6857dd95c9ae8 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d376a0e189ea411f9e43e273837c346a : SUCCESS in 3m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c785ff5e1b9a42b9a15b33929f5c99a2 : SUCCESS in 5m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ecad589be5ca4d209a3f425c4efba79f : SUCCESS in 5m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7a5f8cf2919c4967a1b82e6b98109ece : SUCCESS in 4m 52s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/e92f3797f39e40e7b9754d817b304a03 : SUCCESS in 2m 49s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/441c17df7a114977b8c935ea0b1fdacc : SUCCESS in 4m 21s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/a1d3d37350b74f18b7cd0e20ddee4863 : SUCCESS in 5m 07s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/f528b7efb8e941d5bc10b2dfe768ffb3 : SUCCESS in 8m 48s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/862954363cc449d18fd0062e4394794f : SUCCESS in 4m 29s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/924666c5c64f481cb43a495870533126 : SUCCESS in 2m 38s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/781f70e86d1a4c4bb644ef8ab57efc3b : SUCCESS in 4m 13s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/d39f833acee54bd88520ba0c2c597e5e : SUCCESS in 16m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26a48874ccb94f9f837ed3f72b42fe8a : SUCCESS in 5m 54s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/0ad52cd1b8974b549af1a78727bb7958 : SUCCESS in 5m 20s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/c20b65bc29064b55bc608115147619c3 : SUCCESS in 10m 41s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/00eb3b2a73404f71b024b39cdff57f57 : SUCCESS in 2m 16s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/68bc98de89304e0aa6eaa317f18e65a0 : SUCCESS in 6m 18s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/b8de7f0c530343e990b72620b5c6cceb : SUCCESS in 3m 59s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/799ff2ea89cf42e6a86e24b3e9147965 : SUCCESS in 5m 13s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/60f160d6ce76465cabe2b3a351a43f7a : SUCCESS in 11m 13s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/858ebee2c68747009912f1b1bbabfc3b : SUCCESS in 20m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c36ebfad7cb3407a9e11cbbd6e7eaf87 : SUCCESS in 3m 19s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/00a2d1a167ca4690ab075bc7d9eaa884 : SUCCESS in 11m 00s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/b9e6d1a7e6ef436d91010a041c8a9b96 : SUCCESS in 4m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9ac98d95276d49fbaf9c806db30d8860 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf2e050701164737ab12fa42431ebdd3 : SUCCESS in 27m 14s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/295d495c3e064810ba1bb1fc7bb71b2d : SUCCESS in 5m 14s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/2709aab7aaaa4ec2b7f4d5ec4ad3b6e9 : SUCCESS in 2m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b176aa9714d47d59f5e6c1014f22907 : SUCCESS in 5m 35s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a473ac1b157740d291c0dfd59d3eaa07 : SUCCESS in 18m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b916b629ede947c09cc7cdce4818940b : SUCCESS in 3m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f5d2a2a97c64e2ba67349b408cca1c3 : SUCCESS in 1h 01m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2332ac1c6420464b9031322deb558f3a : SUCCESS in 1h 50m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2a33c43c5448f13020bc7f059543d65eb5e29cd1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-12 08:34:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1a853ce3a1e84d981dee815c48bc12f8303cad3f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-12 08:41:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91cc1b54af148d14eb36ce106602ff31695379bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 08:42:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0c92937067a99ba6a92e2cc13c04fac5ef7f7ea","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-06-12 08:42:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ca4c52827e0b2fb1d912d1c92e2dc71904380ee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 10:45:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5d0f9b1feed471dae36c8b20303de0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b07c72c6ee64b8dab5633f171f45f56 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca8157382368470b9887d3a999d082e5 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30e04b9d5a754576a8b9bb0a67443f8d : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4317377c0afd4853bea46f12b230223d : SUCCESS in 3m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f787e2cb1bd443c89cd1d199bec0759 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9de84b72af74c5aa085cc420e5c12a1 : SUCCESS in 3m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3730176055ed4f2387ce7d758fba589b : SUCCESS in 5m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/be9002fd5e014be8a8526f8c9cf2ed37 : SUCCESS in 4m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ad75366a2f3548bca9a7b3a61a0855c7 : SUCCESS in 4m 57s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/9a1157f9a701479cb5240dd82f610118 : SUCCESS in 9m 45s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/d8a4b0cffe21462dacfabc8c08a88d61 : SUCCESS in 4m 41s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/2e99666094024676b854935f8253abc6 : SUCCESS in 6m 51s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e873182a713043bebb6a6c3b4cfe77a0 : SUCCESS in 7m 14s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/ca8866def41645f98ddaad45c52b6528 : SUCCESS in 17m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b5e2ea95f4f40c68bcf8051c6511ccc : SUCCESS in 9m 40s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/39b08900ae3345c9ab9c5000b25115c5 : SUCCESS in 4m 23s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0d723a0f49d344fcbc6f97fc12040577 : SUCCESS in 6m 42s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/dfdfa18815fc4fbe8fe38da3f1724fb0 : SUCCESS in 2m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/4d86c414991b4327a974f1b52728b3bf : SUCCESS in 9m 24s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/e3306feb5e1647199d3b8fa7c5c8c0bb : SUCCESS in 16m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4c7cc9d6271d49dcb72e224fb38b7c4c : SUCCESS in 6m 18s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b3bf09c7eba34f9b80aefea3f4cc0564 : SUCCESS in 10m 06s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/58c33b3b76a547b89e0884c76e0a7d5b : SUCCESS in 2m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/73feaa8f89cb4fa5a76349c7c1b75c59 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f5a9b464815f4d7f99e47127ed73c9d0 : SUCCESS in 25m 03s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/17d2cf499632403d87448ea001075264 : SUCCESS in 4m 46s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/7680c35c9aa34473b64efb8a9c1344fc : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2517d71bf0844be8e8d6cc606f380d2 : SUCCESS in 5m 24s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/154a2429ad644abe925ad5484e477848 : SUCCESS in 11m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f895fb540dc04661b88fdfb01a0e9a13 : SUCCESS in 5m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/387674102dfd4080b348d3a42a6ca08f : SUCCESS in 1h 08m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/446e8773984a456fae162b2a355306e6 : SUCCESS in 1h 58m 10s","accounts_in_message":[],"_revision_number":1},{"id":"472511b6d1a92304a3f2311306e3eba26dcba65d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 10:45:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4e2ac44f452cb8cb92772559472c87c61f121697","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 10:46:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f108941f5ef46cd93b9c5e58b8781df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e36005cd32914cd79b8e7b151ddbca83 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eba28fa2422034f326790c0b9e34526573f2ae36","revisions":{"eba28fa2422034f326790c0b9e34526573f2ae36":{"kind":"REWORK","_number":1,"created":"2026-06-12 06:23:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/993074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/993074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/993074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/993074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/993074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/993074/1"}}},"commit":{"parents":[{"commit":"68f087c3e34429cff85a9b72c52812e94c2d3faa","subject":"Merge \"Remove logic around Python 3-only branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68f087c3e34429cff85a9b72c52812e94c2d3faa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-12 06:13:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-12 06:13:53.000000000","tz":0},"subject":"update constraint for os-ken to new release 4.2.1","message":"update constraint for os-ken to new release 4.2.1\n\nmeta: version: 4.2.1\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I782bf231cdfb799962dfc3eeafc6da1a9ea9e769\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I9e39ba8a71a2414a3028862308b41ba5a12b1c15\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eba28fa2422034f326790c0b9e34526573f2ae36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eba28fa2422034f326790c0b9e34526573f2ae36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
