)]}'
{"id":"openstack%2Frequirements~993204","triplet_id":"openstack%2Frequirements~master~Iccaa9553d658295165cf678cec2fae94519b06ff","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-06-15 19:13:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iccaa9553d658295165cf678cec2fae94519b06ff","subject":"Updated from generate-constraints","status":"MERGED","created":"2026-06-13 08:19:11.000000000","updated":"2026-06-15 19:17:19.000000000","submitted":"2026-06-15 19:13:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993204-openstack/requirements/constraints","meta_rev_id":"738259030c74f40879c50f8e4ae9d169d67e05c3","_number":993204,"virtual_id_number":993204,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-15 19:13:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-06-15 15:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-06-15 15:22:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-13 09:27:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-15 15:22:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"7d44160b730d70e55f068fdf8fe99a7f56bbc96c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-13 08:19:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc7daa3ef8b4d4684a5ab762b30e85d132190e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 09:27:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1abe6bc5855d4e44bc5716a984a609a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc8a9ec4edaf4199a2c7ef71c574efb5 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3599851100ec41cc9af825ba3f34f328 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c32abd5d52af4b8692d8e404d7d06022 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b0fd07babea4d4fb13e5f387d9efbde : SUCCESS in 3m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b4100c2db9d7460597c40e1d94d42ed1 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1669a8daa11a49fd87652348f5edb2c6 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92a0dba0dd1449f293060ce52de99b4a : SUCCESS in 3m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/56044f70c6e946d4ac291ee3900c6120 : SUCCESS in 5m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4cccef3c7df742649d13cdbb6489a24a : SUCCESS in 7m 17s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/53aeca76e90649a2afbe850c6294cd38 : SUCCESS in 3m 46s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/6788fda3d05e46a6be0c914af4908d36 : SUCCESS in 8m 03s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/5e49dd5abeb24fa6826d47e56c161944 : SUCCESS in 2m 58s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/dd9561411daa4e699d1e9c2a5f8c6e5d : SUCCESS in 5m 04s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/e13e0c7f4a45484586603c9c48094190 : SUCCESS in 8m 57s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/ed2b44f732724bc6bee30bce000fd012 : SUCCESS in 3m 11s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/5b0a3d1dfa8b4025b9103c4938c06470 : SUCCESS in 4m 59s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/d38131dd5a0f4f14bb6d3e113404ba6b : SUCCESS in 5m 47s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/50d3b2519a3a498a843c5cb73dec00b4 : SUCCESS in 7m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98ae6d8990ea481ab88dc5b6593e5271 : SUCCESS in 6m 27s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/2e75b8e459c042568178ff7d99cf1ddc : SUCCESS in 7m 26s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/cfbfc1fc1bc9461d94a5fd7e65a52366 : SUCCESS in 7m 00s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/dbba8ef195a845b287978d82d7efa1b2 : SUCCESS in 2m 29s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d7fe1456ddb34cb297be22dc996ace44 : SUCCESS in 10m 52s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/052a5df443784c3b9822327f7b24ebff : SUCCESS in 3m 49s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/f0a84a9d7e2b40d7b53ef1ac011d0135 : SUCCESS in 5m 08s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/b22b35cf010548a69bb52b9b3e8f424f : SUCCESS in 10m 39s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/d59bb039ba8f4ffaa6e7239f0a8cbf2f : SUCCESS in 16m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/af5133d662b54056921406b2207bf727 : SUCCESS in 7m 54s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/499962e43da74379932fae343013ca05 : SUCCESS in 10m 46s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/8c925fda3aab40929060eb8e05871a9c : SUCCESS in 4m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1d0df57680334ce1b7331056e8423f9d : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d37759ffbf2450285cab7e0a0b49620 : SUCCESS in 24m 36s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e7990b06414c4400b98234c4b7b4fcab : SUCCESS in 7m 04s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e2ee91770d1f49e2b65f023bfd1a5bb2 : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae59c4fc7c9e43cd9fef66db5b92f3b7 : SUCCESS in 6m 40s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/664a6f9cfdc942f7a8a6358a21e8b524 : SUCCESS in 18m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3f2fe9d0e7e7401e91ff842283a23cea : SUCCESS in 5m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/152c260aaf9b48e2abf2f0d5810db1cf : SUCCESS in 1h 02m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b463b984a324acc9e9bfaadd5b6fb9a : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5e508b630a3f272abe515d812e257b19ec5ef864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-13 20:16:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8f4c934fac4e8dbbc54576f67d1d19455bc89db0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 22:24:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/599411f21b3147f9afd55363f3892b66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4aa1cada1149a1b894ceb2660b8f8b : SUCCESS in 2m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddf0378f22dc4874b8757e7c807ad0fb : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a064a78dede341aba8ddcb26f87985b3 : SUCCESS in 2m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d68c38277e9140788c9ee53d4137e167 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eeaf4528f92648a6836936d62a50a5af : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6fbfc808989b4db98028c1d294aec1de : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0060a06f14444ef9816ea2abae41126d : SUCCESS in 2m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b0c8d655296e465b810ce3249bd94a98 : SUCCESS in 5m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/75e07a5d17544d098be387dc92f459b3 : SUCCESS in 3m 17s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2dc5fb7590684d90bd02ad846adead05 : SUCCESS in 3m 25s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/0d92c6c34c7d42acae69308c651712f7 : SUCCESS in 2m 41s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/d510ef62b7e04eb2a6a2ca05145b0114 : SUCCESS in 4m 42s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/4f9f240876c84ba39dc5045bcb1b29cf : SUCCESS in 5m 01s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/79bf70980c0f4e769898fa6574f92d12 : SUCCESS in 7m 31s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/2b4550ec05fa42d8abe824d255c2a381 : SUCCESS in 4m 36s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/1bc9c5c7292b4da28b786a34565a40d6 : SUCCESS in 4m 40s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/8c206d86c3794a449c0c847c632d5070 : SUCCESS in 5m 01s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/3ff6fb34f495477c86338100652d71e4 : SUCCESS in 13m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/105e6a4f3552465d889d5e99058c3200 : SUCCESS in 5m 45s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/efe5de38736c4b8fa5d4e34903b1f109 : SUCCESS in 7m 54s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/a393546c9591406ab281485bb31b8845 : SUCCESS in 8m 33s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/d007865c77ff4694a8447badfafd4f60 : SUCCESS in 2m 01s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/2744536c08f74729a66ce770fa95080e : SUCCESS in 7m 51s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/b72194fdeba349f48f271974c6e5e0f5 : SUCCESS in 4m 03s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/833d547d74f74b9db5b761970e805151 : SUCCESS in 4m 24s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/dc045ae2da644534a5c1c5cdaa0c75c7 : SUCCESS in 10m 07s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/cd91406c57674908855a340bb5a3643e : SUCCESS in 20m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/125d2d5cb95d4db5a69419a1fabdfc1d : SUCCESS in 6m 08s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/18be241f606f4d4b9806794cbb70fd5b : SUCCESS in 12m 54s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/5e8d17fbc88847a68fded8826276cc2f : SUCCESS in 3m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6641b7ee6c144305b5f0d2aa38e77bd2 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a07e283f3785405fb3eace9e48af25c8 : SUCCESS in 25m 00s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/d98a82380b524022a5f62f08dd600da7 : SUCCESS in 4m 56s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/333144a3a99743d6b5a6a03a94c9785f : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/908a0a21b0b84d7dba9bceaf9ae3cf4a : SUCCESS in 5m 52s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/6de25b651fb94494b121a5ff56d0c149 : SUCCESS in 18m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/02adcfe612fd4af399ae78e62ed445be : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96ffcbc452224105a74eb41901c3fd8f : SUCCESS in 1h 02m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/273ea677702f4b818aac4f54298a7d76 : SUCCESS in 2h 00m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2d811539a79a7f016167fc3fb513f19fc617fa58","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-15 15:22:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b9b324d8b0c3ca8f369eeba4a195a236221ae3f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 15:24:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d7ebc91574477e951eb0596be65dae303b7f2f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 19:13:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faaab590aca94225afdf3ca85cb9a5a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2f5cae9f9f940d0b8a9bc5c366e5eb1 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18e4bf085ad84040b0f9d0e6e065c473 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a8d760aa0be4a58bf3c94d5352db1f0 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89c9b9259218477cab53dc9df1f77ca1 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eec86b816a054230abbe8e17236117fb : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c62996e48e5c46ed82cec125db6c8d0f : SUCCESS in 3m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/465b687cfd494e9db36c91e2d0c5cb12 : SUCCESS in 7m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/28f4ebc8e39942be8f2d1be984716da0 : SUCCESS in 4m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ac735e9200ec432d8e9e9e8aded5cb92 : SUCCESS in 6m 25s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/ebbaeec4d5a94636a04ac934b8b77c3b : SUCCESS in 9m 00s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/66bc5dbfc221497aac69b2e373d2548d : SUCCESS in 4m 30s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/a41451d427484b639a48e0c9d8f01891 : SUCCESS in 4m 49s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/33079f38c0854ef599f2a49ba0c05318 : SUCCESS in 6m 51s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b0261d5532584a6393f431bd8ae2a120 : SUCCESS in 19m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98df85b6ac004fbf94157f385f3c96ec : SUCCESS in 7m 06s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/9bb32ffea1cb497b94d5d8887cbc8520 : SUCCESS in 8m 34s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/bbbc4cbb27eb48f686ba4f32d2fb137f : SUCCESS in 13m 48s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/49c6e7cf1d6d40beb964bc373dac8a0e : SUCCESS in 3m 07s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/172c8f4456d5451b8ffdd205b5f83792 : SUCCESS in 8m 45s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/91d5ac22516440db8bd4f82d67d1bc29 : SUCCESS in 32m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0b2c1cc7f20c4b92892869c88b65e7e2 : SUCCESS in 6m 36s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a34f6b6cdb15463b9503385655e0727d : SUCCESS in 11m 49s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/ad7012b6dcfe4088828e6202d40b0f0b : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3f9a7551d8ac4ab993ee440e8acd6ab3 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/548afb67474a41e8ada4d28138c37648 : SUCCESS in 25m 41s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/a2461df40d0a4982ae6b23970a244e41 : SUCCESS in 5m 52s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5c0be1d1808a48bf8892657fe5787857 : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ad85ccb24e545ed90f41857d8d05e6c : SUCCESS in 5m 32s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/228cebbc4e7c45e78bc1f459070148dd : SUCCESS in 19m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/af42503ad69442e190fa869a124380c7 : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88aa2b8a14e0456f9401baf6311502a1 : SUCCESS in 58m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6d05ac39e26472f9818ccbd6a22ea39 : SUCCESS in 1h 44m 20s","accounts_in_message":[],"_revision_number":2},{"id":"942ca4a905c2446340ec1acf83e18a16d27a5a30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 19:13:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"738259030c74f40879c50f8e4ae9d169d67e05c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 19:17:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c29810a37d24482188827f2a9316450a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe964ac88a043f1bbca85474a00a1e4 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b83fbd73926b84db0c913b943f72c71c1cadc641","revisions":{"88400025bbabc3e98f7f9a7fc1f9713cbc00d7ff":{"kind":"REWORK","_number":1,"created":"2026-06-13 08:19:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/993204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/993204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/993204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/993204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/993204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/993204/1"}}},"commit":{"parents":[{"commit":"3c52a72d47377080503df780bf96c3ef778bd1dd","subject":"Merge \"Insist on minimal pyproject.toml file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c52a72d47377080503df780bf96c3ef778bd1dd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-13 08:19:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-13 08:19:09.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iccaa9553d658295165cf678cec2fae94519b06ff\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88400025bbabc3e98f7f9a7fc1f9713cbc00d7ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88400025bbabc3e98f7f9a7fc1f9713cbc00d7ff"}]},"branch":"refs/heads/master"},"b83fbd73926b84db0c913b943f72c71c1cadc641":{"kind":"REWORK","_number":2,"created":"2026-06-13 20:16:47.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/04/993204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/993204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/993204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/993204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/993204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/993204/2"}}},"commit":{"parents":[{"commit":"3c52a72d47377080503df780bf96c3ef778bd1dd","subject":"Merge \"Insist on minimal pyproject.toml file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c52a72d47377080503df780bf96c3ef778bd1dd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-13 08:19:09.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-13 20:15:42.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Iccaa9553d658295165cf678cec2fae94519b06ff\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b83fbd73926b84db0c913b943f72c71c1cadc641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b83fbd73926b84db0c913b943f72c71c1cadc641"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
