)]}'
{"id":"openstack%2Frequirements~993722","triplet_id":"openstack%2Frequirements~master~I39ba20ac333c5334417261081adc032d7fa7e352","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I39ba20ac333c5334417261081adc032d7fa7e352","subject":"update constraint for openstacksdk to new release 4.16.0","status":"MERGED","created":"2026-06-17 09:01:35.000000000","updated":"2026-06-17 13:05:25.000000000","submitted":"2026-06-17 13:03:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993722-new-release","meta_rev_id":"59d5eb75fd27d1fad7349db0a9d8499cc1e79bf5","_number":993722,"virtual_id_number":993722,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-17 13:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-06-17 11:08:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-06-17 09:35:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-06-17 11:08:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-17 09:35:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-06-17 10:34:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-17 11:08:45.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1c5cd90cf2c56faf0a0615c71da31b09137c94dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-17 09:01:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35f8b89294ca5785a51ecdbd1088e502d1b03aa4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-17 09:35:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3aeeed30754a7d70755f80ffd40309da2f91a450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 10:34:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e15abdede60447ea942e8d748093bbe3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1579c235357f4d879a86ae55d3235936 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d10eb95c18e4350af31742774e020e8 : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51839dcd3caf41e6aea6f0a90798fa5b : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5b4b2663eb94c3e8535f8021d1635ee : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b53d9ca264624102ae5a4b34baebcc2a : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6693e1002e34e759dc67e564956149c : SUCCESS in 2m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbebb80a82394145a097400e07681b23 : SUCCESS in 2m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3a9ffc06bbf04d31acb20af6541e292a : SUCCESS in 3m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d0c52efa1e45424ea1368bdc713ed104 : SUCCESS in 3m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5eb2cdc7dee04d7a89e4c40882cf1eb3 : SUCCESS in 3m 04s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/b464632668884301997b76669f51b067 : SUCCESS in 4m 44s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/04a6d52a7dcf419dbc128223da99fd5e : SUCCESS in 3m 44s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/5c91d3c389af4487a3e2a97efe154098 : SUCCESS in 3m 06s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/0d5679e2e6b1442e9f5c6137114242ee : SUCCESS in 10m 36s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/9c5bdaf802a34cf4887b3ea2e84e793a : SUCCESS in 3m 41s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/de8d4670a20b4490bcdd20bd2b9bb93b : SUCCESS in 3m 03s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/5ef646cffd6e439586c2f95473b2a123 : SUCCESS in 4m 05s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/cafd7526e6144565b06ef262c1c6c929 : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3886da0941b14d89b94e3d7665a01a2f : SUCCESS in 6m 37s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/66dce17e0283492a92df7f9d6f853cf3 : SUCCESS in 5m 45s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/9bbcb1c608cf473e9146ed7198cc6935 : SUCCESS in 7m 05s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/d47aea6de39746c888007c034168fa11 : SUCCESS in 1m 46s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d698a18b850e48ba8ed799168e9fa867 : SUCCESS in 4m 38s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/21750d14024a44c0a696d091ee9839fa : SUCCESS in 4m 19s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/5af5ca77cbcb463a8f5cf9ab7dee62e8 : SUCCESS in 2m 59s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/2c530b6ca3bb4cb7bc58bdaa714e5a63 : SUCCESS in 9m 35s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/6500acd7ef774d5c867c514620cb2687 : SUCCESS in 16m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e1e76ce39606412681fd8522be8636fa : SUCCESS in 3m 55s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/29d444eae6f84500a6fd923cd9e0713f : SUCCESS in 8m 13s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/fb5362e44b184331aaf77349dd4d9d05 : SUCCESS in 3m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/590895be22204c37994e3d4674fe4b56 : SUCCESS in 2m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2fce8f724f1445479f95b0cca23ecf5c : SUCCESS in 23m 02s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/1bc648bd05bb4f4fa7835bdb7c15ef1d : SUCCESS in 4m 52s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/64bf296a40be49a39dddb3f43569e0e7 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/840d344b58b745c0b46516dae09f2bf5 : SUCCESS in 4m 06s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/15869972e9c74f37932c59194f13864a : SUCCESS in 16m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e351e9d8bf5c453f84462271c437b28e : SUCCESS in 4m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9569860f5dce408791a363df6d5edbb3 : SUCCESS in 34m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b835af04b15f4a4f9372ad676793ba5a : SUCCESS in 1h 30m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1549f590223dc3e66bda33a2f1d071334eb545de","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-17 11:08:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31710ed9568b75df43a7afda30c9c0ac6b33e00c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 11:10:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6073258585267a9d44f2ebaf5b5c232a06a2c3d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 13:03:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c51db9af2e214f28858409e92fc0e8b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f65dd7d32c4747a83a530012913d5e : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/247b024561844b2f915cae7b20b10466 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/824e6e6a7e364847a261d8dad7cc76c3 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21c630bc148b455688c70ca0f8948600 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65669d3db02444ed908c2af2bdd484ae : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4cb5fe6f3af4dc4934fc1f9c3e4858f : SUCCESS in 3m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8fa8f99b396a4d6e8077ebd122a0a606 : SUCCESS in 6m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b52388ef24f644038abc77f4de6b967c : SUCCESS in 10m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f795f10001094c7c909c347d4c763664 : SUCCESS in 5m 33s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/4d526680f7744d85837853e5c496274f : SUCCESS in 10m 43s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/46a0653774eb48f385eeec262aa7cf8e : SUCCESS in 7m 32s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/f1561f3a48ee48398a9cbcfe6359a16f : SUCCESS in 3m 46s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/19a88b9438384459a89b409c2afa2afe : SUCCESS in 5m 25s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5e46ae9742d2481eadaf0450e7b8d847 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c31ee1801d047c79b048ec240cbc2c0 : SUCCESS in 4m 19s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/52913d0bce674afaac3bc00df2557c92 : SUCCESS in 6m 22s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/930971b7ba0441e192066b037ae025e8 : SUCCESS in 11m 13s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7f136c055d4a40889c7a0774406a7dd7 : SUCCESS in 4m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/cef5969f5dbe4eb7a96656962e2afe6a : SUCCESS in 9m 31s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/0d9a3674f0ac4a259ee7df0fb555a942 : SUCCESS in 24m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a74afe26c66d4d06a9cc7117e47473f6 : SUCCESS in 6m 19s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/8095b3e5fd42415bac53d45e077e1d2c : SUCCESS in 14m 12s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1985aef325b34d5a99eac8bdbca5b5b6 : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dc637ff3001044718fbf235f8229de4f : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ee4b4f051c954eafa20939cb3f5ddaed : SUCCESS in 24m 17s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/5145ea9aa2974eb7a3617bf77b891828 : SUCCESS in 5m 32s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/4abbe1557ee340fdb51b53435e96bd5d : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fffdd78821f4bdfa0571d9d1b5d9093 : SUCCESS in 7m 43s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a2ea8fd075cc469481052a15ddfee75d : SUCCESS in 18m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b45d3694cb0c47d48851056a0171cf03 : SUCCESS in 5m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40b993a341274c4a8a2a6546ab24c27b : SUCCESS in 1h 02m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2bb24aa76754f7f944dd3fa7ad0426d : SUCCESS in 1h 48m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f31cadc550f2621a01555a9f74a8b93c2d3a058e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 13:03:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"59d5eb75fd27d1fad7349db0a9d8499cc1e79bf5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 13:05:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d306cac0e9f45a1ad70279ba05cf0da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e01c8a47dd490198e6b3a40e54a53b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72d979fb9480ad935ba5110ad9fe91019144df23","revisions":{"72d979fb9480ad935ba5110ad9fe91019144df23":{"kind":"REWORK","_number":1,"created":"2026-06-17 09:01:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/993722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/993722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/993722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/993722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/993722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/993722/1"}}},"commit":{"parents":[{"commit":"b83fbd73926b84db0c913b943f72c71c1cadc641","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b83fbd73926b84db0c913b943f72c71c1cadc641"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-17 08:51:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-17 08:51:33.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 4.16.0","message":"update constraint for openstacksdk to new release 4.16.0\n\nmeta: version: 4.16.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ibc3d9c7113f4ed9b53a0767bc85db1a969324ec6\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I39ba20ac333c5334417261081adc032d7fa7e352\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72d979fb9480ad935ba5110ad9fe91019144df23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72d979fb9480ad935ba5110ad9fe91019144df23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
