)]}'
{"id":"openstack%2Frequirements~994610","triplet_id":"openstack%2Frequirements~master~I741ee6f4a3b716309017a96830cf8325065eba4d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I741ee6f4a3b716309017a96830cf8325065eba4d","subject":"update constraint for openstacksdk to new release 4.17.0","status":"MERGED","created":"2026-06-24 09:37:50.000000000","updated":"2026-06-24 14:07:16.000000000","submitted":"2026-06-24 14:03:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994610-new-release","meta_rev_id":"b4dc4da6e7da036c631eaa60207ff60c95b6a162","_number":994610,"virtual_id_number":994610,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-24 14:03:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-06-24 12:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-06-24 12:32:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 11:26:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 12:32:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"65d36c1476480880bad657c70c34df8add26b641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-24 09:37:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f2ae8fd90ac5eb544e0dd687f4f0d754948e958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 11:26:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d3adac11001469aa937bcda07188380\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65bdc88d0725469c9f69838c3d63fd16 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af5a058eb5149b787c70476eef36786 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b57fb38306c4e0fb2a9c412069b3e45 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33e494f0ca0f4925b2b256a7407d05e0 : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b47edce71ea64d2985e25cba03ba8d6f : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7cd0332a269a4e50b697d118c08753f5 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/493fd9e4792e4b4fbd0a816a7aa0e546 : SUCCESS in 3m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/946b636672b64844845a4de88452e9d0 : SUCCESS in 9m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e750c479a00446eb9129c0babae8963e : SUCCESS in 5m 46s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/73c2324621b64112b6dfa0ed3bea76ba : SUCCESS in 3m 50s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/fd88683acaf04b5d9450c890073cacf2 : SUCCESS in 5m 40s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/ffd518a6d1064881a67f626c9803e949 : SUCCESS in 4m 35s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/ba9649cc65284ca78434c226abaf8e46 : SUCCESS in 4m 34s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/25e717d91406470ab45c0763a08276d7 : SUCCESS in 9m 50s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/736dd6f1f24b482590f7586abbf9ff74 : SUCCESS in 4m 38s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/742e5dccbc774156a605d09738b633f9 : SUCCESS in 3m 13s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/f02147fea4c4456195bd0c7017f3d192 : SUCCESS in 6m 05s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/073cc72d1fc244ae96582776916bd076 : SUCCESS in 17m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72ad1f82e99d4cd8954680ca740bb733 : SUCCESS in 5m 41s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/58a11a89578b43a0a85d8416b538000e : SUCCESS in 9m 22s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/b3911940846a41c4a0c6ec8da6575b92 : SUCCESS in 9m 59s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7532ead3949547458328e6a1da365aa9 : SUCCESS in 4m 13s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/a9a5b273bdf24423bfc8a9a05aeadaff : SUCCESS in 7m 40s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/91e8a3acdfaf47ee9c9c921a5be6eb87 : SUCCESS in 4m 18s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/2e78621e790443d6bb36d4ff9a8762e9 : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/4f5d7311a52e412982dbb835280335b3 : SUCCESS in 10m 24s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4fe6fa3533e74ab7b6dd21caae522761 : SUCCESS in 24m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8d9bb127ce534df1ab5442095c36f18c : SUCCESS in 6m 09s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/e880ccc50e954f7f932038e085847420 : SUCCESS in 14m 04s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/2b098e78ffba4f05b6eda0156d7b3867 : SUCCESS in 2m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c73ef1bcffd2432db6636e20c9b9b8df : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c7a5193fcff4550b9e5ef831ecc9b6f : SUCCESS in 25m 51s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/58b16fe65cf84e7f9453e5605d340f47 : SUCCESS in 4m 33s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/a69dee6b884c49a7a632461f9651be21 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f04243a86667418eaf44fd9162f46b61 : SUCCESS in 6m 13s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/10183f1714384375b446713055e34b10 : SUCCESS in 18m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/95900f3afe84439cbba8211b63e359e6 : SUCCESS in 5m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07de736b076c40be95cb578c939290b9 : SUCCESS in 57m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51c2098148914ec7a845ad8bc1cd2d78 : SUCCESS in 1h 41m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3b44a1b703ff429eebe80b755c274d04f3f360d3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-24 12:32:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"046cc366226bbabab79dc50476b579f7cd80ea3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 12:32:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06f8bb4fab6937250c8678fb748f392610d20ea7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:03:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d1e43339eed4cdca592dcf22888d67e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07e4963f6f1405b91205108332a1a08 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a52052c62fc2490ea912b42b9c63b6cc : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e37f2d8a8d094276b7108dbd9981a02a : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ea04880678846fbb16a1c7f32dc0693 : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc3bf8d9d79d48569404e04bf594a3af : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c6551d3ebe7b453388b4cb8af5a3b989 : SUCCESS in 2m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b7221b1b011049f2a9f969e738706978 : SUCCESS in 6m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8b1181805388430a88d63f2a5eb8d514 : SUCCESS in 4m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4fe60b8b013f491d89e17eb9b17ad4d2 : SUCCESS in 10m 32s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/fb006101e3ec4aa091a3f66ae83fd089 : SUCCESS in 7m 11s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/6d4550b99b1d4fc6bf84d44996dce246 : SUCCESS in 8m 04s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/38c88f8d9bc34be0a142e23b892d5e74 : SUCCESS in 4m 46s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/04e7cd1c09134dbfb50aaef2109c6608 : SUCCESS in 4m 52s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5553b20feb11427d9915c5a4d3514b25 : SUCCESS in 7m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29b8cb7e607540f5b9e17d22e5258a5c : SUCCESS in 7m 34s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/9971a37b81f6433fb49328d3c6e4c873 : SUCCESS in 7m 01s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e012002a8b714a0ab50a25dd2076528c : SUCCESS in 11m 01s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7d5735fa539d456a808e28187280b336 : SUCCESS in 2m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/36b05c87ee5147169a75825ac6cb4b6b : SUCCESS in 5m 12s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/3e7be4b8104e4bdd87ee173e26310c7e : SUCCESS in 26m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/38bb144dc1f84b33bd8080c777dae6e8 : SUCCESS in 6m 33s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/fc05d0ab606a4c24b227ee06baeb7751 : SUCCESS in 16m 40s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/d65e0b727ec24d7993cc6bfb076dd36e : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/408e10efc2a64ea2b41c2f37eed17894 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9036bfa0de0240f282785b5f7dd44874 : SUCCESS in 25m 31s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/76cfe2daddd14980a9e7ab89173b929e : SUCCESS in 3m 41s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/2531583ea68f4427a10df269263dd241 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c42c5a8a33684d29845a5842a8422df3 : SUCCESS in 6m 24s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/7ea87f2bf1dc4033819dfcbf29404b50 : SUCCESS in 20m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/365c749755ba48e89aa1098397a3211c : SUCCESS in 5m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88d5603d6b7c4a8c9a202cd36ce38062 : SUCCESS in 34m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/213a7bec5a9a4e1ba8d2ca97d75dab84 : SUCCESS in 57m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6f9f72fd76eacac7bec303dc132223fb4e53e7e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:03:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b4dc4da6e7da036c631eaa60207ff60c95b6a162","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:07:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3ae0a939cd24b03b061357cbc60b697\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/815ec29c38724b3e890dd951c532d99c : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e1b7265f0ccccdfc3b82d8ef87d239503f9bb74","revisions":{"9e1b7265f0ccccdfc3b82d8ef87d239503f9bb74":{"kind":"REWORK","_number":1,"created":"2026-06-24 09:37:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/10/994610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/994610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/994610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/994610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/994610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/994610/1"}}},"commit":{"parents":[{"commit":"f84d342ae1587d9d3898839af306b76f40eb6a19","subject":"Merge \"Add hypercorn\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f84d342ae1587d9d3898839af306b76f40eb6a19"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-24 09:27:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-24 09:27:48.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 4.17.0","message":"update constraint for openstacksdk to new release 4.17.0\n\nmeta: version: 4.17.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ib2298a038396d43d8e68f7b48ca405f62ea0d34b\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I741ee6f4a3b716309017a96830cf8325065eba4d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e1b7265f0ccccdfc3b82d8ef87d239503f9bb74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e1b7265f0ccccdfc3b82d8ef87d239503f9bb74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
