)]}'
{"id":"openstack%2Frequirements~994857","triplet_id":"openstack%2Frequirements~stable%2F2026.1~I21d10395d23b66233aad6aaa30a6182b430dc4ed","project":"openstack/requirements","branch":"stable/2026.1","topic":"new-release","hashtags":[],"change_id":"I21d10395d23b66233aad6aaa30a6182b430dc4ed","subject":"update constraint for os-ken to new release 4.1.2","status":"MERGED","created":"2026-06-25 08:05:20.000000000","updated":"2026-06-25 11:02:31.000000000","submitted":"2026-06-25 11:02:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994857-new-release","meta_rev_id":"6332f1cf7ac7876b277227f326bdbb73801f0288","_number":994857,"virtual_id_number":994857,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 11:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":39262,"name":"Rafal Zielinski","display_name":"sq4ind","email":"sq4ind@gmail.com","username":"sq4ind"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-06-25 09:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-25 08:08:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":39262,"name":"Rafal Zielinski","display_name":"sq4ind","email":"sq4ind@gmail.com","username":"sq4ind"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-06-25 09:06:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":39262,"name":"Rafal Zielinski","display_name":"sq4ind","email":"sq4ind@gmail.com","username":"sq4ind"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":39262,"name":"Rafal Zielinski","display_name":"sq4ind","email":"sq4ind@gmail.com","username":"sq4ind"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-25 08:08:58.000000000","updated_by":{"_account_id":39262,"name":"Rafal Zielinski","display_name":"sq4ind","email":"sq4ind@gmail.com","username":"sq4ind"},"reviewer":{"_account_id":39262,"name":"Rafal Zielinski","display_name":"sq4ind","email":"sq4ind@gmail.com","username":"sq4ind"},"state":"REVIEWER"},{"updated":"2026-06-25 09:03:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-25 09:06:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"6485774b3d0ee3334a3468511b8fb902e918e7f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-25 08:05:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3484e62f0257162e439067fed07d751ae69a4ec","author":{"_account_id":39262,"name":"Rafal Zielinski","display_name":"sq4ind","email":"sq4ind@gmail.com","username":"sq4ind"},"date":"2026-06-25 08:08:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"05986dce511ead9b14096a3138fe3d9c75567658","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 09:03:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30e9d81904464963b4275817986fc53a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6078b3929bcb440a9229d9bf7a302de6 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e250813968d4bbe95b8ead61deadf33 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0122d7ab18fa47f9806db605b120c3c3 : SUCCESS in 5m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a3f350e46cbf46a5807d2790c2cfecab : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5505471fe4194dfe86d566c1a077784e : SUCCESS in 2m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d17f207cf7f747aa812c58abf5add914 : SUCCESS in 2m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/587bb497fb6b4ae2b1282334d1d5c8a4 : SUCCESS in 6m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b2da67308cbb45329c65b01700238d06 : SUCCESS in 6m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/98ee47166a574ed5b0e42623e5bbecd3 : SUCCESS in 6m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/670e0ad6a9564854af1b6f9bf402b7b1 : SUCCESS in 3m 07s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/76577b52d3be4a25a580d62bc3a16604 : SUCCESS in 5m 05s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e069d5dbf9b74a558e1356a5a3da288a : SUCCESS in 5m 12s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/64277d58c5b34a9f996b52a54d457517 : SUCCESS in 4m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0459bd21406b4e5db25702ea85b01aab : SUCCESS in 11m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/005c6e8985f44349b35a43915abf8c64 : SUCCESS in 4m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/67d430cf6d494552a401b04b6fe26132 : SUCCESS in 5m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ae3551decedf4889920213dc3e245213 : SUCCESS in 8m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b4b5e958419a431ab2cf38967c4061d4 : SUCCESS in 8m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40f7dab80f6c46f3af2d402deb0bb885 : SUCCESS in 6m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/13a2b1011a8f4851bee08d98955d451d : SUCCESS in 9m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4e2ba7ae4e4f4397917c4a8edfa60a0f : SUCCESS in 7m 41s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8c4941b41289429fa48e65372a012052 : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/76e1f77b8e604ced9bb62d1c9322490b : SUCCESS in 5m 17s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3eec74d159714ff582b5463fc9007988 : SUCCESS in 2m 54s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3fa889f6c89f42568664aa6658e3ad32 : SUCCESS in 3m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/92200b7fe71a40e38ff16fb84bf22969 : SUCCESS in 11m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/36487e7e20214851a36a5dcaaaa3af2a : SUCCESS in 22m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5e80e54102f84e9fbed35cfc15da5071 : SUCCESS in 4m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8bacc21a19794d75afe55a4c01b18fd0 : SUCCESS in 11m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/641fb30a9b694ae6848ce5f79955c87c : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/743c53ed80234cb4a4f4da08e3a018c6 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c642ff3c3b8f4edc876f3e733dfc8cb5 : SUCCESS in 27m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/aeaaea3f6ed64d90b57bee3414f50d61 : SUCCESS in 4m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3685ab02af054110b4ef396638f2abcc : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3940c5f056f2442fab0b21ccbe81d490 : SUCCESS in 5m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e48311917cbc425db749ce1e16606bc3 : SUCCESS in 7m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6202da27e0aa4354988685655b3e0a33 : SUCCESS in 6m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59dbe7f41a3b484db1e2a95f9299c6c9 : SUCCESS in 54m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9b2cae789bc4c24918d2fcb7cf3712f : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7822498426aff7e123c01fbcdb7d6605c69b0d45","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-25 09:06:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1765502d3cce9384c4fca6761e2855c3497eb777","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 09:07:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c72ed9f7212e49d4d19696355a53a62384667984","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:02:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccc349466e3240ee8e700e118452b3fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b201008832cb4b629b9438092cd862f1 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0253bf2efd944976986f0b1a97290ab0 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f627f742004c4e2281390c190b4d9db2 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5aa38058ea2543759fc973805456fbe1 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b9a76662e0a402192d0637ebf162fa6 : SUCCESS in 2m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d65dd9af57f541c09415ca27ceee3402 : SUCCESS in 2m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/23bcfd445453431b8be9a459cefbb6d7 : SUCCESS in 3m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e9df8c649be6416bafea85ceccce6df6 : SUCCESS in 8m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ec4f0d4cc60742c8abc18154c2f6457d : SUCCESS in 3m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/402b0cdbff3b49fdb6d7585dca7c77db : SUCCESS in 7m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/881c7aece1cf493a9cb3c1392f1c136e : SUCCESS in 5m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bb54a4f57a4249bfbefd8955075a66ed : SUCCESS in 6m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f0823812e54e451ebdee6de23a035360 : SUCCESS in 4m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a4253ef4d0aa45f28d5e01a63c0ff63b : SUCCESS in 18m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a252fb2b2d9b49ac993cc886465a4ebd : SUCCESS in 7m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4574ac9654694a9d90ea93d4f27b1eba : SUCCESS in 7m 23s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9ff4165781a040529999b849da0155c7 : SUCCESS in 12m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e44ee2b5deb341788eca8e3c0532a9db : SUCCESS in 2m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/716de62fd03f4bd3a1f66d950baef9a7 : SUCCESS in 11m 01s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2cb97a81ce504d1ba9d0fa916e648229 : SUCCESS in 20m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/817ecb9d08d747f7a97c5fd4d31156e5 : SUCCESS in 7m 36s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b769aa3580ef40e5ad3875e0bd53db42 : SUCCESS in 10m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a115452d3db84c28bd1e316a582b32ca : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e6f40d6c57e34cd9bcf580a857175f1d : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a0348b4644c24230b28453cec8a5cc67 : SUCCESS in 27m 28s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/53f284937caa4b099b40a557b2d50a48 : SUCCESS in 3m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6d2aa743c81a4fb8ba482d724b339bdf : SUCCESS in 3m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/768431bb29ec4ff3a09a9db6aa289e09 : SUCCESS in 8m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3503a3a8487e4ede92a84846a54d6746 : SUCCESS in 18m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5485df186d68438d8a854720d8122c3e : SUCCESS in 6m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/625d832217044194a1714a8777d43ae1 : SUCCESS in 55m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/450a4765df4744b085e480e09794a317 : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6332f1cf7ac7876b277227f326bdbb73801f0288","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:02:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06cd4e8523cbade25fb93efc4f8ea77d6d97064f","revisions":{"06cd4e8523cbade25fb93efc4f8ea77d6d97064f":{"kind":"REWORK","_number":1,"created":"2026-06-25 08:05:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/994857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/994857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/994857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/994857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/994857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/994857/1"}}},"commit":{"parents":[{"commit":"c4c55d5279d824dc261a43ac51b56146ccc4dd4f","subject":"Merge \"update constraint for horizon to new release 25.7.3\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4c55d5279d824dc261a43ac51b56146ccc4dd4f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-25 07:55:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-25 07:55:19.000000000","tz":0},"subject":"update constraint for os-ken to new release 4.1.2","message":"update constraint for os-ken to new release 4.1.2\n\nmeta: version: 4.1.2\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: stable/2026.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rafal Zielinski \u003csq4ind@gmail.com\u003e\nmeta: release:Commit: Rafal Zielinski \u003csq4ind@gmail.com\u003e\nmeta: release:Change-Id: I88ecf9b168bb45fcf78b7234e1a9c05f10ff4c6f\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I21d10395d23b66233aad6aaa30a6182b430dc4ed\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06cd4e8523cbade25fb93efc4f8ea77d6d97064f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06cd4e8523cbade25fb93efc4f8ea77d6d97064f"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":39262,"name":"Rafal Zielinski","display_name":"sq4ind","email":"sq4ind@gmail.com","username":"sq4ind"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
