)]}'
{"id":"openstack%2Frequirements~994865","triplet_id":"openstack%2Frequirements~master~Id58f90ee8f5379f2369db53cd0ffd9d9ebd290a4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id58f90ee8f5379f2369db53cd0ffd9d9ebd290a4","subject":"update constraint for networking-generic-switch to new release 10.0.0","status":"MERGED","created":"2026-06-25 09:24:10.000000000","updated":"2026-06-28 11:09:32.000000000","submitted":"2026-06-28 11:08:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994865-new-release","meta_rev_id":"c2f9a3ba98275431a6b88cc13df0cd18b4e11579","_number":994865,"virtual_id_number":994865,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-28 11:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-06-28 08:38:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-06-28 08:38:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-25 10:56:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-28 08:36:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-06-28 08:38:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"e1ec0b87250c258b90e649a082fe030aac3d6bca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-25 09:24:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25be54f84dab2cb7de4fbd6e31e0943373a8e2b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 10:56:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/91d1e48163434a5d8a05e89c71772dcf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4e1f3315e324465a45697b4a5c5bc29 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42987898e7264e668c78dde7f57132df : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4493c187d734ae9beeff9ec22622b4a : SUCCESS in 7m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8280aea231d4fd1bfa6423d2506337f : SUCCESS in 5m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1804d29e2b134a458c6b80bcae97ca03 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48a112fea0f94760a9a57fb803d853cc : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/25f26492db9a42f4b7314daca857f18e : SUCCESS in 3m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/669c6503ce98472795014ad596ff0714 : SUCCESS in 5m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bf0938468f0040a09c6b888dd895bce8 : SUCCESS in 3m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cb748ff5a9e54e14a83c7710f3a0ae6a : SUCCESS in 5m 48s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/b0d9e1d481394e16a9ac049991c4e7f2 : SUCCESS in 3m 02s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/b657b2a64214427f9ae7651bca39bfe1 : SUCCESS in 5m 01s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/c6a5951c466843eb91c920627fa35625 : SUCCESS in 5m 08s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/9ba5e0441e1f40e6af125758ca60d837 : SUCCESS in 10m 26s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/ca55ce50d1c94454b0ae13ce33669193 : SUCCESS in 4m 38s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/0c3ad4f6653a45bfa5f20b07740c332b : SUCCESS in 5m 31s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/7c7314fb9c26414cb6f605c58734a32b : SUCCESS in 6m 51s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/01898ff6996045e19cee86847a53da85 : SUCCESS in 18m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f11bac31e7d34f99a5d9ec65d5aa0bd0 : SUCCESS in 6m 03s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/f7abd2bd090c4159a33d97823e006611 : SUCCESS in 7m 20s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/decac20dcde444f599accdfc69e7cd14 : SUCCESS in 11m 08s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e33f04fd63c74b96accb834f994a1a13 : SUCCESS in 2m 30s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/a049b0e302834346a55983807dc73680 : SUCCESS in 10m 03s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/a9bdd1175b434a9591d29b3c721b17a2 : SUCCESS in 4m 10s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/870ef5ed862d474f8d881c3de7598cb4 : SUCCESS in 4m 58s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/d3985fe2e041416f9adb5b5380a502c1 : SUCCESS in 10m 36s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/32269fbaffe143109932faa05d8d1258 : FAILURE in 24m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/27f2af63468e4e9aa6af9818fa7f8089 : SUCCESS in 6m 57s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b94e4f77ce884bbbbf9243475d611ec2 : SUCCESS in 13m 57s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/f6bb686e0d16463f850020639897b979 : SUCCESS in 3m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5163970217c34ce69f7335133e1fceca : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89a141e999394f889b4265e2cf13336a : SUCCESS in 23m 47s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/efb922fc63494314875a690c96f66049 : SUCCESS in 3m 39s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f042dc2b9e0244ed9e3f5ccfb5049d94 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cfcf3fcbee004a08be83410cddafbda8 : SUCCESS in 6m 46s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/74ad30caf80548dc9f6c448165686bab : SUCCESS in 18m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/301d113ca7f3466598a2a99970192ad7 : SUCCESS in 5m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eeea14f2799f4172bd3f159c24d55c8b : SUCCESS in 35m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6d2761b24be44b68684abaefc38d183 : SUCCESS in 1h 30m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4d0d479edccb145019e9395eb4d74d26eb648ae7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-28 08:36:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c72660a06655722354b729ffb44ff6b7b23b3a9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-28 08:38:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7d5dc4312cc093c1e89804a5ae1fd6f784a32eb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-28 08:38:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb018f7c9bcb479ac9073eb15009e92264f63d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 09:49:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acbcb209c12a4a6cb5ce0aacc54da4df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b20ccfd7d824ff987fb67ea4cabb37a : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dac4fed16244bb8b1b7eb3bea68f207 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a103ebb082ba42d895e7684da8ce1215 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4e1d97c46154649b28416b2aed21ee8 : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe3f19badea54915990b02e8a4a01b1f : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/edd5e5db5c8e4bc0bac47eaf74e829bd : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e34dec55d3a4aeea03937ae091e62c2 : SUCCESS in 2m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1b6abf191fe64650802d9ca7c6a9729c : SUCCESS in 5m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8fcadb033684423797376ec139e04a64 : SUCCESS in 5m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7c10a3aad501431f985cb27f9dc73630 : SUCCESS in 5m 07s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/cdf992e46d0c4aceb122aba15adeb4a0 : SUCCESS in 4m 45s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/038b9a2074f24fed8ba80fc10eaf668f : SUCCESS in 4m 15s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/83710744bc524456a06ed468dd27f520 : SUCCESS in 4m 53s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/68278e69191a456ba82f9584570d5b67 : SUCCESS in 10m 06s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/70f4bdcfcc164288aec85853f064c76a : SUCCESS in 4m 37s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/3bc1221d197b48cf8cee03aa0640b854 : SUCCESS in 5m 24s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/beea88854dd94f59a080176773842c81 : SUCCESS in 4m 09s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/60dc2ca16c654d53a9bdda608ac09f29 : SUCCESS in 17m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1f299a95be14592ad0ead09e23b8aec : SUCCESS in 7m 06s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/89f4b79c11b440868fb4b65991505c6d : SUCCESS in 7m 57s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/eedd751085dc4420a57c700035cb680e : SUCCESS in 10m 20s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/1506069fd7d54634805893d968f23cf8 : SUCCESS in 2m 36s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/4bbba35e21b64ecf8fb256a68ef34b2a : SUCCESS in 8m 14s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/0d2356d0761e42529e92fe53d8aa039b : SUCCESS in 3m 38s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/f59dc79ddbc1429fb2a78ea8838d7773 : SUCCESS in 4m 44s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/04bcb3c3f2d14b13a465be95ece72b89 : SUCCESS in 10m 31s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/b897be558bd44cfda9317058bac31d5b : SUCCESS in 25m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1e1123b70247487882d769b6bec0b64d : SUCCESS in 5m 52s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/98eab167a99944faa5c43baf8b601691 : SUCCESS in 9m 26s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/acb30db66cef45dcba1260aaa6c6d1fc : SUCCESS in 4m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3d71fe1b22fe4bed9fbab16fc46d927e : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce7df89250004e688c2d87680e0188d5 : SUCCESS in 24m 47s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/d4949ea1273549a1909134cbacb66047 : SUCCESS in 5m 11s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/182a6405d56b4d50aba427e94dc1d80d : SUCCESS in 3m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/51fa6555ecc84440889423bcdb761fce : SUCCESS in 7m 22s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/863feee1059242d8b79c1872df1d1fac : SUCCESS in 17m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/029a71fea7484800bfb7ad8d6d35a546 : SUCCESS in 3m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d814ff45e7b14267b391c90e1d82a359 : SUCCESS in 1h 07m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a593ffdde234fefb032d4227314e38e : SUCCESS in 54m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4227d64faae7d96296b9e86db6ee2cc44509ef16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 09:50:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9228812c079d224051223ee9d598c4b3dd976957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 11:08:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b72933ec0e4a4dd69ab57d88e56726d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d985337ac6441592c88d3226a750ad : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61085b91f2f14b4bbb82c9c4436306e1 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7801990a0c3b4226a61924c2965e6fcb : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c95431123d0f46fd8e5e191521bcf6f4 : SUCCESS in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73c205e02faf45eebe6023f913f8a52c : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/29d8c5bd37cb440c8d6160c8e0eb9f9a : SUCCESS in 2m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7fcf611a594a4739873d750a796f8b3f : SUCCESS in 5m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/061468327c9647a7b0cbc78c8bbd7eef : SUCCESS in 6m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7567605cd0d5425a9457a0058423cef6 : SUCCESS in 5m 02s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/f95f8204c74f42c1b46f18f7ec0e51d4 : SUCCESS in 10m 23s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/5750c7face4c4037a4691d4c2f884733 : SUCCESS in 4m 10s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d92c0aaf92194d50b8f73cc1867d91a4 : SUCCESS in 4m 28s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/b470df9f3fbf4156a2f89b3b2ce1ddc7 : SUCCESS in 5m 40s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b9d0a5cb100e4c32b1591011fbe17da5 : SUCCESS in 17m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ba99770b8e2494c9ccf5abb2bc1300e : SUCCESS in 6m 44s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/59db774328ce4f38b781f4099791907e : SUCCESS in 8m 11s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/9dc87c57adc64206b6b330dc3eb905dd : SUCCESS in 6m 43s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c3cd788610de47b9828d3db5ba098228 : SUCCESS in 2m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/b377a1786eea46c68d9966e029e879cd : SUCCESS in 6m 27s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/3f6b85a3b3104c28b7709be0cc6db892 : SUCCESS in 21m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d986a43372cb4755b58a6f9fe2d873cd : SUCCESS in 3m 24s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/dca1fd7751154417827128f8bda1df2b : SUCCESS in 12m 42s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/4b601761cfc64eecb69c0fea3968ba7c : SUCCESS in 2m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8460ec8881764d05b2bb4816db90ad39 : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5094c20771554c4c9830b220466d7f82 : SUCCESS in 25m 54s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/1d8f1a65076e48a99d4cabddccf6a854 : SUCCESS in 4m 56s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/dba9c32b732a4cd78d4bfda7855806a2 : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0aa126df95c84c2db6880db3440cd215 : SUCCESS in 6m 04s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/56adef8e871d484b96dfe7f4bf1c012c : SUCCESS in 18m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e8d2ce8899404b4e83ebca6cfa66ddaf : SUCCESS in 5m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5d8a310bfc64b799665bdae570c4354 : SUCCESS in 34m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b89db75632a4f76952b91af194ddc9f : SUCCESS in 1h 16m 18s","accounts_in_message":[],"_revision_number":1},{"id":"090fdbb012ec3f737fec069ee49d1c92741c54ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 11:08:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c2f9a3ba98275431a6b88cc13df0cd18b4e11579","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 11:09:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55b3d52d4cc14afda9a0b5efb967d849\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed69be1bff0245a2b2164613f625e2f0 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18c1fecd7e7cc6e0c9a3826a70612ece1e4ba203","revisions":{"18c1fecd7e7cc6e0c9a3826a70612ece1e4ba203":{"kind":"REWORK","_number":1,"created":"2026-06-25 09:24:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/994865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/994865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/994865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/994865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/994865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/994865/1"}}},"commit":{"parents":[{"commit":"9e1b7265f0ccccdfc3b82d8ef87d239503f9bb74","subject":"update constraint for openstacksdk to new release 4.17.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e1b7265f0ccccdfc3b82d8ef87d239503f9bb74"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-25 09:14:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-25 09:14:08.000000000","tz":0},"subject":"update constraint for networking-generic-switch to new release 10.0.0","message":"update constraint for networking-generic-switch to new release 10.0.0\n\nmeta: version: 10.0.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: Icea3dcd66212829b30ba365b4c39e40b7700f41e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id58f90ee8f5379f2369db53cd0ffd9d9ebd290a4\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18c1fecd7e7cc6e0c9a3826a70612ece1e4ba203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18c1fecd7e7cc6e0c9a3826a70612ece1e4ba203"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
