)]}'
{"id":"openstack%2Frequirements~995066","triplet_id":"openstack%2Frequirements~master~Ia60eed445fa9cb687d67c0bb0d7e521a19692256","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia60eed445fa9cb687d67c0bb0d7e521a19692256","subject":"update constraint for futurist to new release 3.4.0","status":"MERGED","created":"2026-06-26 14:01:27.000000000","updated":"2026-06-28 10:18:12.000000000","submitted":"2026-06-28 10:17:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995066-new-release","meta_rev_id":"5806b3188d39d6dc5e69393fea466b26378dea16","_number":995066,"virtual_id_number":995066,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-28 10:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-06-28 08:39:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-06-27 01:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-06-28 08:39:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 15:49:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-27 01:04:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-06-28 08:39:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"250752b2333f650b10bc9e451e08d02cadd65891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-26 14:01:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f985e9e5a5d158987e1cc9ab583e7bc799c5b20f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 15:49:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c96bfbeb7eb4ebba942bc836b1e6754\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b54f02747ce40b6a1e82d4b228456b5 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d60480c49bad4333814ff52faadf49e5 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a52e33202584fed866e6cb5660d9ad2 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77299a0b5c5d46e28a68fc8243709774 : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f0d3fe6719fe43b89ff3976067d7e2a9 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e11582532de47b2becb52f518e28496 : SUCCESS in 7m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c3ab9e52f6b1417ab3afb969feb84058 : SUCCESS in 3m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d68e24e809ea4cdab5352ccdca230bfb : SUCCESS in 6m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7285fe01e3fa46f1a61360a98761b730 : SUCCESS in 6m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/998a9f894f824b20a6f427c3d03ac2b6 : SUCCESS in 3m 35s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/8caa154332464caa83f5594745bedad0 : SUCCESS in 4m 17s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/4bf2868b73ef498bbab12450bec151b9 : SUCCESS in 3m 37s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/944fc25ba73c49e195c60332d595dfcb : SUCCESS in 5m 52s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/8bdde4590fdf47c98f5cdeb54a8491b1 : SUCCESS in 8m 00s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/903fecbc1cc74e53874d7c9eb4a0be9b : SUCCESS in 5m 00s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/11a8aa9b045a4c12a616ccc401b3e4f0 : SUCCESS in 3m 02s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/b15f420d473b4fb59d8c1fa8d4025a69 : SUCCESS in 5m 57s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/1cafe26136db450390519a6daba31639 : SUCCESS in 18m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a54602671cba4b21b7aa0c198c70ad3d : SUCCESS in 6m 51s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/e5ddc2f83c47468893d87aa51a41f597 : SUCCESS in 8m 55s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/899efa42b13948989d1c94fae76802a4 : SUCCESS in 7m 07s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/0e2a66746384471d95ae20839ac33a2a : SUCCESS in 2m 49s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/0f630e59817047a4bd04f7f2f3b552ec : SUCCESS in 8m 53s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/bebf2b7f1c054cc8ad35066e5c5c14e8 : SUCCESS in 6m 25s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/e842c393b69a43578dbfbe89e9706aaf : SUCCESS in 4m 46s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/e9a8614e10c74a68b78ef59e44cb564f : SUCCESS in 11m 33s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/280ea4281c274a02926a02a99320aec0 : SUCCESS in 20m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/605c2a5e1ed844649cd06f6661be4e10 : SUCCESS in 4m 06s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/96e6c8c104684d2e92122ec34fde934d : SUCCESS in 13m 51s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/92ac0c7c50be4c9d89aa24bec47545b2 : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/03768f5e3b0f49b3a9b71b7a5513554f : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3276174a3037444fbaf3d408ac6ffdc9 : SUCCESS in 26m 20s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e24b33ed12764b30a314ad18103dd0d2 : SUCCESS in 5m 10s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/be91e85be9bb4b12996fddd5cc8d36ca : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5dcf7acd7db6494a80acdba897776e5f : SUCCESS in 6m 44s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/257d4c2aafc6480aabc2b6cf45b0683b : SUCCESS in 8m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7f8665c477174d53bf4440fac75d693d : SUCCESS in 5m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecdb9a0aaf234e289aa2f5ed1debb60a : SUCCESS in 1h 05m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc9e983f694d42109555521178aae412 : SUCCESS in 1h 42m 12s","accounts_in_message":[],"_revision_number":1},{"id":"594700534bc29cec5ab1bb7d347fd57f7f64affc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-27 01:04:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"33b87c485379cd8289bd137b43eaaf1e605fa857","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-28 08:39:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f7b307bed008d4cc79f09438a3d768ee445c494e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 08:40:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6428daea0012814f9a70bc6a674273a96c5b700","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 10:17:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2812760d1da34ebb9e614907d5085b9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58b0aa46a5a04e1fb8f5c0bf83505d01 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bca85f54b9184090ab577553c66bf498 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae1bded13be34caf96b49101788ba09c : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06bb25139e254d8fb037883b2265d265 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8378ce3676f84b02b34166ce296681a5 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c90ade979bf433c807e60a3302308e6 : SUCCESS in 2m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/825d6006c3bd4388ad0b669126675962 : SUCCESS in 3m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6e3c93dc2ae04066b577ba32162188a6 : SUCCESS in 5m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ce3f80b08dbd45dda4b8a1716899690d : SUCCESS in 9m 20s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/096e85ee22424d27a5eb3a6598228bd2 : SUCCESS in 10m 04s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/a714bff63b7f448bbf9d38aa2a7a99d9 : SUCCESS in 4m 26s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/3e318489016147e9a983ab496f27ad38 : SUCCESS in 4m 36s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/3ca58069854a4bbab2c4b9950483e064 : SUCCESS in 3m 46s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/6c729a16624e454d965781738313b0fd : SUCCESS in 12m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6fbb131d0df4a4ea4fa41b4229ac2c8 : SUCCESS in 5m 06s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/22de6946eb6946a68d54837372df11c1 : SUCCESS in 6m 48s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/654cfd3a205640f59690f377c076091d : SUCCESS in 6m 58s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/63f81594b2354654a712da8c1b753b6f : SUCCESS in 2m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/ca00e1d9ceb34b93928de0250c2695ab : SUCCESS in 8m 32s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/7a7d2b45c6e34248ac46f2c321ffb12a : SUCCESS in 27m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d08b0b3f32c54928a67b0191f055313f : SUCCESS in 9m 14s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/4dada7d0685c4680b31ef33f6beb4b6f : SUCCESS in 8m 13s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/a10343fffceb4dc69a1adb1823d05293 : SUCCESS in 3m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f00e2cfd4b7a40e788a8119272e414c0 : SUCCESS in 2m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de9ca5ad41c94046a941417b57e7463b : SUCCESS in 20m 10s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/ae5178fa4ed04867826160451fdaa548 : SUCCESS in 3m 27s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/ad522e2de12d4985894d47cd551a3497 : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/141afdf07424477ea4ad16c169219611 : SUCCESS in 6m 09s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/1fb745217135401b9fe65c6cbfaaab41 : SUCCESS in 17m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c8eba47666814a948109e401352f4070 : SUCCESS in 3m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc73a790a1f9418eb09ec1beb49534a7 : SUCCESS in 48m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6e5c43f04e6478b8b3834c95e6538eb : SUCCESS in 1h 35m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e9e9736db4b37c81f03cba66e3a902486b3993ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 10:17:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5806b3188d39d6dc5e69393fea466b26378dea16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 10:18:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a8d93cec10e48468f84ece27ca0ba50\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60eedfc751f342519f998986ce68fb08 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"962f65894e810743a62e9d9dac5fe2f5b9bb891f","revisions":{"962f65894e810743a62e9d9dac5fe2f5b9bb891f":{"kind":"REWORK","_number":1,"created":"2026-06-26 14:01:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/995066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/995066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/995066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/995066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/995066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/995066/1"}}},"commit":{"parents":[{"commit":"655fcff9ab42f99367686b3d93a8ae1384c29801","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/655fcff9ab42f99367686b3d93a8ae1384c29801"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-26 13:51:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-26 13:51:23.000000000","tz":0},"subject":"update constraint for futurist to new release 3.4.0","message":"update constraint for futurist to new release 3.4.0\n\nmeta: version: 3.4.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Id3e91003824addefedd2b805c718e0661d9c3f80\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia60eed445fa9cb687d67c0bb0d7e521a19692256\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/962f65894e810743a62e9d9dac5fe2f5b9bb891f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/962f65894e810743a62e9d9dac5fe2f5b9bb891f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
