)]}'
{"id":"openstack%2Frequirements~995174","triplet_id":"openstack%2Frequirements~master~I9bc4dd9b8f7379040328b435cb479912f006ab8d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-07-03 07:23:07.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I9bc4dd9b8f7379040328b435cb479912f006ab8d","subject":"Updated from generate-constraints","status":"MERGED","created":"2026-06-27 08:18:49.000000000","updated":"2026-07-03 09:05:53.000000000","submitted":"2026-07-03 09:04:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995174-openstack/requirements/constraints","meta_rev_id":"e223dadbba8ac869c52d63e08ac94bbce37f6372","_number":995174,"virtual_id_number":995174,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-03 09:04:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-07-03 07:23:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-07-03 07:23:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-27 10:16:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-02 17:17:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"bb43f4806f6858e2502ca4a0176478ac0a12cb19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-06-27 08:18:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f4fe1170caf0640ab0ade05c10b6fd5a27a67e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 10:16:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2aadf57ff1604929b3da7ce975a87845\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb6bc11cc7d45d0967b9cf190193a44 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa6290754d11439f9a2fe62249eb9bf5 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ac20a6ac4e84c1aa737bdb57928f9e5 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/729bbfcb455d4cd3a5c0898894287396 : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2062c1d03bd5408d857f92626bba1417 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1fcde4e3d29a4b3e91e6de76b68ad09a : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06d7a568245544a2910d176043a28f2f : SUCCESS in 2m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bdc72945eb5540639bd7cb48bd941f3e : SUCCESS in 10m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a366a3f680dc47c183ed03b90f3a1d89 : SUCCESS in 8m 59s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/bf3f1a27e7ba4ddbaacb18470a1f917e : SUCCESS in 10m 42s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/7ef0669a475949cc8cfb7c9dc67d674a : SUCCESS in 4m 48s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/750f015718cf4caeb0d23f7d36b2f05e : SUCCESS in 7m 49s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/7141c31bf67e4e1984f23f80f3384a25 : SUCCESS in 3m 13s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/e4dff449a4b2422992117152c3fc9877 : SUCCESS in 10m 29s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/381404ccc908478babdb23f393842cd1 : SUCCESS in 3m 35s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/5d9fd28732a04d599851ebcbef43f42d : SUCCESS in 5m 43s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/3a3806bc5e7a4ccebfa21064f9d09be3 : SUCCESS in 4m 05s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/300a191379464b0d9093e02bf2c40db0 : SUCCESS in 17m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd1dd64d151941b59b206ff11f5b5e83 : SUCCESS in 4m 11s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/7d9181d73bb0492bb057f0258d2b2cb1 : SUCCESS in 7m 05s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3914616c257646f0a248934e47139c61 : SUCCESS in 10m 57s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/491ae1d0bd5248a6aecb7c4a1f9a22c3 : SUCCESS in 2m 16s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/947c8c4ef8bb4df6966323d1085ee8fa : SUCCESS in 9m 48s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/17d4c5bba4604ffdae3e42288100b598 : SUCCESS in 4m 10s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/76706ea9d6624e3c910bbed08cf5305f : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/55fffe0b63bd4191843c82f079366a37 : SUCCESS in 10m 34s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/aefaa950a1d84a12adc22eab47c6d5b3 : SUCCESS in 19m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2dac67acba504509a66b6e5e0c0788bd : SUCCESS in 3m 45s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/98f0e8c0b0124e829de3c9e50d4a9f67 : SUCCESS in 14m 34s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/ebaa83e50f264e74b1c54a6a76db683b : SUCCESS in 2m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/078de2081d294ff3b7743adf877dc9ec : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a2e3174838f4da79cd12259b01f1ccc : SUCCESS in 20m 50s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/ec9c40bf9ab743ceaa05a5333c14eaaa : SUCCESS in 5m 13s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/fd4b47ed74694784bd31bf14dd8f69cd : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3b102783f38b4768888bbd6028860bb7 : SUCCESS in 6m 20s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/80b45be9008e46d58463611ab262306e : FAILURE in 18m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1b53c142a38b428b9eb124947129a700 : SUCCESS in 7m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/559c723767ec4b50a9e8d30c8b024d5f : SUCCESS in 1h 06m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76ff0d5378274cb0a18daa318ea13a92 : SUCCESS in 1h 51m 09s","accounts_in_message":[],"_revision_number":1},{"id":"770e34294c379f4db2a7e6af096cc9c1a0cece61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-28 15:48:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"173f92c2d6bf9657c06a4272aa09d44e9fb4ee81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 17:09:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f20a54391060412593e6c343427ffbb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/527fcb1e350e4a6399ea875363f657ab : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ba444ecfe494a63888c2d5db8d8199c : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61ccb546ad8c4fedb5f5f44f4a6a4021 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ed7b2f22ba64ed58a0a03f1c0296f6f : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6d1d29912f8047c59017b884709f2a8d : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7bf3bd755324d8c901d2a0f176d0615 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/11b8b064177242d5b6f36e31c1a76cd8 : SUCCESS in 3m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/67535eb91f184788b1e01be5147d3fae : SUCCESS in 3m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1c31ea21641f4f6aaa8d1555b23b3d3c : SUCCESS in 7m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7da42ac60923462c8526394abb04f99a : SUCCESS in 4m 58s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/316582a4634a47819dbe5dff3d8b708a : SUCCESS in 4m 13s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/675f8fbd9ac74760aa4cf25ee45ac49e : SUCCESS in 5m 29s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/ea879630750f4696be770d5024911cb2 : SUCCESS in 4m 57s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/e535c1ddd4b4472686b46749ca317f15 : SUCCESS in 8m 44s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/bfcc68bd08e748f3892f5568ecf67ef7 : SUCCESS in 4m 45s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d8c5631fd7ec4916aedbca20a4f4be36 : SUCCESS in 4m 43s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/00d6421ed8a0422d906c2980fc61aa83 : SUCCESS in 5m 53s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/e5798bb79f6241ec98a6208b86cdf6f8 : SUCCESS in 7m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e2009d911064ed0a1d8bda1423ae15d : SUCCESS in 3m 41s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/f47f6dff61e84436ba4a9a439c3b05d6 : SUCCESS in 8m 23s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/b834c24f8e5d4c44b511f6ee38a8065d : SUCCESS in 10m 12s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/be01673708d5441f97975739285b680f : SUCCESS in 2m 17s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/538b8bd8b308484abf1bbe55cbd28a64 : SUCCESS in 5m 34s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/251e88595a3a448880e1e2fcdaa0ac57 : SUCCESS in 5m 49s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/2a065d7707fc4590b0faad72625a42c8 : SUCCESS in 4m 12s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/d9c3e1540d434548a000c09b3f9f7b27 : SUCCESS in 8m 45s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/d777faacf7c7496d99435a10bc2df518 : SUCCESS in 19m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/91fd3582c0a142aa9f25bdda28d49068 : SUCCESS in 3m 52s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/e02a063030524beda60968b785224bb7 : SUCCESS in 10m 19s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/e0a93acae0e24211922f09d56448362f : SUCCESS in 3m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/93069e75c4a74a168e1154fca8803e3e : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d1f51ae190a4f5d9dfebd423aef238a : SUCCESS in 26m 04s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/43ad08858f19488eac94fd49f769efea : SUCCESS in 4m 21s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/9f5c80c99f484c3995eea6efbe0b2d3b : SUCCESS in 3m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/42a8fb0a2c6b4617bc36dd677cdf7c6c : SUCCESS in 7m 11s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/5734791c9ae94580900a2d7bb1edddbc : SUCCESS in 19m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e8d3ee02f94549d692e3caacdd1d46a5 : SUCCESS in 6m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/742f0dd519f04fa5a2cabb9525774f8a : SUCCESS in 1h 09m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/282832fe731c46fc8e2704ec0da3b3c3 : SUCCESS in 1h 19m 44s","accounts_in_message":[],"_revision_number":2},{"id":"4d75f9336802027da6afbd2e9fd99817b70d5c8a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-02 17:17:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cfa9046a461af68db3e31286aa29a8023cae7491","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 18:58:03.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 995174,2","accounts_in_message":[],"_revision_number":2},{"id":"2fcb4e95849bc448ebe8bba643163f938ab65b9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-02 20:54:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"cf635d51c2b97bf3223e5285fb8548dd288b092b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 21:55:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df51ab62a34d44b4abde1ae7f2315182\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d3c3d10135426d9d2fa7a80f6f890e : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d326ab60b1e4c92bfa077784233241b : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9245330d41a486d9a11d233cb4aa592 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8760e69998664b3f81cf06dd6fc53494 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/68144664fbb0446a9055d51512c5c57f : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b974fd7581543938e43babbef010b83 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7362110ba35b442fa346185be50c5690 : SUCCESS in 1m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/df0a13849c404464871f974207c9f4dd : SUCCESS in 5m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3563d84329db4f80ab61a86d282926d7 : SUCCESS in 3m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4e4306718b8f4b699f2f4b72679ca49d : SUCCESS in 6m 19s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/dc73661fc1ba4aa28bbeb41fe9c657e1 : SUCCESS in 4m 50s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/60aa8e0d02de424b9f1a82e54dcc0d14 : SUCCESS in 5m 28s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/5a4ec3fae80546ec8271fd3450a5f0bf : SUCCESS in 6m 03s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/04a9f5136c8647fe9533b48e2b7028ba : SUCCESS in 9m 53s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/17bb41670bd64a2ca30a6fc25bcec65c : SUCCESS in 6m 48s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/61dee4481fe445d18378b2966b905de5 : SUCCESS in 5m 02s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/f018cbea405d4afa9a63a53b555d5582 : SUCCESS in 4m 29s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/2a16ffdab65e487d9a246b4f86546f26 : SUCCESS in 17m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8040361d01d84e62ba014302f6c53071 : SUCCESS in 5m 57s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/636667c94d234d2f9158ca74d210c653 : SUCCESS in 5m 36s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/5bb90df30561403eb6d9c2a4129845e9 : SUCCESS in 10m 29s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/8b13e8ca89e5435680c998f68cc3cab6 : SUCCESS in 2m 26s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d64d241761c54ba99685f6e5ba71e861 : SUCCESS in 6m 18s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/7cf88bd5ca8a4dbaaefa4cd7b927485b : SUCCESS in 3m 46s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/55a9ecbbe7304b0491eb81edd7dfc5ba : SUCCESS in 4m 32s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/b5b655ed58224a8db6930eb87ddbb78c : SUCCESS in 10m 51s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/0dcf47079f8c4676923f15ea39162e1c : SUCCESS in 24m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e68ddd4969bf48418d629d1ffc6e9c1b : SUCCESS in 7m 04s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/621ca84cbb8d404a91db2a5c1a610568 : SUCCESS in 9m 41s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/26da39e4022c45cba45013227a493039 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6f940b2051214b8fbffd09247d95c9ec : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/917734c627964e7f90b195bb76a24738 : SUCCESS in 21m 37s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/551fd820809843438b16db28ce309d2c : SUCCESS in 5m 51s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/00e927a829df4dd5b79c2faf85b6bf26 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52920d99e381472dbdc471793d498036 : SUCCESS in 5m 19s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a87256f187534889ab5aca27ca976f83 : SUCCESS in 19m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/19146e8f991d4e30a15811069a4afa26 : SUCCESS in 7m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a04375a1ea4448ec84298a0b840533de : SUCCESS in 37m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65f326373ee347ba9d24d7830e53c242 : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":3},{"id":"9470af559d451407bf170bd70e70654ff13f9216","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-03 07:23:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3000b8995d4b099f3e1d7633592dc76da2111f6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 07:25:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"80f40a174aaa9c253fb0985f065848d6fbfdf131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 09:04:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5a480cf09b54b88a2132d09fbfc74b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbac23e0fabb47119ef4821c0f9b38e6 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8281350f3b3b4713bb7fd13ede380b1d : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/788313f2b33f48548554f53bc1ae7e79 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/baa0605beb07415795dd76f2e0aa7421 : SUCCESS in 2m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1908078b39cb42d8b6ff9d8885923bf1 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af46ff8a7131481885b0ee5cea5869b5 : SUCCESS in 3m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5ab9fd316817413e94667a59341d4d15 : SUCCESS in 6m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ea7f84429d6746f68118db9033e5a70a : SUCCESS in 5m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/bdf53d0cd93c4d4ebbf0e3d95a162731 : SUCCESS in 6m 10s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/62fb46a262324f80af308658e9468a8a : SUCCESS in 10m 29s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/89b501df9d584e578430434210e8dc44 : SUCCESS in 5m 54s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/51ed4087a6a44b3b8db99410e6871cee : SUCCESS in 6m 38s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/36a0ccfd93f046d4ae7d7a92078e9ddc : SUCCESS in 4m 06s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/c94506aacf4145cfb8878664c2b13949 : SUCCESS in 17m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3240bf383176407288c70d2c2f1ad9ba : SUCCESS in 6m 03s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/8555c6f5821c4d7cb3c2b4cd88ba0b10 : SUCCESS in 6m 15s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/fc1c3b0a3a9b4104b8b0064f447ce1e9 : SUCCESS in 10m 19s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7b59a863234646558ae35b66ef37ea2b : SUCCESS in 2m 14s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/f3448c093016401381f37b86f5548876 : SUCCESS in 10m 14s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/ca3eac195ab54b16897325f7ee4eaa69 : SUCCESS in 19m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0c5e795e442541c49970366b16a86308 : SUCCESS in 6m 31s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/f0afe256e4844a3cbd17a7559a3aa5f7 : SUCCESS in 13m 29s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/b8abf2ffdee94e4cb91d7909cec4f8b2 : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/90a43222c1994ff1bce3ae177d0bd012 : SUCCESS in 2m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb3bc46fe4a646c9995577aa590a34b4 : SUCCESS in 23m 46s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/5fcbde75133c4dcc8d0bb59af732a2f6 : SUCCESS in 3m 37s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/dab909a3e3a54897a0fa247fc197af13 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab7895b9b1564bc0b77110e38fdbdd91 : SUCCESS in 6m 58s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/35e10bfb3e884711bbe2faf70a53a719 : SUCCESS in 7m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/36c129be322d4dc2abcd8a8fcb22b7c6 : SUCCESS in 5m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0873d163ae3444eca9c06f5f78f0a1e1 : SUCCESS in 33m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48960d5285a84fd0bfc55d4a0768e780 : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":3},{"id":"631f19a8cf083519683c5ce7874b3b0e598b2b3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 09:04:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e223dadbba8ac869c52d63e08ac94bbce37f6372","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 09:05:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c36aa3925e6484a962d659078409a8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ad4fc6106ab41ab8743b7c7766b761b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2c5b9965ac06aadfa063e316bdd88e45dc470908","revisions":{"e61b7bc9d919f8d1ee8ea2cc2fa85b4bb751b9df":{"kind":"REWORK","_number":1,"created":"2026-06-27 08:18:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/995174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/995174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/995174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/995174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/995174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/995174/1"}}},"commit":{"parents":[{"commit":"655fcff9ab42f99367686b3d93a8ae1384c29801","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/655fcff9ab42f99367686b3d93a8ae1384c29801"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-27 08:18:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-27 08:18:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9bc4dd9b8f7379040328b435cb479912f006ab8d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e61b7bc9d919f8d1ee8ea2cc2fa85b4bb751b9df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e61b7bc9d919f8d1ee8ea2cc2fa85b4bb751b9df"}]},"branch":"refs/heads/master"},"73e71ed93274daeb416603ee478c2755153d7bf5":{"kind":"REWORK","_number":2,"created":"2026-06-28 15:48:30.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/74/995174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/995174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/995174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/995174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/995174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/995174/2"}}},"commit":{"parents":[{"commit":"655fcff9ab42f99367686b3d93a8ae1384c29801","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/655fcff9ab42f99367686b3d93a8ae1384c29801"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-27 08:18:47.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-28 15:47:20.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n- pytest: swift unit tests\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I9bc4dd9b8f7379040328b435cb479912f006ab8d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73e71ed93274daeb416603ee478c2755153d7bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73e71ed93274daeb416603ee478c2755153d7bf5"}]},"branch":"refs/heads/master"},"2c5b9965ac06aadfa063e316bdd88e45dc470908":{"kind":"REWORK","_number":3,"created":"2026-07-02 20:54:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/74/995174/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/995174/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/995174/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/995174/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/995174/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/995174/3"}}},"commit":{"parents":[{"commit":"625339afe2184001102c5783080acb954f40bdbc","subject":"Merge \"update constraint for stevedore to new release 5.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/625339afe2184001102c5783080acb954f40bdbc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-06-27 08:18:47.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-07-02 20:53:42.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n- pytest: swift unit tests\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I9bc4dd9b8f7379040328b435cb479912f006ab8d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c5b9965ac06aadfa063e316bdd88e45dc470908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c5b9965ac06aadfa063e316bdd88e45dc470908"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
