)]}'
{"id":"openstack%2Frequirements~995651","triplet_id":"openstack%2Frequirements~master~Ia176f35ab2bd0b3d2772672ead7e8ba0e2332431","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-07-14 19:16:08.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-14 19:16:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia176f35ab2bd0b3d2772672ead7e8ba0e2332431","subject":"update constraint for tooz to new release 9.0.0","status":"MERGED","created":"2026-07-01 09:06:39.000000000","updated":"2026-07-29 19:36:03.000000000","submitted":"2026-07-14 19:16:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995651-new-release","meta_rev_id":"7337b5d1a612e1335110a28098bd413fc2127061","_number":995651,"virtual_id_number":995651,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-14 19:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-07-14 08:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-07-14 17:34:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-01 10:27:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-01 15:00:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2026-07-02 20:43:54.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-07-14 08:53:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bc88cd8b67632989aaf77e644495ddd69fbe805f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-01 09:06:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8008f5d984eeceb225a5da63c2274506a3e6bf10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 10:27:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a2b21c109224348a4a0f7ac2d7671cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4961fae2f25d4bd8aa619c431514d236 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77968a6911f342f0b68ba9117ad76a10 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af0669fd453c4e33bc6d162165eabf70 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3dfc5367bb514ebaa243a141a2a23e27 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/714725e0d9bf4bed9f3ee98f69845a84 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69627db58ca34f66ae0ae60d444e8710 : SUCCESS in 2m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/abf17d5617a34190990b76eaddf334e2 : SUCCESS in 3m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/289cfff0b5384daa9a778ecb5a09f3da : SUCCESS in 6m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/07b35b106ddf44e28e80a9c87abbca65 : SUCCESS in 8m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6efa987e9f62485991fabb6ba38e20f4 : SUCCESS in 3m 13s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/16633fd1ea124ba19004d700733967c8 : SUCCESS in 5m 14s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/837ac03dea18485fb0ac1499d6ae767e : SUCCESS in 5m 58s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/dbed23ebde2c41a3909d7390d2bd504f : SUCCESS in 4m 59s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/eb66f61adfb3405aa81e8085d48a8850 : FAILURE in 8m 30s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/f0f1aee01b9344dca5ca7436088c308b : SUCCESS in 4m 35s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/6702b861de9b4dc5be71d93f133d1772 : SUCCESS in 5m 05s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/f4922f1985e446dcba79d7f4f1dec755 : SUCCESS in 5m 24s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/9ca90a7f1bef4a2c988e5e3ded2e9bbc : SUCCESS in 8m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da3c4f46d65845f3892005a1d78870c0 : SUCCESS in 5m 45s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/994ad336f518429390e386f7c55e8ee1 : SUCCESS in 6m 25s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3fd006b5324d4902b770dd94ecf8f15c : SUCCESS in 10m 09s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/20c025306b8a48e6a0ccee81014565d7 : SUCCESS in 2m 22s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/8915440e20ba4f3f893fcd2125a97024 : FAILURE in 5m 15s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/8e5441421990410aac0bbf0d64b42812 : SUCCESS in 3m 59s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/55cf1a713e864a97a6fc8cc6b4142ab6 : SUCCESS in 2m 54s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/b4b38cf29a0c4e498374edf8c0b39c46 : SUCCESS in 11m 10s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/2d88bed28964480cacf4a882ae790d1e : SUCCESS in 21m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6ae0abd05ca44c50b19568fe6c206b70 : SUCCESS in 6m 22s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/02e15e36dd43494094c7b64c5e908ab3 : SUCCESS in 13m 00s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/27e7bc979ef047e693183319998bc9cd : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b500acb38c24f8a83809d3a8fed2b79 : SUCCESS in 2m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/49d7200d8e0240c4b056d53ffcf43fd2 : SUCCESS in 25m 53s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/141ff39583544eed9236983d2183148f : SUCCESS in 5m 29s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/69c8993ce42b49ce8ab727429c8623c8 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f5c26d0e47c4ec79c41706372f4c480 : SUCCESS in 4m 04s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/4564dc48e126449fa836af834f252199 : SUCCESS in 7m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9149c2067d494c97964a6fcc262f8598 : SUCCESS in 5m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e33ab64c0bfb4d9f944784713c2cc003 : SUCCESS in 38m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7f2b6453f274bafaaea1e543466e90f : SUCCESS in 1h 14m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9fc29a282f18e119cc607db603f90aac8a9abffd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-01 14:24:33.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f62565357522d6d6b6f0bfc1c47c17f5f86e1805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 15:00:19.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/995691 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"f34b766446f5e7972036e45b3bfb9d13da462fbd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-01 15:00:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6722af78c376e8126a1305aa178f051010cb3f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 15:12:23.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/995691 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"df861e2eb2a3461e9955765b4adf831940dfeaa8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-01 16:40:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9cdc317439708feb73e8365f5a4b27e49e54e973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 17:53:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fd8008429bd451faa978907f00dad80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b98b51fec5ed493e84b3371fba48d26d : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f56b135f9594488a6592feb893b52db : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b7630ae486643268e1ea3061d911e32 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05d06208f11b4a18ae093bca8ad7ce62 : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/357bc5f1d1d6493088263800c15b3f4a : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7be9f2f45c2c45a1ad79cdf1a69ddb70 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9de6e96ed6154fbb9b0a787f1e21704d : SUCCESS in 4m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7b96212d69ec4351b15e68f481eea5b4 : SUCCESS in 6m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/445c1c7c703445a8a437030c14233c32 : SUCCESS in 5m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cb80d032168b4d4ab82df53294e23841 : SUCCESS in 5m 22s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/0329787a1e8349ba98bf43fb62749a88 : SUCCESS in 4m 58s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/0240e0b9640f4f4f8ae1399c29234f33 : SUCCESS in 4m 53s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/597e31be222d41019182e34a3faf271a : SUCCESS in 5m 49s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/6a207dcecb824223965cb5ecdd80accb : SUCCESS in 7m 45s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/370227ee477b4f95abf243a7758062cd : SUCCESS in 4m 50s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/c2eec226aac5430dabe4a57d9da2edf0 : SUCCESS in 5m 19s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/08412e42aac64e6aba333a5f6b10e65c : SUCCESS in 3m 26s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/01b7057cba8c4655a925b72d880d99aa : SUCCESS in 17m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97612bc6d60c41fda7db8ac58d8c90d2 : SUCCESS in 6m 07s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/697229c50cdb44eeb258c51965732ccf : SUCCESS in 8m 37s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/559f0d4a93224204814c5e60601b5ead : SUCCESS in 12m 49s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/6452baedff8c44b1ab1e80b902c27262 : SUCCESS in 2m 34s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/875291f6afb34b6b87e4cbbba531c066 : FAILURE in 7m 54s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/8f18c197309144cc89d56da7ce1dd28d : SUCCESS in 4m 21s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/4f1c9e5a4c694bb1976cb83f4494999e : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/8335c902a2594bcf92ce5264b9c56fd1 : SUCCESS in 9m 52s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/7ca4635c971d4a4d9139edbf2e6d61c5 : SUCCESS in 27m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0d58dd23f20147618c54490295b1e8ba : SUCCESS in 3m 39s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/85f0cd8c5a74417d8e2396aa45b414c0 : SUCCESS in 8m 13s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1d5c1f9646d949b098db120eac9cc70b : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/57e0b6d5451340a1a29e2a8285bb95cb : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/532e82c63d6441e7afa9d91e8b32bfe6 : SUCCESS in 27m 35s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/78a04dd5d4194f7e87e1cf4b4cce0b6d : SUCCESS in 4m 48s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/cb96206a822a476ea9d166eab2437994 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b2409ce24a674bcea59f82138efc10ce : SUCCESS in 4m 07s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/7112b4ae1648454e9d1b65c9e9dfc723 : SUCCESS in 15m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/550a2fa17b6242adb9e13d1607fb522e : FAILURE in 5m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e85bca4ccd1e4d7e8bb494274b47d718 : SUCCESS in 1h 09m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebbd76663cb248cda37571266942488b : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":2},{"id":"5b0461e233f37bea5b3af9e92142d176c9a530cf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-02 15:56:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"576e056eeb86945b797ef2a9113b73fbca32f939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 17:53:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d220cdd0f8e4dd88b6d7ba4be6f2de0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b971c62f81a49d8a76cd6daab606bcf : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae68fe10bce34d2ea16b1d20e73382d1 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9f454f35cb34da49539595fd9db738a : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da11189571e645098666a48ba3aef38f : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ac7c0e9ad3d1484d91ebf33a018e6c6c : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/123929cfb5704507bfa8fdeba63ce645 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/466703f79b0f45b7835cd022e16f422d : SUCCESS in 3m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d90e07001d7d43d593fb6da95245388f : SUCCESS in 4m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d35f6e36b3ec4032a0e659207a83f5f0 : SUCCESS in 6m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/933eed665d4c4929985e53f0f02f925c : SUCCESS in 6m 02s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/43b73b73969648a38b769755384ac177 : SUCCESS in 6m 27s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/e38578335bf14725a9e56595c5749ae0 : SUCCESS in 4m 29s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/2c5b428c9cf340beb40f6f2168639fd2 : SUCCESS in 6m 55s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/05783135768e4bf7ae19a6ad35b2a754 : SUCCESS in 9m 49s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/8c1bbff16fa8470888f8722ebb0540d1 : SUCCESS in 4m 59s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/309860e431e04edeb208223d83f53960 : SUCCESS in 4m 37s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/138e92b015d6401a9061acb818533989 : SUCCESS in 7m 49s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/25884b1fea874e60ad34545e0a351a38 : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c6987c55aea6482f93c6ee443bfd0742 : SUCCESS in 6m 58s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/9fa76637155649aba6cab3cf51eb01cb : SUCCESS in 8m 16s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/20b8e49d1fd24fc6ba74a5dd7ede3580 : SUCCESS in 6m 50s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/fcdf74b548eb40888683b9003b757040 : SUCCESS in 2m 46s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/102f50d4351c402fbb8cd566c17d616f : FAILURE in 8m 36s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/b8a99482bd754e5c8675b0976d0f6e15 : SUCCESS in 4m 24s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/1caff0ef47a74e9bb637b56a3da15082 : SUCCESS in 4m 24s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/50b87302966e4a309ca2f590bdec6e74 : SUCCESS in 10m 22s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c5902d1f1ecc41ac810139b5092c6dbc : SUCCESS in 23m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dc71cdb218a948adb2edee142e1b1c88 : SUCCESS in 7m 07s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b4c402d222684f61b683d49a9bebe761 : SUCCESS in 14m 45s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/5cc32b95e6a848389932acda3586d7b2 : SUCCESS in 3m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/70e6fe309b9e441a8a5f681a322fac4c : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f09bb30c46094559827526fefc951398 : SUCCESS in 29m 48s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4ba949933f9a426fa16701a3372b9a83 : SUCCESS in 5m 48s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e9f2641033ca4bdeab54bbe0d69080d4 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d2c298fe151b4c5bb1d66b0ef2a37886 : SUCCESS in 7m 02s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/84001f9d9fbf491db8142d3309e6b37b : SUCCESS in 19m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e953c3a6e4cf44959756867c412f27f6 : SUCCESS in 5m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b4b5a2fa9ea4c709125e7c06679184b : SUCCESS in 1h 12m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44f902c9e1d64607ad3fbdb695f189d5 : SUCCESS in 1h 50m 47s","accounts_in_message":[],"_revision_number":2},{"id":"7e36b75d60cab0698cd3d96f5df547d052b83bac","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-02 20:43:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54c7015869079a864e51b361c17f96b0ecfddc4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-08 15:15:43.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1f2afda6277735367e8316e73851da53cf0b1add","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-08 15:16:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d666dbc25441395a4f10a8bacc477bd57225b4bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 17:01:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6667a8bdc3d14e2e83457bd5e79ae8dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ee4c910c6f472c8dd89d0245679468 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3c55085abc249e58f9b20c7c56d9d23 : SUCCESS in 2m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec6664f08bc74a7eb41e9108bc270da7 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c09259fb92d74d778e75fc22417f3a86 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/052cd74a6fb5437f9691eb3f51955f82 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6e0a51ed31449b88ef2eccad9ceeeb7 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b1a5b0c2a024481bdc8bdbf939b3ed1 : SUCCESS in 3m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3a575b6a451c41f8a746c37ac478f88a : SUCCESS in 5m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d95d7375b974498eb52b63db11735e30 : SUCCESS in 9m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5e30f8ab9e1d4938b0e50f42b9e8eb3f : SUCCESS in 7m 11s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/5790029d5dc84b64aef7532c27dcc72c : SUCCESS in 5m 05s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/4eac3bbda0424db3a7b72ac354643ae8 : SUCCESS in 5m 28s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/758029a2e5f3435fa5c1e16db4bfe6a9 : SUCCESS in 4m 31s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/30683c25223e4f699b4a5e2acdf843ca : SUCCESS in 11m 06s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/156f7644cf614a10b70cc8ecfdb3a228 : SUCCESS in 4m 43s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/682911c52be34d30b85b92a26104d425 : SUCCESS in 5m 27s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/0bd76623c8184cbcbc8d641ff35f7457 : SUCCESS in 6m 39s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/3235641a298c411c84753658b04a13d1 : SUCCESS in 18m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e21ac4450b854dd1ab492b0931fb640f : SUCCESS in 4m 15s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/260d7b46bcf0414685214f0ddb429360 : SUCCESS in 5m 16s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e768582cfe254d22bb20ee6e999a4837 : SUCCESS in 13m 18s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/66157379012245edaa46a86ea1ba8ef5 : SUCCESS in 3m 22s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/4b1a617eb91b4a35af474bdecc862c60 : SUCCESS in 8m 16s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/8b44cf086ab84feca8656522777a9ed1 : SUCCESS in 4m 06s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/a6cea4711c3241c192e50741050ca0a1 : SUCCESS in 2m 56s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/338218e5b92944d4b181441ca866d3c7 : SUCCESS in 10m 06s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/54a60be424b0455f9f22a2306254e913 : SUCCESS in 25m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/725365ebf5ea47f3badebb3662572790 : SUCCESS in 7m 33s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/06c5247e021f451386ec54a93095042f : SUCCESS in 11m 44s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/72b54cf7bcd242fb9d598509e0a2ed92 : SUCCESS in 3m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ff28ef553a764008b1f03400ae3b05ae : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c8d085b6b5de44ee9f2cc941e4b4ab1b : SUCCESS in 29m 06s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4472bfed93a94b159f25def71065fcf7 : SUCCESS in 3m 26s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/cc17ae81738a415c9b0c78fa1b94f97a : SUCCESS in 3m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/54df619ceace401798cb9198f176bbd5 : SUCCESS in 6m 29s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/62bfd17db2c642aab61e84804f1b2cfd : SUCCESS in 9m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cf1769b941a84a70b1db2c21b9d4e867 : SUCCESS in 5m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0c3c2781ecd480fbc90700e9bad4ca4 : SUCCESS in 1h 03m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef6df0da7cf14d5e8c6c14fc74eacd2d : SUCCESS in 1h 40m 29s","accounts_in_message":[],"_revision_number":3},{"id":"ddbbbc52e093a21c27bf067efbb7fab92efc9023","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-08 17:22:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f4ace518f91e7d6abbc5cff59c44c8577563372c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-14 08:53:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"15e2ebe6a78be56b7ca70763ce93ea35305f44ee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-14 17:34:49.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9d53e198bdcb4154cf62563bfbff8c5af8d9286d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 17:35:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9f613b06133e0c97f76dbe1639e4214a9dcd5db1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 19:16:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a95dc2edfcfb427d9770f3bdac78b57b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f04f03bd974121a1e2d07cee63ecc2 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c87e1571cf34478bbf7cc743e30e5db2 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1cc03594f69e46ca9ab28f0757517e06 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e5ed2b01d2d47208791cf1406cd2442 : SUCCESS in 2m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6871a5b35c8547579149eaf6b8c521f3 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/031a7990c85c466c8ac672ab63b1f168 : SUCCESS in 3m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66f5142eae364308bf7a85dc572a545f : SUCCESS in 5m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7c6a4238b1124727bca172cd595fce4f : SUCCESS in 3m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/325bbce34a394687810a49c0c4e71704 : SUCCESS in 5m 25s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/8f0ce49d10814c62ab092ce2b490c6b6 : SUCCESS in 10m 52s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/0fc3485e465f451eb761380b8d2d1980 : SUCCESS in 4m 33s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/1c5d4088a3e1495899e2d9054c2e2a98 : SUCCESS in 2m 41s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ae248e4ad924478fad8a473f9a63a9fe : SUCCESS in 5m 13s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/93a2579a559745dd83c1ff877b0dd216 : SUCCESS in 15m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74c6e53e8cca434dbe88e5399806d50f : SUCCESS in 5m 58s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/a2a831a881084514a1c8384f40b9e9b8 : SUCCESS in 3m 56s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/c765841b49c641699322be1bde29948f : SUCCESS in 10m 52s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/07077eee48b147e1b39a8d64ad2bfab5 : SUCCESS in 2m 27s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/6df8802cba6d4543bea47146a5e07c16 : SUCCESS in 4m 29s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/2a2ba0c979d04d249d06c7ccadac3ab8 : SUCCESS in 39m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9bc5ba3db2f24b5fa505428b1165dacf : SUCCESS in 8m 00s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/66bdb18014984a4ea282258b681b82d8 : SUCCESS in 13m 04s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/d31911ae3c434026a25cc522eec6ac09 : SUCCESS in 2m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d3a650251ba147b394ede1fa878333f7 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/85aac61c1ba943898b8ed013fa16e1da : SUCCESS in 28m 38s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/7df1c048298a42a6a78f5a5b04a53d16 : SUCCESS in 4m 36s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/28d81f1a543644e4a46b6bba6988f1a6 : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3ab160403304f7386fc45a3fccea9e5 : SUCCESS in 3m 31s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/e249172d0375495582c3b78db87a5a4e : SUCCESS in 18m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7bbdb042a3ea422a99f07f2fa9b0f7d7 : SUCCESS in 3m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c9dc84c71bf4f038f2d517cc85211c9 : SUCCESS in 1h 02m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c72ee3107ab14cad9884cb0f5490d204 : SUCCESS in 1h 25m 07s","accounts_in_message":[],"_revision_number":3},{"id":"162b6ed01c17db749c8a5fb4392b92e52fb2ecb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 19:16:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"01322c4f816718afb83de80a43d5e2266904ce01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 19:17:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe3848a6de4743f3bffc74240325c069\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fca4f2c286e435a8ec6e543f783219e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3},{"id":"7337b5d1a612e1335110a28098bd413fc2127061","tag":"autogenerated:gerrit:revert","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-29 19:36:03.000000000","message":"Created a revert of this change as I5d35ad179dff8d4f795ae71bb4568e024117aa9e","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6125a0544382f14a5bcae3922b5ec4871a2bce5d","revisions":{"56f0448e75900c32c77e73845a58879274c0590e":{"kind":"REWORK","_number":1,"created":"2026-07-01 09:06:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/51/995651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/995651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/995651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/995651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/995651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/995651/1"}}},"commit":{"parents":[{"commit":"e93565a60e34357734bf5c7e2bff23d49969addd","subject":"Merge \"Stop vitrage requirements sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e93565a60e34357734bf5c7e2bff23d49969addd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-01 08:56:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-01 08:56:38.000000000","tz":0},"subject":"update constraint for tooz to new release 9.0.0","message":"update constraint for tooz to new release 9.0.0\n\nmeta: version: 9.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I8f37c9fbbf3fe38f27942a51492efb3d72f8b35e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia176f35ab2bd0b3d2772672ead7e8ba0e2332431\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56f0448e75900c32c77e73845a58879274c0590e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56f0448e75900c32c77e73845a58879274c0590e"}]},"branch":"refs/heads/master"},"b01f716ca624fbdd158a3da52fa3f5107f293415":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-07-01 14:24:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/995651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/995651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/995651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/995651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/995651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/995651/2"}}},"commit":{"parents":[{"commit":"e93565a60e34357734bf5c7e2bff23d49969addd","subject":"Merge \"Stop vitrage requirements sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e93565a60e34357734bf5c7e2bff23d49969addd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-01 08:56:38.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-01 14:24:33.000000000","tz":0},"subject":"update constraint for tooz to new release 9.0.0","message":"update constraint for tooz to new release 9.0.0\n\nmeta: version: 9.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I8f37c9fbbf3fe38f27942a51492efb3d72f8b35e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nDepends-on: https://review.opendev.org/c/openstack/cinder/+/995691\nChange-Id: Ia176f35ab2bd0b3d2772672ead7e8ba0e2332431\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b01f716ca624fbdd158a3da52fa3f5107f293415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b01f716ca624fbdd158a3da52fa3f5107f293415"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6125a0544382f14a5bcae3922b5ec4871a2bce5d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-07-08 15:15:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/995651/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/995651/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/995651/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/995651/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/995651/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/995651/3"}}},"commit":{"parents":[{"commit":"e93565a60e34357734bf5c7e2bff23d49969addd","subject":"Merge \"Stop vitrage requirements sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e93565a60e34357734bf5c7e2bff23d49969addd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-01 08:56:38.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-08 15:15:43.000000000","tz":0},"subject":"update constraint for tooz to new release 9.0.0","message":"update constraint for tooz to new release 9.0.0\n\nmeta: version: 9.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I8f37c9fbbf3fe38f27942a51492efb3d72f8b35e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nDepends-on: https://review.opendev.org/c/openstack/cinder/+/995691\nDepends-on: https://review.opendev.org/c/openstack/manila/+/996489\nChange-Id: Ia176f35ab2bd0b3d2772672ead7e8ba0e2332431\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6125a0544382f14a5bcae3922b5ec4871a2bce5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6125a0544382f14a5bcae3922b5ec4871a2bce5d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
