)]}'
{"id":"openstack%2Frequirements~995662","triplet_id":"openstack%2Frequirements~master~Ic823212a0f12ac300fc729d13aa04f6b99570514","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic823212a0f12ac300fc729d13aa04f6b99570514","subject":"update constraint for os_vif to new release 5.1.0","status":"MERGED","created":"2026-07-01 11:01:52.000000000","updated":"2026-07-02 13:52:22.000000000","submitted":"2026-07-02 13:52:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995662-new-release","meta_rev_id":"772837e339eec7a2fc9a6c5f10241ca83590fd36","_number":995662,"virtual_id_number":995662,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-02 13:52:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-07-02 11:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2026-07-02 11:37:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-01 12:32:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-02 11:37:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"4cfc6271baf6142ad413c41aca8768bee32f20bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-01 11:01:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ac6fa6c34773b7c83a39d632f558c40c1af3104","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 12:32:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bee339f9f3fa4d4fac11b8073da07464\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5369fe12ec254ae3829df655ba8b629c : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6b9969f21c4caca6f5733ac6b66be5 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7858ff95dd924a41b2e99a34c7d8bb89 : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4486202011d94e0ab32a8a14f2b9bd54 : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/513168d0f4314a66a9247205af2e37d9 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94274272d1874b3abe173ae03797c51d : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2aed8f930e05496ebe50710e37129cc5 : SUCCESS in 1m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/aae47b0bf33042eabbe250680dad4f1b : SUCCESS in 11m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4f7008ddfd4448e4bbf035771609a57b : SUCCESS in 5m 29s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8904df9f5fd4412ebc4d61a5862f0eb8 : SUCCESS in 5m 10s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/f0a1adf01c8d4ac1ac3b1a6e2cb0bf43 : SUCCESS in 4m 32s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/f21bd0c1d2484342b3ed65637775d99f : SUCCESS in 3m 13s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/67e86be14a60466294076f9db08285b7 : SUCCESS in 5m 33s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/7b7365a8f1ae4c8cac326f6d5ccea10c : SUCCESS in 9m 56s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/f5c979e587704b69b9e674497c132eff : SUCCESS in 4m 05s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/77672c8a107b4ea8a6adb0c55d7120b3 : SUCCESS in 5m 10s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ce64c47aeb874a499d26dda95a33a0d0 : SUCCESS in 4m 02s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/0979f214725645cd82546234ed2894d3 : SUCCESS in 8m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/970cab38a1e843d5b2e90f4bdfdabee0 : SUCCESS in 4m 41s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/acb49521595c4b9c98539903e8a82ef4 : SUCCESS in 5m 41s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/8f88d544acff43e8b42667d92b77c582 : SUCCESS in 11m 34s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/10c2d8b4016d47fb91007ebfdad5f42f : SUCCESS in 2m 41s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/4d6c4df9da6f49328bed685d64f44f88 : SUCCESS in 7m 58s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/c8d5ab76432a461eb55a61d7b442f4d9 : SUCCESS in 6m 36s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/acb8e60b42d549f7a0cdec52327053d0 : SUCCESS in 6m 41s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/bf7c6e783d09472c9f9a590b9dc05295 : SUCCESS in 12m 30s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/445245fc23fa4e55b2ae6fd00e8e4df5 : SUCCESS in 19m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d2f38685a6694fef9df0f961861073b4 : SUCCESS in 6m 35s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/8d62e9a46d244319a0c197635694c4c0 : SUCCESS in 12m 51s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/b40adf3a1e364830961b4c2cfae07690 : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c14be1ea8d854f248fc50e2b41cf5163 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/85d48cad2fc4426d80db9656773aa28e : SUCCESS in 24m 48s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/667740aadd824fe9b72c86eedf339665 : SUCCESS in 3m 20s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/c41839f728cf4e58941bbbf7ea1b8b01 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0111a21116bd4de7b7db76746861b2d3 : SUCCESS in 3m 23s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/54c37a6630ad4bf88aef016c10486f56 : SUCCESS in 17m 44s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/90e05901332d4fcdbee730fff6facf12 : SUCCESS in 5m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53cb12a2c93447b893d68dc304ccde31 : SUCCESS in 1h 03m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2961542063f94e9abeddd9d21cbc7dc9 : SUCCESS in 1h 28m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4dafd902b90ff2678772d1a69eb245871e669eb9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-07-02 11:37:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d0cea19447d3e4fbaea6086540eee5624b157a7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-07-02 11:37:52.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d211091ecb651700c7e6a8b33188b91fd92353a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 11:38:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a0dbc5cc5c0c972ebd6155b03168a8b482ec990","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 13:52:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a8dbca9ee36433b9f9185b6768c7614\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4b95c76253499fa637355b40965fd6 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad4db039891840b59d872d7ac7eccfc4 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8fcbf67adfc49759a19c16c5a030053 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8dd5d21ac4ed47c1b9024c0050bf77c0 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9da16f5b4c54378acbee7aed6528e35 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96dcffc7e32c4bedba42720b13210cec : SUCCESS in 3m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f268775a81ef40c4850a47aabb83f6d5 : SUCCESS in 5m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9080454081054b899394d165b6949fd1 : SUCCESS in 5m 42s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1ad375778a1a4e3d98e9b652e5ab051e : SUCCESS in 6m 55s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/3641bbcaf80e48089afdf5b89d608236 : SUCCESS in 11m 11s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/25bee4193ddb42438a17d685904121ba : SUCCESS in 6m 23s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/f79fe1b28f1843168fa4a6cdef1a47b0 : SUCCESS in 5m 11s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/bb03872db0594107a67749396c0962ff : SUCCESS in 6m 54s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/0494ed3a680c4fb493c20c07dae763c8 : SUCCESS in 17m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d4ba3c99d004d46891df31e32d5cd43 : SUCCESS in 7m 18s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b6b97a510dee4aff94918b9ce03664e2 : SUCCESS in 8m 31s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/b6b7e12aab87448995c1c3e51a6e2de0 : SUCCESS in 12m 12s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/de5f107a049145048d130c31f5472e74 : SUCCESS in 2m 47s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/25a8818e9c5e4f7382a7ca8222509cec : SUCCESS in 8m 08s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/f07ef3ef39d34c35b99007eadb3d9786 : SUCCESS in 34m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fdefc29bb1e446d1a1cd5f2f982839a8 : SUCCESS in 6m 28s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a9a8289a4ec444efae77b7a4653ac3a3 : SUCCESS in 16m 41s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1b3abbfcbbb041fc9d85d4f987b89aac : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0866cd06c6b242f49291ad9756041cc4 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0c95d73846e46be94d7e8e505c56272 : SUCCESS in 25m 57s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0ac281573edc48768e6c036db9868efc : SUCCESS in 6m 12s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/efc1d6dbcb3d4bd7be020516f1df03de : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b91942c8fb94d8b9a11bff3fe2b0342 : SUCCESS in 6m 59s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/eee6afc65fa440a780c88908c48240c1 : SUCCESS in 18m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e8ef9281c98d4956b3f554cc3e72db40 : SUCCESS in 6m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2d0cd93b2934ef0a27f8d1af0237654 : SUCCESS in 1h 02m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ce3bf50d22b44a1a44187a872bde860 : SUCCESS in 1h 52m 16s","accounts_in_message":[],"_revision_number":1},{"id":"772837e339eec7a2fc9a6c5f10241ca83590fd36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 13:52:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a95c953ed587f5414be16d8a75dd8fabc36f194","revisions":{"1a95c953ed587f5414be16d8a75dd8fabc36f194":{"kind":"REWORK","_number":1,"created":"2026-07-01 11:01:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/995662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/995662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/995662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/995662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/995662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/995662/1"}}},"commit":{"parents":[{"commit":"e93565a60e34357734bf5c7e2bff23d49969addd","subject":"Merge \"Stop vitrage requirements sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e93565a60e34357734bf5c7e2bff23d49969addd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-01 10:51:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-01 10:51:50.000000000","tz":0},"subject":"update constraint for os_vif to new release 5.1.0","message":"update constraint for os_vif to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: team: nova\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iacd17f4f337706c7c7d37057fd11d5765d0f4558\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: ribaudr \u003crene.ribaud@gmail.com\u003e\nChange-Id: Ic823212a0f12ac300fc729d13aa04f6b99570514\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a95c953ed587f5414be16d8a75dd8fabc36f194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a95c953ed587f5414be16d8a75dd8fabc36f194"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
