)]}'
{"id":"openstack%2Frequirements~995671","triplet_id":"openstack%2Frequirements~master~I58fa24709808b24c11766b042b1c330facb51059","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I58fa24709808b24c11766b042b1c330facb51059","subject":"update constraint for python-designateclient to new release 7.0.0","status":"MERGED","created":"2026-07-01 12:02:30.000000000","updated":"2026-07-02 13:54:11.000000000","submitted":"2026-07-02 13:52:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995671-new-release","meta_rev_id":"136e614007d1f34ba4c5ce281728ee0e5a9c08db","_number":995671,"virtual_id_number":995671,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-02 13:52:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-07-02 11:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2026-07-02 11:37:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-01 14:19:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-02 11:37:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"00c98bcc767af973df25fc3af1951eaf31c368eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-01 12:02:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0240e416eab3733b1b6a42b5636af09803c8bd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 14:19:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d18d3889115f4283b97b72cce9eb2991\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88e48228135a439b9a0edeacf361d958 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a74592dde09941c5ade6b907ba556909 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff2b5ebf24644d9981e17d6433895c29 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98f2cc18acb54655b57c62bf71908116 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9baa2c1de8e84992bc3c736e60a735b0 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0841acdc0b746709762a52284740980 : SUCCESS in 2m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4940c87a1e264991bf3393ce0e1c42d5 : SUCCESS in 3m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5f8b864136ed4bf48e1047b558ad1fbb : SUCCESS in 9m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/db69ba2dd91e4a0e8d31ca0426421061 : SUCCESS in 4m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8d0feaaea6d74c868d0ccdd01e03517b : SUCCESS in 3m 08s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/22f56d1d7981403691cd38351920d4a9 : SUCCESS in 5m 54s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/2d0625020b4844e5863194ddb72cc36e : SUCCESS in 5m 57s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/bec809a1f28647e881e195fe0acb5dc8 : SUCCESS in 3m 03s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/8a07525d7a1c454b8f90fde0633358df : SUCCESS in 7m 15s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/905dc466e9924f0ba676119dd9139f81 : SUCCESS in 5m 34s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/7cf674b4fd8548c2b0972ab306be947e : SUCCESS in 4m 48s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/da59f4d736094f35ad1e11c137c6f1b6 : SUCCESS in 5m 19s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/90f37bbd8a5040af8c0e78009e7550da : SUCCESS in 14m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da2d27fc42924bc98ec1708f9625ca32 : SUCCESS in 6m 14s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/0d46e8bd09b04b8287a98a1cc7af712f : SUCCESS in 8m 20s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0a0a5ed952b64e4db9c6b300bdb5e4eb : SUCCESS in 12m 42s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/0bce278b4481485e907044bd59f7180c : SUCCESS in 2m 30s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/f1e937ac616c4ed695d3b15d18345dc1 : SUCCESS in 8m 30s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/6181136b9ea542aabe632fda1b7c7235 : SUCCESS in 4m 54s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/ee303ca3b5bc461eacb30a8d2ce64db8 : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/b08b0b38195f4761a515a2ce6100ee30 : SUCCESS in 10m 18s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/cb7355dfff674b6abe2fba265592218d : SUCCESS in 20m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/afb8ee2079244a739a1e7d81eb16b087 : SUCCESS in 6m 27s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0094ecab3ce74b929ca09cef6334f008 : SUCCESS in 8m 42s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/75fd9b2d77a941d5b491a3f1701b5caf : SUCCESS in 2m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46f7e928b8fb4a2897ec91e21e716c93 : SUCCESS in 2m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c55ca40903504082880ecbf1818b8e14 : SUCCESS in 25m 23s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e538c8571e01403698bf958edc424c69 : SUCCESS in 4m 42s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/4c956734cbb54137ab43c1dd25a57e8a : SUCCESS in 7m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e7ca3a1acfb409d9a1a53b2f73c6c72 : SUCCESS in 7m 15s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/99c56e856272466ca78782a78ef3d212 : SUCCESS in 8m 31s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/88f172bbe958413b9fda3caca554bfae : SUCCESS in 7m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/130d3a2fedf84d61b9db79dcd6095bf5 : SUCCESS in 1h 05m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95e2883c93be4285bd072fcc17bbe7a4 : SUCCESS in 2h 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"63251ae919b2b532c0de1131e65f45f7beefc026","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-07-02 11:37:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d977df9967538b47af2161079953aac1b9d3362c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-07-02 11:37:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d28a0368e68598310ca6ffcd9c95095deaaeed5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 11:38:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6165ac640ef96b354b372f0bd44998aea3d2ff89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 13:52:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d116ecd7e9fa4f1a8ee9acb0a9433d17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73efd5b98fe644dba3ca2575ff510455 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb62bf0aa83647329a0815948a08b114 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb0c7b70da964568b0797f4bb35aa65b : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c999973a3f254ecfa393df801c56d529 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65b2d4606890422991f4dcfb0bc9baec : SUCCESS in 2m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0201eda599624bda92e6f8b7f5d6f0c0 : SUCCESS in 4m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7ae9ea7522c64033a34179d6d53a68a4 : SUCCESS in 5m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/768aecf188944ba1859015918e9f9815 : SUCCESS in 3m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d6127126f7a2408d957c5cdb421b0062 : SUCCESS in 5m 15s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/a69bf0eeb7034236b405d6cd9c73e440 : SUCCESS in 14m 03s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/0e1abd0e62804e3da623bac7063c7af6 : SUCCESS in 3m 15s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/edc72015a0ba462db262767f0fb17537 : SUCCESS in 4m 23s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/dc2b03e82bf249c08368a21c635d762d : SUCCESS in 6m 49s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5177531abb194e07bb4de4b4dc77d6b3 : SUCCESS in 17m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6072c661ab8845bc876117df51a96e22 : SUCCESS in 8m 40s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/30eadb13765546c8908d0cdea8a9df50 : SUCCESS in 7m 39s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/537087bcde274a0a97d44d44725e9e7c : SUCCESS in 11m 44s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/8cda4f3d530943bab9bc65e03b6fcda8 : SUCCESS in 3m 56s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/642e5349a220438aab89d9d63f1a7f11 : SUCCESS in 5m 55s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4be101d11f394fccbc3643e9b8c9eb14 : SUCCESS in 31m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/559edb24675843a189a87a5c9abfa80b : SUCCESS in 8m 01s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0ae59e09a7db4956afc72ee39f7ec2f4 : SUCCESS in 13m 06s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/a699ecf0b2614c37899c7af86f6fd9b8 : SUCCESS in 4m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/241507dfa4b240ba9b687df1e13442f6 : SUCCESS in 2m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f040ec988a7d4d22a6707fced6f41d2a : SUCCESS in 26m 32s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/2810b4a7f92c4e3aa7e1ea3e636a661a : SUCCESS in 3m 24s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5b075d11bb24462da530553438e49f83 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e3b7baacde4c408ebb036454de99a552 : SUCCESS in 5m 18s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/eaadb5674627409b918834996629c5c0 : SUCCESS in 17m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d0e1e80524624f5482d5865b7d64efe1 : SUCCESS in 6m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aed20b353c9741e49ed7c6257e21208b : SUCCESS in 1h 10m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98b946e649914e21bc31d1d2f011220a : SUCCESS in 1h 50m 18s","accounts_in_message":[],"_revision_number":1},{"id":"63510a2b794a182462794150eda17675bc12815b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 13:52:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"136e614007d1f34ba4c5ce281728ee0e5a9c08db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 13:54:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6c832ee77294cb7aa0df5c590f1bb9d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca952d11750747f2b0a5a9dd2dfba8b5 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"368b467dbbf78935f811cc83dfad30d0b55e2a10","revisions":{"368b467dbbf78935f811cc83dfad30d0b55e2a10":{"kind":"REWORK","_number":1,"created":"2026-07-01 12:02:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/995671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/995671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/995671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/995671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/995671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/995671/1"}}},"commit":{"parents":[{"commit":"e93565a60e34357734bf5c7e2bff23d49969addd","subject":"Merge \"Stop vitrage requirements sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e93565a60e34357734bf5c7e2bff23d49969addd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-01 11:52:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-01 11:52:28.000000000","tz":0},"subject":"update constraint for python-designateclient to new release 7.0.0","message":"update constraint for python-designateclient to new release 7.0.0\n\nmeta: version: 7.0.0\nmeta: team: designate\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5fcce3b82c485ee88614f4bf40ab10aec4a76d50\nmeta: release:Code-Review+1: Omer Schwartz \u003coschwart@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I58fa24709808b24c11766b042b1c330facb51059\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/368b467dbbf78935f811cc83dfad30d0b55e2a10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/368b467dbbf78935f811cc83dfad30d0b55e2a10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
