)]}'
{"id":"openstack%2Frequirements~995786","triplet_id":"openstack%2Frequirements~master~I84a35d51f79a07598553b1942935897fae580a39","project":"openstack/requirements","branch":"master","topic":"drop-sqlparse","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-07-03 07:17:42.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-07-03 07:17:34.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I84a35d51f79a07598553b1942935897fae580a39","subject":"Drop sqlparse again","status":"ABANDONED","created":"2026-07-02 08:53:49.000000000","updated":"2026-07-03 07:17:42.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5d28792815d4682dd181b60ab54e2927ee34edc9","_number":995786,"virtual_id_number":995786,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-02 10:34:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2026-07-02 20:48:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-02 10:34:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-02 15:26:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2026-07-02 20:48:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"eb6c0c4c5e2881610f0d69b1ce6d819a3ced581f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-02 08:53:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b161d3a5ad1d414b3064790e992f4d4f41d8648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 10:34:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/775c1fadb8b64c8a98477996578045f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d67923c21541418bdaa06f57e75c2c : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeffd566f2034088b674264184a68b01 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2a0f4f1d945473bb68deb984210e696 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/579b05389d3140fcb65d842e084a9679 : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4934a33fdb454654ba34a2ab98b24cd5 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a21d9c668e54f4c9e09a6b20dffd0c6 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/932e140594514748b7d07172a00b9e20 : SUCCESS in 5m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ef7a83ea2cd945afbb95e685e19c5e15 : SUCCESS in 6m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2eec3395216d49c9b7b1087bbb174560 : SUCCESS in 5m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6c2751278007463f85bb99b60cc4fbf7 : SUCCESS in 4m 55s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/c2f9f73f3bbf41799f60fb36935aefe6 : SUCCESS in 4m 43s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/5055e87c46c94e349ead3e7ef1bb472b : SUCCESS in 6m 37s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/f0f2a1935e2740438901247aabe36e62 : SUCCESS in 3m 44s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/21a90a14f21c4b0ea1591bf924f44deb : SUCCESS in 10m 26s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/8553cfbe8f114be9bdc684f8d3893f00 : SUCCESS in 6m 10s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/99db6414f8c64222a9c8ccf7688950a3 : SUCCESS in 7m 42s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/a535e9a397fc481baa596bc68a068e45 : SUCCESS in 7m 40s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/14dc6c8c77d64404b9e9644f2b3b77bf : SUCCESS in 16m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/21617abece0c4e52869e74decd4d4b28 : SUCCESS in 7m 38s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/80118b967d134f43a688fa87646755c4 : SUCCESS in 8m 23s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/81767be4d16f4444ac514bf45d2c61a3 : SUCCESS in 10m 08s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/2b7b4e43c5ff4fbcb9ffc342979aadf7 : SUCCESS in 2m 49s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/cee1171fedc34f2e8a51a291202f952e : SUCCESS in 9m 09s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/d56a23314ede46e9b5442ced62dc2098 : SUCCESS in 4m 48s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/d27cd748955f465e97a902a45f7c1b76 : SUCCESS in 4m 07s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/cbc17b3ca44442c29abe49feca092117 : SUCCESS in 9m 23s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/691fef2ec41a4ceeb6817acd9a69fa2d : SUCCESS in 26m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d80282615ebd47248d01eb39b1a8a4a9 : SUCCESS in 6m 22s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/c0bb6d0ab1a54e4685a9305fe7e15815 : SUCCESS in 13m 16s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/4d3a92ec386349a6aa1fc963187b3156 : SUCCESS in 3m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/94fc92f87f3a40b384aca81761922323 : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90bbc37ef69a4350903d734a8080e29c : SUCCESS in 24m 02s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/1e6d9fc6f515470a90b3349c476951dc : SUCCESS in 5m 36s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/595b0a7b092c47bf8208e8bdd25f8ff2 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cfb06c039d8047c99d99862301713589 : SUCCESS in 8m 03s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/707b28b3e391488e8cd51d0a7cdbb4dc : SUCCESS in 13m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8ea0cda14efd4606ad486208fef8babe : SUCCESS in 5m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d03102b3bf7d40ea94d029ea927a6466 : SUCCESS in 1h 05m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5e76127e9ae45a4bb8724c0db2bd646 : SUCCESS in 1h 38m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1e825cd80cad316dd7a694b45a7ae2d11cbd9913","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-02 11:14:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73c6027eea053d3a9d634ece69606ec17055c356","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-07-02 15:26:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76b7fa88619be184262e6d5bc7512de4b2364214","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-02 20:48:18.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5f0f39ee4d1e396c455b1c34d82d4ebf384499d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-03 07:17:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d28792815d4682dd181b60ab54e2927ee34edc9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-03 07:17:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2e5ebbc8a828e45466e42f61b196b04999902f0","revisions":{"a2e5ebbc8a828e45466e42f61b196b04999902f0":{"kind":"REWORK","_number":1,"created":"2026-07-02 08:53:49.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/86/995786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/995786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/995786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/995786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/995786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/995786/1"}}},"commit":{"parents":[{"commit":"e93565a60e34357734bf5c7e2bff23d49969addd","subject":"Merge \"Stop vitrage requirements sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e93565a60e34357734bf5c7e2bff23d49969addd"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-02 08:50:17.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-02 08:53:01.000000000","tz":120},"subject":"Drop sqlparse again","message":"Drop sqlparse again\n\n[1] already removed sqlparse from requirements but I suppose\naccidentally it was added again, see [2].\nIt seems it is not used, only references are pointing to\nkolla and rpm packages:\n$ beagle --server https://codesearch.openstack.org search -f grep sqlparse | grep ^openstack | grep -v \u0027/requirements\u0027\n/home/lajoskatona/venvHomeLocal/lib/python3.12/site-packages/beagle/app.py:16: UserWarning: pkg_resources is deprecated as an API. See https://setuptools.pypa.io/en/latest/pkg_resources.html. The pkg_resources package is slated for removal as early as 2025-11-30. Refrain from using this package or pin to Setuptools\u003c81.\n  import pkg_resources\nopenstack/kolla:doc/source/admin/image-building.rst:419:RUN {{ macros.upper_constraints_version_change(\"sqlparse\", \"0.4.4\", \"0.5.0\") }}\nopenstack/kolla:docker/openstack-base/Dockerfile.j2:173:\u0027sqlparse\u0027,\nopenstack/rpm-packaging:openstack/glance/glance.spec.j2:54:BuildRequires:  {{ py3(\u0027sqlparse\u0027) }}\nopenstack/rpm-packaging:openstack/glance/glance.spec.j2:133:Requires:       {{ py3(\u0027sqlparse\u0027) }}\nopenstack/rpm-packaging:openstack/keystone/keystone.spec.j2:47:BuildRequires:  {{ py3(\u0027sqlparse\u0027) }}\n\n[1]: https://review.opendev.org/c/openstack/requirements/+/922587\n[2]: https://review.opendev.org/c/openstack/requirements/+/910040\n\nChange-Id: I84a35d51f79a07598553b1942935897fae580a39\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2e5ebbc8a828e45466e42f61b196b04999902f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2e5ebbc8a828e45466e42f61b196b04999902f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
