)]}'
{"id":"openstack%2Frequirements~995795","triplet_id":"openstack%2Frequirements~master~Iadb829f20ea7a1914074f3111d0bdb3acbae550c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iadb829f20ea7a1914074f3111d0bdb3acbae550c","subject":"update constraint for heat-translator to new release 3.5.0","status":"MERGED","created":"2026-07-02 10:16:45.000000000","updated":"2026-07-02 19:30:05.000000000","submitted":"2026-07-02 19:29:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995795-new-release","meta_rev_id":"7a5d98742a3cc6e426445fd540fad8f25d8c3f2a","_number":995795,"virtual_id_number":995795,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-02 19:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-07-02 17:18:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-07-02 17:18:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-02 12:23:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-02 15:28:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2026-07-02 17:18:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c3a97bfe41fde95b6b7231a4522f91c513e4da43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-02 10:16:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19fcf263171c42fa51603cf7f20811086da3e8c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 12:23:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4211123474e04f41bb04d970d1a917d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e21e11d8ece6488e8efa37c1b56dc080 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58391ce02cd44d3397acbd65f43908a1 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a7a358e14ae400bb9d92a26a969d93f : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/427c644c5f8d4c21a75326d9dfb97d35 : SUCCESS in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2af715b0cb4848cb92830ea8ddd0f1c8 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b24e20203f2f4c75a200e66e6f24c07c : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76be2ba02f4348b48d43cfa79bc48d55 : SUCCESS in 3m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c098a4f2ae964202a417eb842a21fd2c : SUCCESS in 5m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7d1dfebf4496439e986b0ce543b7d7ea : SUCCESS in 9m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4e2042f91d134ba7acd61df7f380b5ff : SUCCESS in 5m 15s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/587d7c1b506047b48f10147cb90e05c1 : SUCCESS in 4m 31s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/79053e0f5505423cafa2d4531f4a2671 : SUCCESS in 6m 02s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8efc37b7c92d43c4843440cf2c7a5ed2 : SUCCESS in 5m 06s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/affe773d90c147c08644806fafe03bde : SUCCESS in 9m 06s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/db8f7c2cc937486e98f1163a0a47179e : SUCCESS in 5m 57s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/4f71df6a24f8411f81fa1dd9d8d95588 : SUCCESS in 4m 45s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/62693fc29d6b45b2b5be214e4a9674e5 : SUCCESS in 7m 05s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/0d7e7331bd5d4d3989233d09e59deda9 : SUCCESS in 17m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/43eaf99794194681b72ac1f985e2a843 : SUCCESS in 5m 42s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/72f4865fe3a24f55ac4c7525e7bb6b7f : SUCCESS in 4m 21s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/ec6e8cc16b4a4be8a1523b280a098800 : SUCCESS in 11m 28s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/2e1d48a81fa5490691763d1bae622a49 : SUCCESS in 2m 08s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/05853b279ca14c4b93d73552e71d7cb8 : SUCCESS in 8m 13s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/19214923a7914a58a37ba959fdee2de6 : SUCCESS in 5m 27s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/8a863b346a754576b9cedaff8e7a7b9e : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/b401be34f4bf42498d77f3812df68f85 : SUCCESS in 10m 14s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/732a0c2e70b84f95971bea78dcb653d1 : SUCCESS in 21m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c0a7eac69e394bc88d2c00cf22360b44 : SUCCESS in 6m 19s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b74209b8276f4d33ac24383137e7c16f : SUCCESS in 13m 12s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/24f27742236c43f3872fca291ce0991a : SUCCESS in 3m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/df3ee88db2ab4d1d8223a385898a42c2 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ab429f2a93f443786c805bbbeb779b3 : SUCCESS in 24m 50s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/35b7500941874275b9c28bd2bcbaa232 : SUCCESS in 5m 08s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/2fa246c94ab9442d9ddf199fe0edaf94 : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d8cab2e6c764306bb5936abc9d84b22 : SUCCESS in 6m 32s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/b4459d7a12d84c1b882934ed92e301f0 : SUCCESS in 13m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5c662b1f8f9f40248bffda757a41e4fb : FAILURE in 4m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db78b1ad39394d9c84a526a8c69ebb94 : SUCCESS in 1h 08m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5459a85c09a34d249144ab06ba82a8b7 : SUCCESS in 1h 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4f748e11c51b80e027a73856324ea6c0314e8e07","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-07-02 15:28:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4dc1931a83b8e5e5a9323945bcbadac71f616853","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 17:13:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f945ae0997594d70b25f2f57ae4adb91\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/182dfadcf6474e868d5253f7533be646 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea6282ceee4474c8d14e80a0a5b8f14 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a1dd03143e44b71a6f3c475dc35953f : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d29763561ebc4095a860a725712b2016 : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c5a8106215ee451f8998bf7116b1095f : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31b03a3466d04d10af6a5fbdc1d9e1f8 : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c184c7b9e22540f2b7418d2433ceae9a : SUCCESS in 3m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f048feda40244f6abfdea0f8a1c305df : SUCCESS in 3m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1713ed4f803245138d0ecf4700b9019f : SUCCESS in 4m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f6e6e5a9d58d426a80524cc7105f982a : SUCCESS in 4m 55s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/66f924ae26884caf83b5e56d05d68369 : SUCCESS in 3m 15s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/f873dd21a701414cbce34c5cef7b0911 : SUCCESS in 3m 25s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/fe1401aa5d264aca92b9bd1760dd5bcb : SUCCESS in 3m 59s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/9af20006dda94090a517e792a722575d : SUCCESS in 6m 43s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/00baae674b3f4fda9cb6a4c87cfed4cb : SUCCESS in 3m 34s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/a94f6e06efd8403097b5a08c7fad5221 : SUCCESS in 5m 54s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/421776e6b3614935a7b4f61151927fd9 : SUCCESS in 7m 09s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/0e775699a823471b95aaccffdda18dae : SUCCESS in 18m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/730b058ae77d4e758d18c58419742334 : SUCCESS in 10m 01s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/7cc38221ea7948478fcd829378e58ce5 : SUCCESS in 9m 54s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/d4f45290689747ceaf57a83b18fde274 : SUCCESS in 12m 35s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/be10671bd38e42c29faa44bae7d5efa8 : SUCCESS in 2m 00s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/0936b56bf2414318a7ec7969be6a2237 : SUCCESS in 8m 47s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/54f51b1065b1450289f11c23e8257a61 : SUCCESS in 5m 02s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/e5d12a14e4864bec936b467b16123566 : SUCCESS in 5m 42s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/bd770e50481542889b5aea14f8ad8270 : SUCCESS in 11m 08s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/40abc520890948b1bd75c8e7a131e688 : SUCCESS in 25m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f74e8f9442024a2c878a9a0bb7273f6e : SUCCESS in 6m 19s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/5119c4c7112748748fda235c964e8c8e : SUCCESS in 13m 48s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c3e7bcd31bfa4d93993a0120732061d2 : SUCCESS in 3m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/40255f953a084176a654ea551c134db5 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88e8239ab55744b3b145122840f9ec2c : SUCCESS in 32m 13s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0487fd123501437bbd0e03642d84ce46 : SUCCESS in 7m 08s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/2320cc06b98d46f08261eb0177fe4f16 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d31d06e6c2c44183b747f0119793a6e3 : SUCCESS in 4m 35s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/41049614595745af91e0bc8f55afd3d8 : SUCCESS in 18m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e188be4250fc480cbb0dfcc437ba7acc : SUCCESS in 7m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9a73904b1344358b0c65f6b9bb4963f : SUCCESS in 39m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/084cca264934407cb591f259ffc855c2 : SUCCESS in 1h 38m 19s","accounts_in_message":[],"_revision_number":1},{"id":"53302efcff04b455c71707126647897388b70c18","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-02 17:18:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"958c20578d9e5408f68c9bc76178d73e0502dc49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 17:19:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f725e7139caeede3d497434a4e14013ab4fdcd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 19:29:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91f34f295c04474c816a3e6012be5801\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08293f2bcac645cdb18e0b3f7cd4be21 : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/702dfe81e9eb49cabcaba6a1209b3b69 : SUCCESS in 2m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ebff7beddd4b4f4296647b352f7cfa0a : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9200c053c8974a7d9493d609b2d57407 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d35b3a8cb1e424696f754f895176a00 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/362102e38731404d8b57477eeb2956b1 : SUCCESS in 3m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5f98594aaa5a4a29bfa31efc506fd857 : SUCCESS in 5m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/634e6009036f426d967c4d52de92637f : SUCCESS in 6m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/47c1ddba838547b6a641211c6f9cb550 : SUCCESS in 5m 54s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/57d8096814a64fecb869c881c17d9027 : SUCCESS in 12m 02s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/f806704a978d49cd998eb264df1a7fdf : SUCCESS in 3m 44s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d151166d893546c0b7dfc35704736bc2 : SUCCESS in 4m 46s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/0a7a882828bd4b0c8fc93cd29472859b : SUCCESS in 8m 30s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/64645fe664c24b248fffa5bb6d2391f0 : SUCCESS in 15m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fe48a8fa3e4b4218bc26c8d001d85811 : SUCCESS in 7m 16s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/9b14016b6ac949deb09175be8378c359 : SUCCESS in 9m 10s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0431e35186c9459f9241a57cee6bc621 : SUCCESS in 7m 16s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/0fe728efb05e4b2f91e7d173e8da1482 : SUCCESS in 2m 00s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/9272696bcff94a8193403da77af6f4a4 : SUCCESS in 8m 13s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/6911dcb827e546e0b437eaea75a3a04a : SUCCESS in 28m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bd87714696c14197a23fba11f8b63246 : SUCCESS in 6m 37s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/ac4a4cdc523a4548ac2b7840104cb1c2 : SUCCESS in 8m 46s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/0a2638604975458b93450a8c7fd5da41 : SUCCESS in 2m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/94c193559e7249bdb940cce11ecdacb7 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3ce9421d33044a2872b2a8c8fbd35c2 : SUCCESS in 25m 22s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/7bf322a19cb341d1800fd20220d98597 : SUCCESS in 5m 34s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5a82bfbe7406442fb6bafc4783b884fc : SUCCESS in 3m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b3a155050a64e579737694869404c0a : SUCCESS in 6m 24s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/229b68bae45f4ba99486070574ee1524 : SUCCESS in 8m 05s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/277b98f3d88d4257a49f314c390a6725 : SUCCESS in 4m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f5d900fb109434ba54d9f953ce32687 : SUCCESS in 1h 04m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c73fc077fc144354beaf23cad39a4d70 : SUCCESS in 2h 04m 09s","accounts_in_message":[],"_revision_number":1},{"id":"de3b36a3a002400f502ef56cf0ffc8998ba0281d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 19:29:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7a5d98742a3cc6e426445fd540fad8f25d8c3f2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 19:30:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aecb4ba7ca564242b377c7d9938cee33\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa75b65aa1224b4dbfd7ebdba3b33f40 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75d596ffb03e92156dcf62a1d0dda63dc3a20651","revisions":{"75d596ffb03e92156dcf62a1d0dda63dc3a20651":{"kind":"REWORK","_number":1,"created":"2026-07-02 10:16:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/995795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/995795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/995795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/995795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/995795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/995795/1"}}},"commit":{"parents":[{"commit":"e93565a60e34357734bf5c7e2bff23d49969addd","subject":"Merge \"Stop vitrage requirements sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e93565a60e34357734bf5c7e2bff23d49969addd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-02 10:06:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-02 10:06:44.000000000","tz":0},"subject":"update constraint for heat-translator to new release 3.5.0","message":"update constraint for heat-translator to new release 3.5.0\n\nmeta: version: 3.5.0\nmeta: team: tacker\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nmeta: release:Change-Id: I79433d343aeeb42e90fe0e730bf647fb7ff73d5b\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Iadb829f20ea7a1914074f3111d0bdb3acbae550c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75d596ffb03e92156dcf62a1d0dda63dc3a20651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75d596ffb03e92156dcf62a1d0dda63dc3a20651"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
