)]}'
{"id":"openstack%2Frequirements~996441","triplet_id":"openstack%2Frequirements~master~I27af77991b850249d9fef3aa60c3645b83ad6cc4","project":"openstack/requirements","branch":"master","attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-17 12:30:47.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-15 17:27:02.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-15 19:28:26.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I27af77991b850249d9fef3aa60c3645b83ad6cc4","subject":"Do not fail on optional dependencies","status":"NEW","created":"2026-07-08 10:58:16.000000000","updated":"2026-09-17 12:30:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9904ac0eb30a6c14ec4085ed93558e0321e58ef3","_number":996441,"virtual_id_number":996441,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 02:21:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-09-17 12:30:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-08 12:57:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-07 10:11:19.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-08-07 10:11:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-08-20 18:08:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-09-17 12:30:47.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"6d5552277c7e52cf392f9a1a8440c0b0a8a2bbce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-08 10:58:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"180533f58d9ad1d10a7b7ab5eb16f920619b2040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 12:57:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eca425a50558416daff9e03d7be8e89d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b34d46fd8a436fabb6e17f31e75684 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bb7574abf4240efb30efe263bb3073f : SUCCESS in 2m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af3fdd56b1b1473eb72ec82782ef6367 : SUCCESS in 2m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e708c9ddff9c4dc8a487f498a343ff7a : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/97497c45169e49069842a7bf6d7a9d28 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/25c78523febc4cb38160aa62799faaea : SUCCESS in 3m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f773cda7954b45caa6345c241eac99e8 : SUCCESS in 1h 01m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffd12118cdf54bdb92adc87c2e3e9691 : SUCCESS in 1h 52m 57s","accounts_in_message":[],"_revision_number":1},{"id":"1d38faea8ec2f3a4742101d7b18cf9a165038f97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-09 11:12:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b13b65cc733ed121b09af53f906e040daf2f53cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 12:55:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466d20be8a194cecac6daea7f0fae567\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef596536692f45ed9ed797828f2e35e9 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f4a333b90af44beb38be872f28591d2 : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad32ab33f0ce46bf9f60d576a9d23624 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db4e166639df4e46bf942ce065cdd179 : SUCCESS in 2m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/67732c73afe44c6f9f97d460880cbfae : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/efc946814e7b487e89105b5c2b1c5770 : SUCCESS in 3m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a8fde13c28f437e8050c4417a7b1132 : SUCCESS in 1h 08m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2db34f14c66d4ef5a8c78c8a93878bc4 : SUCCESS in 1h 36m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3db2bc59e7e15745b55785addfb167980dd500a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-07 06:18:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"185f44a12a9cf3c53689151cd10570cb8a8c91ac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-07 06:19:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nThe following files contain Git conflicts:\n* openstack_requirements/check.py\n","accounts_in_message":[],"_revision_number":4},{"id":"818bf984d3ab7ba88b79773d677008d3351bdca2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 06:19:17.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 996441,4","accounts_in_message":[],"_revision_number":4},{"id":"6ddbd8444c204bed818bd594dae6f988ffe8a8b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 06:19:21.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 996441,4","accounts_in_message":[],"_revision_number":4},{"id":"aac8047afcb4feefcfd77925f83a829dfc5c0697","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-07 06:19:33.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f7554a9138f705e4625666e36f0ad7693e4bad01","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-07 06:19:39.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"7671f07aa0c2f38b22088bf161274daf987aadc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 06:19:41.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 996441,5","accounts_in_message":[],"_revision_number":5},{"id":"a0140c84b8fa7f1b259596b20d262674938aa269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 06:20:27.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 996441,5","accounts_in_message":[],"_revision_number":5},{"id":"6bc75ff4ace0f794029accc4ba1a8ff876863c41","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-07 10:11:19.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"42544e965ba4a1e4dcd0ba3efcfa1c3c7af72d9d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-07 10:11:34.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b7a951254a5769c92bb7954cddd5e1686741334d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-20 18:08:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"caa8383108beb1b4aa745597714bd62ef8ef5755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-15 17:27:02.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"5470c4986399d6ae0320f5322125025a6c9cc57b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 19:28:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3d6da8cc3154d41be6dd5e03e486846\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d2a70cc7d1a46d6921a91bdf66b2df9 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90270acb951647eb8adfdf7fea15a4af : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72e506b9270942bc95d6742af7988e17 : FAILURE in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7b5731c2c4846b7a46e22ccfaa4697f : FAILURE in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3488cfa10226407e9961c1c3c905ef26 : FAILURE in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44d8c232782f42539fbe326f743e8760 : SUCCESS in 3m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2cfeef300884fc385997f1b3f639628 : SUCCESS in 44m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb1bffa7a9b54f7db3d5fb011738466b : POST_FAILURE in 1h 55m 01s","accounts_in_message":[],"_revision_number":6},{"id":"43b10da8ef0d5976e479cd3f1a348371041be618","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-16 01:16:48.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"53a1c18eb008f8d6957197563dacfb50e5dea90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 02:21:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb328e6c432469ca4cf203b06a3ceab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dfa135c52454e8fbd13138eb31bb72a : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc3c496c752b4b73ac0bc9e06922fb54 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6540305a34e41d5b39ff32b8d5b2403 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7584265f04d4dd19eed585dce102808 : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/330ae64bc22c4c3a980d08126a010887 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ad56074e21d416e8eb28af33e0cfa63 : SUCCESS in 3m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8d12676e9fa43fc8ab01464aee4cbac : SUCCESS in 59m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ad37a4dea774c2080b4d26c196578c2 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":7},{"id":"6ae5588aa80a731c99696f14fcfc6a598d306003","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-17 06:43:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9904ac0eb30a6c14ec4085ed93558e0321e58ef3","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-09-17 12:30:47.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b23e101ff711ed87b1b7cfee75c695f95e9da3ee","revisions":{"daad38ec0a8199c45f30b0ad120074d4a97ec10e":{"kind":"REWORK","_number":1,"created":"2026-07-08 10:58:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/996441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/996441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/996441/1"}}},"commit":{"parents":[{"commit":"7a04938e7e6f130a007e17750e880b6ec3eadd07","subject":"Add respx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a04938e7e6f130a007e17750e880b6ec3eadd07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-08 10:57:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-08 10:58:14.000000000","tz":120},"subject":"Do not fail on optional dependencies","message":"Do not fail on optional dependencies\n\nChange-Id: I27af77991b850249d9fef3aa60c3645b83ad6cc4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/daad38ec0a8199c45f30b0ad120074d4a97ec10e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/daad38ec0a8199c45f30b0ad120074d4a97ec10e"}]},"branch":"refs/heads/master"},"2024ca2ec10a3b08fb9c025e92056cf9a9bb9901":{"kind":"REWORK","_number":2,"created":"2026-07-09 11:12:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/996441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/996441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/996441/2"}}},"commit":{"parents":[{"commit":"7a04938e7e6f130a007e17750e880b6ec3eadd07","subject":"Add respx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a04938e7e6f130a007e17750e880b6ec3eadd07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-08 10:57:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-09 11:12:32.000000000","tz":120},"subject":"Do not fail on optional dependencies","message":"Do not fail on optional dependencies\n\nChange-Id: I27af77991b850249d9fef3aa60c3645b83ad6cc4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2024ca2ec10a3b08fb9c025e92056cf9a9bb9901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2024ca2ec10a3b08fb9c025e92056cf9a9bb9901"}]},"branch":"refs/heads/master"},"1777a61d060ebbf12ce6f461825abf005e3c9ae5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-07 06:18:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/996441/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/996441/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/996441/3"}}},"commit":{"parents":[{"commit":"ddab97be8a4c5fe248db071fde9013ba530d7f72","subject":"Merge \"Add py-consul\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ddab97be8a4c5fe248db071fde9013ba530d7f72"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-08 10:57:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-07 06:18:45.000000000","tz":0},"subject":"Do not fail on optional dependencies","message":"Do not fail on optional dependencies\n\nChange-Id: I27af77991b850249d9fef3aa60c3645b83ad6cc4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1777a61d060ebbf12ce6f461825abf005e3c9ae5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1777a61d060ebbf12ce6f461825abf005e3c9ae5"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"7a04938e7e6f130a007e17750e880b6ec3eadd07","ours":"2024ca2ec10a3b08fb9c025e92056cf9a9bb9901","theirs":"ddab97be8a4c5fe248db071fde9013ba530d7f72","merge_strategy":"recursive","contains_conflicts":false}},"4503bb3ae352d9e579139aec576edfa6ce1b1d19":{"kind":"REWORK","_number":4,"created":"2026-08-07 06:19:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/996441/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/996441/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/996441/4"}}},"commit":{"parents":[{"commit":"3b80226b6025682b891c5ee0218b69f2f29b1214","subject":"Add support for PEP-735 dependency-groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b80226b6025682b891c5ee0218b69f2f29b1214"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-08 10:57:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-07 06:19:10.000000000","tz":0},"subject":"Do not fail on optional dependencies","message":"Do not fail on optional dependencies\n\nChange-Id: I27af77991b850249d9fef3aa60c3645b83ad6cc4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4503bb3ae352d9e579139aec576edfa6ce1b1d19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4503bb3ae352d9e579139aec576edfa6ce1b1d19"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ddab97be8a4c5fe248db071fde9013ba530d7f72","ours":"1777a61d060ebbf12ce6f461825abf005e3c9ae5","theirs":"3b80226b6025682b891c5ee0218b69f2f29b1214","merge_strategy":"recursive","contains_conflicts":true}},"6f778d37e512d7fe9b68e80cf8ae0aaaee94e098":{"kind":"REWORK","_number":5,"created":"2026-08-07 06:19:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/996441/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/996441/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/996441/5"}}},"commit":{"parents":[{"commit":"3b80226b6025682b891c5ee0218b69f2f29b1214","subject":"Add support for PEP-735 dependency-groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b80226b6025682b891c5ee0218b69f2f29b1214"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-08 10:57:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-07 06:19:33.000000000","tz":0},"subject":"Do not fail on optional dependencies","message":"Do not fail on optional dependencies\n\nChange-Id: I27af77991b850249d9fef3aa60c3645b83ad6cc4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f778d37e512d7fe9b68e80cf8ae0aaaee94e098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f778d37e512d7fe9b68e80cf8ae0aaaee94e098"}]},"branch":"refs/heads/master"},"6893e7e8af887bc171c90f1ce943da46fb210361":{"kind":"REWORK","_number":6,"created":"2026-09-15 17:27:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/996441/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/996441/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/996441/6"}}},"commit":{"parents":[{"commit":"9edc04289eca1a4c18ae86e4430416d82a5b07fa","subject":"Merge \"update constraint for os_vif to new release 5.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9edc04289eca1a4c18ae86e4430416d82a5b07fa"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-08 10:57:59.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 17:25:10.000000000","tz":540},"subject":"Do not fail on optional dependencies","message":"Do not fail on optional dependencies\n\nChange-Id: I27af77991b850249d9fef3aa60c3645b83ad6cc4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6893e7e8af887bc171c90f1ce943da46fb210361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6893e7e8af887bc171c90f1ce943da46fb210361"}]},"branch":"refs/heads/master"},"b23e101ff711ed87b1b7cfee75c695f95e9da3ee":{"kind":"REWORK","_number":7,"created":"2026-09-16 01:16:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/996441/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/996441/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/996441/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/996441/7"}}},"commit":{"parents":[{"commit":"9edc04289eca1a4c18ae86e4430416d82a5b07fa","subject":"Merge \"update constraint for os_vif to new release 5.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9edc04289eca1a4c18ae86e4430416d82a5b07fa"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-08 10:57:59.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-16 01:16:19.000000000","tz":540},"subject":"Do not fail on optional dependencies","message":"Do not fail on optional dependencies\n\nSome of the optional dependencies are quite specific to projects and\nwe may not want to maintain these fully in global requirements.\n\nChange-Id: I27af77991b850249d9fef3aa60c3645b83ad6cc4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b23e101ff711ed87b1b7cfee75c695f95e9da3ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b23e101ff711ed87b1b7cfee75c695f95e9da3ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
