)]}'
{"id":"openstack%2Frequirements~996511","triplet_id":"openstack%2Frequirements~master~I23578a2b59545ae9f8c77db0273e3fd5a9a8fae9","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-07-17 01:07:54.000000000","reason":"Change was submitted"},"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-07-17 01:07:54.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-07-17 01:07:54.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-07-17 01:07:54.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-07-17 01:07:54.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-07-17 01:07:54.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-07-16 22:18:24.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":["remove-enforce-scope-flag"],"change_id":"I23578a2b59545ae9f8c77db0273e3fd5a9a8fae9","subject":"update constraint for oslo.policy to new release 6.0.0","status":"MERGED","created":"2026-07-08 17:57:09.000000000","updated":"2026-07-17 01:08:47.000000000","submitted":"2026-07-17 01:07:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996511-new-release","meta_rev_id":"fce9b64a045449e16e76cd0c3bec6cc7b58b874e","_number":996511,"virtual_id_number":996511,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-17 01:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-07-16 17:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-07-16 17:14:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-08 18:58:31.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2026-07-08 21:14:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-09 07:18:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-07-09 11:55:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-07-09 12:01:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-07-09 14:37:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-07-10 14:05:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-07-16 12:17:11.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-07-16 17:14:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"bcdf91c3ee243bd5ec5c88485e87ccb7b034b7ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-08 17:57:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b1d6699231ca654ec39cd8f038df7608e1e0c45","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 18:56:19.000000000","message":"Hashtag added: remove-enforce-scope-flag","accounts_in_message":[],"_revision_number":1},{"id":"b4fddc27e4f94f8c52ecce1d1a498f54c6ece0f7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 18:58:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9724fe3d322fe097e7ed8ef7895d92db443e37ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 18:59:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"4ca8e50479a0b137038ad2273241d24a9ccc62fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 19:12:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"31bc687041c634cc0bb47076c4ffbac0f8dbbd64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 21:14:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3bbb2255a3f4c1f918c830496a11365\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db617501e1e74c21b7874c84a679cfd6 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be96ca0f16d4b778f6d822d081d2f10 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eccf03f59abd4adf972e975a7a776a1f : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9f922a369e44504ac638525d2c1d6de : SUCCESS in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/85a031e1b92d4df1858e9583c579e8df : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c28ecf4df8fb44e381316d9d92a56c2f : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71f0a2cec7314fa491329ad145e5e546 : SUCCESS in 3m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/890b452b0c9741beb760de27090ae30f : SUCCESS in 6m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/127d995145a7416a89e4ebba15eab358 : SUCCESS in 6m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a03261ff7781401d968bf7e358340698 : SUCCESS in 9m 51s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/6348016bb0c442369f2bbe2e4d590af2 : SUCCESS in 4m 39s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/6fd9e0f96bf44038b8c6fd2a0525be4a : SUCCESS in 5m 49s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/efc82ccef9804af4b06d16b933caf7a7 : SUCCESS in 5m 19s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/55630350075040a889c4e67f7024029b : SUCCESS in 10m 30s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/0481979f66d84647a413e200386544ad : SUCCESS in 5m 11s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/912729f8f05d4b209302828c06d33a5a : SUCCESS in 5m 40s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/1b1c86a40e1642248bbc7363cda03eee : FAILURE in 3m 38s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5c7187495b8f4d5cbf90e62ebac23133 : SUCCESS in 17m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29d058e7f7d34db8ad736f06eab5f5f3 : SUCCESS in 6m 35s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/d7c2b0d464d74e3c99bb73e17dab5889 : FAILURE in 8m 34s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/1faf293258344338bfa95218129b592b : SUCCESS in 10m 34s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/34b2948354a946efad2d03dd9b95035b : SUCCESS in 2m 34s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/4049d2d2b8c244af8a7d315967ddbae4 : FAILURE in 5m 12s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/5d1ccca2ad0f4c5ea330adc7b8310fe0 : FAILURE in 4m 16s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/8833dcaa7f4e45d689e498a1aa2a3ce9 : SUCCESS in 3m 03s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/5c9689ef3a6a4795952127c3a81b00c3 : SUCCESS in 10m 43s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/3a5fe1437a934bbbbbb0ef6b0715372e : SUCCESS in 25m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/df83391b516649cf8381696b168a7de0 : SUCCESS in 7m 31s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/d48a8f5e3f6f4fe58931fe55b6c0600e : SUCCESS in 15m 34s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/7d9d53e4ab034629b8137ee2148c6cb9 : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4749743d076a40c7be3a39c5515872a2 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e57500ee98047cfa5b3be4ed5a7d0d5 : SUCCESS in 31m 53s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/c9d1cf124e3543269a703d9a82ec5e69 : SUCCESS in 5m 17s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/8c59c33ddcb445d3aa3e1eab71a651e2 : FAILURE in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d42c57c7a1d40c4bea6f81929b4ef5d : SUCCESS in 7m 40s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/f5d6f6ed8e23429c85c22e6c92328afa : SUCCESS in 18m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/69d0d91c2ee14df1aaaf19fae73722e4 : SUCCESS in 5m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee1ccb00726047b785df288fb8f0797e : SUCCESS in 1h 11m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edc890b769e74b2ca5dfdce9b136607c : SUCCESS in 1h 55m 45s","accounts_in_message":[],"_revision_number":3},{"id":"7836e940e2407c2c8fc726df08dcae00697fc4e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 21:18:42.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"5378c0ff68580d0dcab501c578f8b1b5dc4fc51a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 21:18:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cee2fb0aa705b209c5c1b4ef104fe4cd064349e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 22:38:39.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"a3dca73f18ed47ac4ae9b8b3fa5b922181a57ad7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 22:38:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fd07edffbac172033e67e13a2e90310f0ea75c92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 00:05:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5e4edfd68494e29b645abc660cbe21c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35068d591fbc4cc8969dd3ba4b4bc531 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92cb2e62e135420682dde948c2e7f18b : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6067a721b4814e54a9fe004f571f2c22 : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0dd6dc604cc84d40a719ef743467c0ca : SUCCESS in 2m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/15dd1ec1ac194afb8203d73ff388c637 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a7c6912c568495fbc44b6c1178a33ca : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/633b3effbd8f4d46b94b073b4af05db8 : SUCCESS in 2m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dd8dc2d249de495c845114688b533804 : SUCCESS in 6m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/46787aac59cf4a5886cb2e67210aa3f6 : SUCCESS in 6m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/40091d96b7a54695928274908ccc798e : SUCCESS in 10m 20s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/d797b3fa0dde4a81a950ab46b8a44229 : SUCCESS in 4m 48s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/e979a32b3a6e4f898d1bd868f0de9c7d : SUCCESS in 3m 20s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/a2295e900fb440508ea2d5ae5533f788 : SUCCESS in 5m 24s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/a16c121ce10a48ac8217d1196cfa8f5e : SUCCESS in 8m 00s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/24dbea0c448647469f33cf942d5df05c : SUCCESS in 4m 01s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d50e383135e64f8e843bf85ddf69dd6d : SUCCESS in 5m 22s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/3ba1f4a130804e2aa624eaa9c5fd8919 : SUCCESS in 4m 35s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/bf9735f7719a419b827688188c88979b : SUCCESS in 18m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c7cded19f4d4c549b54d445213be3d4 : SUCCESS in 4m 10s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/344d89c4841140d2b6305dfaba6520f3 : SUCCESS in 7m 35s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3dbca70c4c7e47c4b361e15f8a89ad92 : SUCCESS in 11m 32s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/0971350957c6417b9b8f61e71197f0e3 : SUCCESS in 2m 51s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/ce84c794366840c3bd4a19416502536b : SUCCESS in 8m 22s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/6fbda873e4374c3990ace4ff766bf7ed : SUCCESS in 4m 15s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/cc05a0c8549b4d20b9e170c3385c8df2 : SUCCESS in 3m 04s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/67a1ddeebc164935a16ba70b08bbe9a0 : SUCCESS in 11m 11s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/1d53bc9e46294f8e91691d0ba7b85ee0 : SUCCESS in 26m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/de44c74cec2e413880fd3660a76dc0a8 : SUCCESS in 8m 54s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/56efc2bfc9db427aa25120b54475b501 : SUCCESS in 14m 03s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/3f12b59fa58a4144b34ad199901f349c : SUCCESS in 6m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/785981c9ec9a4a208a3bd69b4e9bac3f : SUCCESS in 5m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/907a6b3a87bb43a88b9ab3380e64ab71 : SUCCESS in 26m 04s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/ca20f64d98eb401d8c6828604b4d53cf : SUCCESS in 4m 56s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/b991a4fd50d44ec7ae1eee0c85d7cb4f : SUCCESS in 7m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/21e93d892e1048b49c644dfcbaf78839 : SUCCESS in 9m 00s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/40747dfc02334c6da2e676e8a26541e3 : SUCCESS in 13m 40s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/aede646ee1b44bd884c8879c9a054ec8 : SUCCESS in 7m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bec1020342cb4f8d873f8874a09a79be : SUCCESS in 37m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ea2a28929b04a99bfd8a567158d3226 : SUCCESS in 1h 21m 50s","accounts_in_message":[],"_revision_number":5},{"id":"bbcf77567de016687c08f8733f2734e128e4e955","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-09 01:53:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"38dd800354cb7fe5a32d2fcc42892bf3330defac","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-09 07:18:37.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a43e49197158707b71014d9bfaaebe1e5f3094fe","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-09 11:55:27.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"efdf2579f1d68f761e43a0a48d3f886db1e9782a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-09 12:01:07.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"079d324da725dd58d1f58661430ab820959bda37","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-09 14:37:25.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"dce6b99f5156939f1a7d5ca240099a266069961f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-10 05:34:33.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":6},{"id":"8fe79ab5025618e8521a9c31e67713b8ec3e1269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 08:10:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ffe7a7e5a214ea8a73f4553283e5d0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9cad354eb54ef6abec283cf4823770 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6fa749c4b444018094669c45252570 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe12682c0a424c88a8b1c6d1e2304ae3 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b60a0aafacae4cbe8cbc9601737970ab : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3948f0761aef4872a9cd1ec489bc578c : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14d6823df254429da6aaec856aea97ec : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7fd3de0663c4498c82738396314742f6 : SUCCESS in 2m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3ff202aeb092480a8be4973ce78590c4 : SUCCESS in 3m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b09f0737a4414ece9fe0e5132add2029 : SUCCESS in 3m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9236768a4dd8451fa42f3766a349c81b : SUCCESS in 5m 53s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/a80a6f75934f4ddf99c6160f56dbb81d : SUCCESS in 4m 43s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/f749ae04ff8a4c5fae4f31bb364dd28d : SUCCESS in 4m 44s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8b610edf06374d9d8f6b2fc74f4579ea : SUCCESS in 4m 57s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/9b3a4c00c4d54747ac1c917016299642 : SUCCESS in 9m 03s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/3c94c13ece2443ab92fed6bbe2537deb : SUCCESS in 4m 30s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/a8980a2758ff43839a2d3121649b108d : SUCCESS in 4m 52s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/017c318a56dd411fb03f4583a9844ff2 : SUCCESS in 6m 14s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/2fd0b6aa80c3434f8409ce3f17cd17e7 : SUCCESS in 17m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4a25ecf7e0f4ddf8ecbbf4b37e9a215 : SUCCESS in 5m 31s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/e6b6cbc1a4eb4b3fac40f1663a7bdef0 : SUCCESS in 4m 03s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/06cf191493db47ce8dc3ba8af388c538 : SUCCESS in 10m 46s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c8c0b67e207f4caaa6507a9574c20f08 : SUCCESS in 2m 44s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/0364c788b36347b0b6c896afc67ab009 : SUCCESS in 5m 30s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/3a51db5407da4fd98c9bac198f9bb740 : SUCCESS in 5m 05s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/c82af17503e7493693b35acfa946b834 : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/4190c8402cfa4681a03b18cf187a56dd : SUCCESS in 10m 35s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/a77203bb40874e92abe1366fd09131bc : SUCCESS in 19m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0f2c5c8aaa194c1c90f8da4ac311bc73 : SUCCESS in 8m 17s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/e52b3ecba5e24ece8c36854bee5e1a71 : SUCCESS in 13m 16s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/5254b158469a40199658c40aa64a6676 : SUCCESS in 3m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/91b2616a41a340b19d69becd2e18da95 : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e8d869a844c4c8c90030d558ac3d13b : SUCCESS in 24m 56s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/ba4499aeb5bc48838aba8ee862ed8ada : SUCCESS in 3m 11s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/4096538f4baf4ceca46851f3169b1ca1 : SUCCESS in 3m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e51940b452f844fb97406f4a9e7d74f1 : SUCCESS in 3m 33s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/e7fd57fe10f6443fa5ec1ef988ae98c1 : SUCCESS in 7m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2ff64d02f1f041a8b7ede5c7d3bc1796 : SUCCESS in 3m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3a867e84e0d49f1bebfbd554936617f : SUCCESS in 37m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f73e3ee2f9b34a1ebe27990e550d7ad1 : SUCCESS in 2h 27m 48s","accounts_in_message":[],"_revision_number":6},{"id":"2df58e334357a99fe6ee254cdbfd6a00bca89cd3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-07-10 14:05:37.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fb51892d2aa66f0af2aab2013a718957434423ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-13 17:05:23.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"51c3f05b6cc13ca60af3149b4246f8f23f303bca","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-13 17:05:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"22b84d14749e0a0275a3562531d01f53f1756331","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-13 18:00:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1695792344bd5edf9b71a7a370d86a89b3f67669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-13 18:00:57.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"f1e64d87d1251e9988e034ccd35efe423911336b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 19:32:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71eaed6c78d7415c82cd4793fbd679f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a1d2ac3841a47f98a889632d07529ed : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b51eacac694474b9dcec0039e133f0 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a10440d83db04543be78c79759c397d2 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e9182728fd942beb128d9cafa5d0340 : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8448b2c3d495422ba53d079ef84ac806 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3bb5877e77144429dbf00a267555306 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7cfd122cd58241db810640c3c2b6a843 : SUCCESS in 4m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e0680a12f8764b52802818459b03e35e : SUCCESS in 6m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ff37b40a706d4174bb36f110138725c2 : SUCCESS in 6m 19s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ba0a4188c1c44664b3b8e3678db6ce85 : SUCCESS in 6m 07s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/c2633a3197744409a119d544368b8979 : SUCCESS in 4m 34s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/ad04a426759c40728a8a3d62610bbf69 : SUCCESS in 3m 32s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/3ae529118a7643cbb3f3c832b8376cb1 : SUCCESS in 5m 19s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/917b577780ea44d18420528b8f31f70f : SUCCESS in 10m 31s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/14e212e46cdd4bc4995ec7777abf6fa3 : SUCCESS in 4m 25s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/66c6c4c225004884822cdbf1df72acb3 : SUCCESS in 4m 54s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/653b8edd80524d4c8fbf37bd6ea1a52d : SUCCESS in 6m 17s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/95d732a494dd481bada857918e3efa1d : SUCCESS in 20m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5080757180604687b6bb7c593f739809 : SUCCESS in 6m 56s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/39a70ab36c0d43c186bc6d82426be7fe : SUCCESS in 6m 53s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/d3a35d3d99684afe8a82de4ebbd0949f : SUCCESS in 11m 31s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e1ba5ffc93094220acd21b12e84fe064 : SUCCESS in 2m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/b6bdc782276344e082af7384d60050ba : SUCCESS in 6m 23s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/915acc5b3c154ca7ae15227cd68b1319 : SUCCESS in 4m 24s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/8ad9db783a974690980261c8efce4b15 : SUCCESS in 5m 18s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/4a2954e8a3aa4784a1fc1c46eecbd827 : SUCCESS in 9m 16s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/178d289474a745b0ac949682ba4ca316 : SUCCESS in 24m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/faf37f773c3d4a46b05ca1f66cc671c7 : SUCCESS in 8m 30s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/cbfa3f3c92b04cbe8ce597cdb7b05409 : SUCCESS in 9m 19s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/44d18a78dc60490c8ec828e4e0b82581 : SUCCESS in 2m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3bc35412b8c040f1a5a7a08f50b9dd53 : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/583fb1c1f19c40d0a7656259c664771e : SUCCESS in 26m 55s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/7ce1961e435e4b50b34896660fc35ee9 : SUCCESS in 5m 28s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/3f942757127941af86033cde14ed988e : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad9ed5a5739545b5826633cadc2375c7 : SUCCESS in 3m 28s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d2fc34321b1645cc9950e2073db4eb9c : SUCCESS in 18m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ca24050b9f0244b8a767c9ff6432afa2 : SUCCESS in 5m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b65af81b7b0b4fda88f5fc02f91ec627 : SUCCESS in 58m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf4978e828a64b41901980de677a1664 : SUCCESS in 1h 28m 54s","accounts_in_message":[],"_revision_number":8},{"id":"0ab9948fdb45923e3b2505682000898ea7822987","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-16 12:17:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0fb2b4116eb1f47b7515db12e05f09244bccc8b0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-16 16:35:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f971cd5d71433b3abaf783c6038f29e2b006a205","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-16 17:14:13.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"43135f932e0c5fb93f5432ca11ac0505b44c3450","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 17:15:14.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"35e0a774b3b82c13b993f4b380de129fe7df27d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 18:38:33.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42c960940b914725af54fd62d03234a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037bf85e51b84224adc7413264fc4ed7 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f282a6ff5cb2417eaaaf106bb6eb8e69 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4d3d74481ac4a9bae407aecccdbaf1f : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01372671097a43e5809c549aa5d80a5b : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7124af581427422aabf23460e0f71880 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/67cf8759fd494f839637a619f019b7c3 : SUCCESS in 2m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/39f32197067a476182f6924673b9a8c0 : SUCCESS in 5m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6722e4e1b41a40269763c874d00c2b49 : SUCCESS in 11m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6d36f5b9f12c4756b6f9a81b2b46b547 : SUCCESS in 5m 45s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/46c4028d75d642d19cc661cb967b536f : SUCCESS in 10m 04s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/e3dcd0dedcdb4156822ce348a3586b41 : FAILURE in 7m 50s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/992ba0d99e3543a9b49c6f3d5c547630 : SUCCESS in 4m 57s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/1bdc13eba7884b958930b1d7bc12208b : SUCCESS in 5m 39s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/6415a59495a345118c0321296b9f808b : SUCCESS in 17m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/538eceaeb0b7451fa0754304686d4428 : SUCCESS in 13m 19s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/607ffe5abbd04f9a81108d127f5c9f82 : SUCCESS in 8m 48s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/13ad93552dff48e9b8c03d93b87426bf : SUCCESS in 11m 58s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/fe6e1fc846454a42a7c9ad990dc118c1 : SUCCESS in 2m 22s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/5a607f3e2de14c6788831538e1daf14f : SUCCESS in 9m 49s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/afe948e7de0640029a640afd99cc1e7a : SUCCESS in 22m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b341345cbafd4610a6139ca5e17296b6 : SUCCESS in 6m 34s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/7d1427ba53c34c24bc7ffd94544c41f0 : SUCCESS in 12m 44s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/bb6363a67cc54da79785463ac14892ef : SUCCESS in 4m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/95a7baa0d7e644b4a49093ab66573b64 : SUCCESS in 6m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8c3992d9dc545cb9fdd9137fcfc1c9b : SUCCESS in 30m 53s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/ff6d916480144ba2a807f5ef542dafb1 : SUCCESS in 5m 23s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f6769ea1af844dc7872086bdb25ce18d : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7da661bc267848b791e77fc9747bf85e : SUCCESS in 6m 46s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/10378623e7484ed5a635ceb7fa4d0168 : SUCCESS in 13m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/27d250505b194db08879d7a6df5e8ea9 : SUCCESS in 7m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff6e85cd677f4dedb1ad0ba1ef3925f1 : SUCCESS in 1h 01m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a806d14f9a3a4ba99e28760a4741e7b6 : SUCCESS in 1h 21m 34s","accounts_in_message":[],"_revision_number":8},{"id":"9312ced634ae426afbc09acfa8b91a701fd3dbe7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-16 22:18:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"92fbe40901993c4c09ea17e49b515ea0b785c529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 23:54:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed71fe4a14e54aee8d5221987d509a0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdc2ed7987c4444db6b6d907e1ed3c63 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caaf3a96da18489e8887b46b6252deed : SUCCESS in 2m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29932108be4d469295f1b8fbee51e5a4 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f54b2a3e179a4fda94ad8d45341200f2 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6f52c53162254f7693d1b11fe8d9469a : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83157499ebe14dbfbd6f532ad05dae03 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a93dff623e4640278e9b1a683172a606 : SUCCESS in 3m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2a76f53561ae4b318d63ccb27ce8ffb4 : SUCCESS in 3m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b5b2249eb2f242dd8bacb353dd3a7ef3 : SUCCESS in 9m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/916301c1328e487d88b85cad647da410 : SUCCESS in 5m 06s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/a209c750c48a4bedb622070803e35e33 : SUCCESS in 3m 57s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/2cba4d37cf2d43529a02efe7c9ba41f6 : SUCCESS in 5m 31s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/da79f5a8d88d4c04b2f97d944d995a8e : SUCCESS in 5m 09s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/f7c977eb8fb34de893db8708079ad7ec : SUCCESS in 10m 06s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/88c61626dcc247f6b46ac89662598d87 : SUCCESS in 4m 03s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/8c60f8c365254d5eb80e061272c8aab7 : SUCCESS in 6m 16s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/fe616e7de4df4ebeae26ee4b8c4d55ed : SUCCESS in 5m 16s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/8d719813145b49d9be098518824d5631 : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7bfd7f4925048e698e0544bd2643c2e : SUCCESS in 5m 31s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/f800395c13d14a1eac395c7a31760032 : SUCCESS in 7m 34s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3f971e1347b14ed4955db54d2f2a1cfc : SUCCESS in 10m 27s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/6fa7a7a216ae4f6593dbd2c9e12d2405 : SUCCESS in 1m 53s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/6f96650d9bc54c46bb133acae65ac527 : SUCCESS in 6m 25s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/47acb81d17b842fc9b1ef95580b70d62 : SUCCESS in 4m 00s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/1c9a28f368134844a037745ad83ecd94 : SUCCESS in 2m 50s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/7c856b3d09ea4c2f9110780b5b703bde : SUCCESS in 10m 43s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/66d7ef8b62464582870ff18996974805 : SUCCESS in 16m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5b54d4b46d844f46ad133f0ec56b2658 : SUCCESS in 3m 29s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/6c0bf6f7561c43fbb71d6eefe99f273b : SUCCESS in 13m 53s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/195cac743b994841a88d583b20b67638 : SUCCESS in 3m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/49f2d46829544952b52af1aa1746e4d4 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73c4ed0771a248d7b8279ab3c9826cc6 : SUCCESS in 26m 46s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/deaa93f2fd694d2aabe00c1ba6b3521a : SUCCESS in 3m 30s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/63d16576d5ad4561ae04d9d56a0dd63e : SUCCESS in 3m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/825dd6fb40434ccdae90d3e26d407d4a : SUCCESS in 6m 19s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/ab0af001a408415095c098a1eef5944e : SUCCESS in 8m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b102a9f87fb04bf6ba43d609a31c3f86 : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2a23651cba4499bb0f537a6aa48a762 : SUCCESS in 56m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36e3f062022742b29eb762712061ab7f : SUCCESS in 1h 30m 55s","accounts_in_message":[],"_revision_number":8},{"id":"a8f27727f7cac13f78d74df24ed3b0eb99370828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 23:55:33.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"3bd4c579d6aa78c9d22a3b7651fe570a8f8a45ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 01:07:53.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d975336f10604b9b95b3f4aa01b5829d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e673974debc41bcb7067fa9ef80e179 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fc9138680ed4e7983dffa3c6616c058 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/06ea3cf1a38e4b9b92af66f7ffec7579 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bee485f0ebd147fe89ac918ebeeb50f4 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7995934f0d51470484c640a09daafc7a : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2af38d5e465444e69b7bc3ab11d4b68e : SUCCESS in 1m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d4b12aa0cfb5412ba57b4c55342dd30b : SUCCESS in 3m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/51f3afdb74c443c3a58192c68a3c33b9 : SUCCESS in 3m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0623623f54914e389a685a3dcc48ff9c : SUCCESS in 3m 00s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/afb642e757f8446592f0e4ec7ff4dc63 : SUCCESS in 10m 31s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/58b6d6e9e73948fdb1c92c26a80a0a3e : SUCCESS in 4m 20s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d86c41e0ba0d4008a40f0d2d4012d14e : SUCCESS in 6m 24s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/85cd088b8b4c4db283fb6e907fffb480 : SUCCESS in 3m 24s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/2077bd6211b847278be76a1e67615b57 : SUCCESS in 13m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e52fc072644b4da79495b0b4090e8593 : SUCCESS in 6m 02s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/1264b2b174aa490b8e90b1e829ae8f0b : SUCCESS in 4m 27s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/7533c2864b8341249e5f2c69526d7f0d : SUCCESS in 10m 42s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/0795ebcc38e343b7a66154ffdf13c0a8 : SUCCESS in 1m 32s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/f1fbe24c152d4da898b8491f0255f4b9 : SUCCESS in 5m 01s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/78d29ae807bc493caaea578bbdf30aaa : SUCCESS in 22m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a636ddb7b7394cffa1242b0ff8af8f38 : SUCCESS in 8m 07s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/f213eab2f474474d8399eedc053c75c4 : SUCCESS in 13m 13s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/9749b1819ebd4591ba9570d12e220b29 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f93e03e9812a41c28d6b61a65a133aa1 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fab47f46dfb5477894b0829d1978d16c : SUCCESS in 26m 33s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/62158c1274fb4d59bc69c952248b0a40 : SUCCESS in 4m 35s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/928d532edcc74cfdab4a9422a78316f4 : SUCCESS in 4m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/850063a555994345b268a57a3fb2415a : SUCCESS in 6m 24s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/6a70e848ce4649778720456143c81925 : SUCCESS in 18m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b5f2e7765b184f3bb875663670f490af : SUCCESS in 3m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb43362ad7db4ecda3fb864ab70e74f4 : SUCCESS in 43m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab6d74076bfe43a984b54c0ca6e3fd0d : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":8},{"id":"aa6d569281660d8da50cc67abc7311d9781ec421","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 01:07:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"fce9b64a045449e16e76cd0c3bec6cc7b58b874e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 01:08:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92442a9e14164c7d85be315094a383e6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e513acabbaf4b62af0a59262c6489f6 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e4c9eded3c914df8c181b42d321b67c4796631b6","revisions":{"21177a92da0ccdb0ec62e37e1b844f5e2bad1ca1":{"kind":"REWORK","_number":1,"created":"2026-07-08 17:57:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/996511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/996511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/996511/1"}}},"commit":{"parents":[{"commit":"7a04938e7e6f130a007e17750e880b6ec3eadd07","subject":"Add respx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a04938e7e6f130a007e17750e880b6ec3eadd07"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-08 17:47:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-08 17:47:07.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 6.0.0","message":"update constraint for oslo.policy to new release 6.0.0\n\nmeta: version: 6.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Commit: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Change-Id: I1a21b336004295a3613ba582faf37227f520a61c\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I23578a2b59545ae9f8c77db0273e3fd5a9a8fae9\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21177a92da0ccdb0ec62e37e1b844f5e2bad1ca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21177a92da0ccdb0ec62e37e1b844f5e2bad1ca1"}]},"branch":"refs/heads/master"},"2ff32d0d85ad56e7ca74e0df48f8f5aa70565b28":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-08 18:59:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/996511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/996511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/996511/2"}}},"commit":{"parents":[{"commit":"2caf755950ac80873991a06cd5a62ec393c10553","subject":"Test nova cross functional job with master placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2caf755950ac80873991a06cd5a62ec393c10553"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-08 17:47:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-08 18:59:26.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 6.0.0","message":"update constraint for oslo.policy to new release 6.0.0\n\nmeta: version: 6.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Commit: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Change-Id: I1a21b336004295a3613ba582faf37227f520a61c\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I23578a2b59545ae9f8c77db0273e3fd5a9a8fae9\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ff32d0d85ad56e7ca74e0df48f8f5aa70565b28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ff32d0d85ad56e7ca74e0df48f8f5aa70565b28"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"7a04938e7e6f130a007e17750e880b6ec3eadd07","ours":"21177a92da0ccdb0ec62e37e1b844f5e2bad1ca1","theirs":"2caf755950ac80873991a06cd5a62ec393c10553","merge_strategy":"recursive","contains_conflicts":false}},"c2a1380c656c21edab551ec4bda34387a523bfb3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-07-08 19:12:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/996511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/996511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/996511/3"}}},"commit":{"parents":[{"commit":"2caf755950ac80873991a06cd5a62ec393c10553","subject":"Test nova cross functional job with master placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2caf755950ac80873991a06cd5a62ec393c10553"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-08 17:47:07.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-08 19:12:01.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 6.0.0","message":"update constraint for oslo.policy to new release 6.0.0\n\nDepends-On: https://review.opendev.org/c/openstack/keystone/+/993852\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/995405\n\nmeta: version: 6.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Commit: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Change-Id: I1a21b336004295a3613ba582faf37227f520a61c\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I23578a2b59545ae9f8c77db0273e3fd5a9a8fae9\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2a1380c656c21edab551ec4bda34387a523bfb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2a1380c656c21edab551ec4bda34387a523bfb3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bc0115f30d46d9c87fba09dd47388975becbf127":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-07-08 21:18:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/996511/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/996511/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/996511/4"}}},"commit":{"parents":[{"commit":"2caf755950ac80873991a06cd5a62ec393c10553","subject":"Test nova cross functional job with master placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2caf755950ac80873991a06cd5a62ec393c10553"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-08 17:47:07.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-08 21:18:42.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 6.0.0","message":"update constraint for oslo.policy to new release 6.0.0\n\nDepends-On: https://review.opendev.org/c/openstack/keystone/+/993852\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/995405\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/996546\nDepends-On: https://review.opendev.org/c/openstack/heat/+/996550\nDepends-On: https://review.opendev.org/c/openstack/manila/+/996539\nDepends-On: https://review.opendev.org/c/openstack/magnum/+/996540\n\nmeta: version: 6.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Commit: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Change-Id: I1a21b336004295a3613ba582faf37227f520a61c\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I23578a2b59545ae9f8c77db0273e3fd5a9a8fae9\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc0115f30d46d9c87fba09dd47388975becbf127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc0115f30d46d9c87fba09dd47388975becbf127"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ebcd33c5b0891391c72e3b16e8a55423beb6c63e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-07-08 22:38:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/996511/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/996511/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/996511/5"}}},"commit":{"parents":[{"commit":"2caf755950ac80873991a06cd5a62ec393c10553","subject":"Test nova cross functional job with master placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2caf755950ac80873991a06cd5a62ec393c10553"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-08 17:47:07.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-08 22:38:39.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 6.0.0","message":"update constraint for oslo.policy to new release 6.0.0\n\nDepends-On: https://review.opendev.org/c/openstack/keystone/+/993852\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/995405\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/996546\nDepends-On: https://review.opendev.org/c/openstack/heat/+/996550\nDepends-On: https://review.opendev.org/c/openstack/manila/+/996539\nDepends-On: https://review.opendev.org/c/openstack/magnum/+/996540\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/996559\n\nmeta: version: 6.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Commit: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Change-Id: I1a21b336004295a3613ba582faf37227f520a61c\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I23578a2b59545ae9f8c77db0273e3fd5a9a8fae9\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebcd33c5b0891391c72e3b16e8a55423beb6c63e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebcd33c5b0891391c72e3b16e8a55423beb6c63e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7e23877444f8037327948afeadffd721740d0f03":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-07-10 05:34:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/11/996511/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/996511/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/996511/6"}}},"commit":{"parents":[{"commit":"e021ea3bd1637d83fe3e17caf2b6946b17537f24","subject":"[to be reverted] Test nova cross functional job with master placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e021ea3bd1637d83fe3e17caf2b6946b17537f24"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-08 17:47:07.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-10 05:34:33.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 6.0.0","message":"update constraint for oslo.policy to new release 6.0.0\n\nDepends-On: https://review.opendev.org/c/openstack/keystone/+/993852\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/995405\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/996546\nDepends-On: https://review.opendev.org/c/openstack/heat/+/996550\nDepends-On: https://review.opendev.org/c/openstack/manila/+/996539\nDepends-On: https://review.opendev.org/c/openstack/magnum/+/996540\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/996559\n\nmeta: version: 6.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Commit: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Change-Id: I1a21b336004295a3613ba582faf37227f520a61c\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I23578a2b59545ae9f8c77db0273e3fd5a9a8fae9\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e23877444f8037327948afeadffd721740d0f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e23877444f8037327948afeadffd721740d0f03"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"2caf755950ac80873991a06cd5a62ec393c10553","ours":"ebcd33c5b0891391c72e3b16e8a55423beb6c63e","theirs":"e021ea3bd1637d83fe3e17caf2b6946b17537f24","merge_strategy":"recursive","contains_conflicts":false}},"6f0ee20105a0c1e5cdec58178fe1132ac9188616":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-07-13 17:05:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/996511/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/996511/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/996511/7"}}},"commit":{"parents":[{"commit":"f09f79c259a8fa6217230b255cfa418d0e7377d9","subject":"update constraint for python-openstackclient to new release 10.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f09f79c259a8fa6217230b255cfa418d0e7377d9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-08 17:47:07.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-13 17:05:23.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 6.0.0","message":"update constraint for oslo.policy to new release 6.0.0\n\nDepends-On: https://review.opendev.org/c/openstack/keystone/+/993852\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/995405\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/996546\nDepends-On: https://review.opendev.org/c/openstack/heat/+/996550\nDepends-On: https://review.opendev.org/c/openstack/manila/+/996539\nDepends-On: https://review.opendev.org/c/openstack/magnum/+/996540\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/996559\n\nmeta: version: 6.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Commit: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Change-Id: I1a21b336004295a3613ba582faf37227f520a61c\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I23578a2b59545ae9f8c77db0273e3fd5a9a8fae9\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f0ee20105a0c1e5cdec58178fe1132ac9188616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f0ee20105a0c1e5cdec58178fe1132ac9188616"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e021ea3bd1637d83fe3e17caf2b6946b17537f24","ours":"7e23877444f8037327948afeadffd721740d0f03","theirs":"f09f79c259a8fa6217230b255cfa418d0e7377d9","merge_strategy":"recursive","contains_conflicts":false}},"e4c9eded3c914df8c181b42d321b67c4796631b6":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-07-13 18:00:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/11/996511/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/996511/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996511/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/996511/8"}}},"commit":{"parents":[{"commit":"f09f79c259a8fa6217230b255cfa418d0e7377d9","subject":"update constraint for python-openstackclient to new release 10.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f09f79c259a8fa6217230b255cfa418d0e7377d9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-08 17:47:07.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-13 18:00:57.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 6.0.0","message":"update constraint for oslo.policy to new release 6.0.0\n\nDepends-On: https://review.opendev.org/c/openstack/keystone/+/993852\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/995405\n\nmeta: version: 6.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Commit: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Change-Id: I1a21b336004295a3613ba582faf37227f520a61c\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I23578a2b59545ae9f8c77db0273e3fd5a9a8fae9\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4c9eded3c914df8c181b42d321b67c4796631b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4c9eded3c914df8c181b42d321b67c4796631b6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
