)]}'
{"id":"openstack%2Frequirements~996517","triplet_id":"openstack%2Frequirements~master~Ieb110e4456768d57ea89f2b412c47c000f1d53ff","project":"openstack/requirements","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-07-10 05:34:06.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-07-13 18:01:30.000000000","reason":"Change was abandoned"}},"hashtags":["remove-enforce-scope-flag"],"change_id":"Ieb110e4456768d57ea89f2b412c47c000f1d53ff","subject":"[to be reverted] Test nova cross functional job with master placement","status":"ABANDONED","created":"2026-07-08 18:55:53.000000000","updated":"2026-07-13 18:01:30.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a096fb6c2b199409f46768aa475dbcfda736be47","_number":996517,"virtual_id_number":996517,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-09 20:03:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-07-10 05:34:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-08 21:06:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-09 12:16:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-07-10 05:34:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"72b7a934eced501deeee9075f7dd1189e1716c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 18:55:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0d061dcd28e50b4953c515e2aac27ada90bfa5c","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 18:56:38.000000000","message":"Hashtag added: remove-enforce-scope-flag","accounts_in_message":[],"_revision_number":1},{"id":"bdb03df9c5364f7b6e8a3e4a50d61884fdad4773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 21:06:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25eee9cb2d304df582f5a0896f804a33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0af3d904d054e08b35dc4c2893acc22 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87376684b9e845d089a677c1797db852 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17f1af87bb7845b0988ad6abd77198e5 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/819c294f29f849d9ab387501d4e9e34d : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/343a987b9baa4bf6b973db025910fc70 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fdc67dd15f084fd5b634f5e0ffdc2134 : SUCCESS in 3m 47s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/4687ea2a2fa04801895fdb2366415b29 : SUCCESS in 4m 42s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/02a17f05efb943ce83aa0b59765611b8 : SUCCESS in 7m 26s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/5e615accba4d477daaaa7231327ae230 : SUCCESS in 5m 18s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/42b48cd2d1594f0fbacbc1c5c823570a : SUCCESS in 11m 13s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/fcd28280e8e64f67a10f90cdf7b5fa99 : SUCCESS in 5m 03s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/3ac964c89527472986846f838c97bf28 : SUCCESS in 5m 00s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/b2bd336c10ac49b9868452240aa12e87 : SUCCESS in 6m 32s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5d641fd03cca471bb7379093e95cec69 : SUCCESS in 17m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b2341d01dce44d33bb81020b147ab652 : SUCCESS in 11m 06s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/c037a657a703419dba26045d7e0e22fb : SUCCESS in 6m 22s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3dd157fdac27423c9a8d444530dffcd1 : SUCCESS in 10m 33s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/6e0cd36ffbad4a6d969e53aed3f50fa3 : SUCCESS in 3m 02s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/5fce6ab387ff4fc69c82acb464750664 : SUCCESS in 8m 41s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/0c2dbffabd794911a6d334f16b550863 : SUCCESS in 4m 01s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/b2e2099ac13c4516848581ac2d58b345 : SUCCESS in 5m 31s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/1ec26f054749414f83d0f9a2210996d7 : SUCCESS in 9m 19s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c4686aa62e6845b19d7f2d8bde6f4901 : SUCCESS in 24m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0766041413d543cf9ef0692968d519fc : SUCCESS in 6m 52s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/41f55e4d8adb40eaa21dabf9eaecb0f0 : SUCCESS in 14m 37s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/b8cee48764ca42d09f183d91e9b4ee23 : SUCCESS in 4m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/08552b18f49f40deac486b20f292a678 : SUCCESS in 2m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eba6c9a49cbd4b728bc68c0c46cda5c8 : SUCCESS in 26m 40s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/698bcb0dcece49538089fce80741bcd7 : SUCCESS in 5m 09s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/4c0999f912404d73910856524ed3dba9 : SUCCESS in 3m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4feb038f0b454c2f8411f5be357c5b3f : SUCCESS in 6m 45s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/4de7ecb6b27f40c0adc0e6d3cbb34fda : SUCCESS in 18m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e03725a588504456984b57bfee472302 : SUCCESS in 5m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f88206364f74365865da2a33b2cd606 : SUCCESS in 57m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9972569a851c45f08867978ac8ebe7a7 : SUCCESS in 1h 58m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9b5dbec1c426b24505c5bb086826c05b75b39134","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-09 12:16:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95bfbf3bb054ecfc1347aaa8d8326ca68a260c3f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-09 18:07:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19fb4bd76ab1b76e5dd30588d8b23be70b6e7dd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-09 18:15:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a42c9fb4e83f6a40d9c44a165b0b06f03a55bd10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 20:03:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c279c50892b4a408f3a6244cd588449\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5e57e0c97724c24816164d832db625f : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dac5566ebe9402b94195f5752db223f : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/204bf26781234ae583511dabcf73e794 : SUCCESS in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14063ada608048e89d69a4df16876356 : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fdc47c560f4340558f70f8c60806eb6a : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ccf62636a0e455a975e1526fd49eba4 : SUCCESS in 4m 47s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/628b7ef5e3d44ff695426c37255fa034 : SUCCESS in 6m 53s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/260f28ca1f5f4ddbbc3335a15bd7f1dd : SUCCESS in 7m 09s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/96ee376d988f426b9e5df0ff235c1272 : SUCCESS in 5m 00s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/afc397f5f5c343d4b11eff0159183701 : SUCCESS in 11m 31s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/2324f67757e64f48b4f15330d99c5120 : SUCCESS in 4m 58s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/ecc179ec895e4ff184adc169412eb8cb : SUCCESS in 3m 18s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/7b3506719606444eaea2fb85cfda13b0 : SUCCESS in 7m 06s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/b237deba8ac04f00bccece5118d6f843 : SUCCESS in 14m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1f4ea42f4c94839ab7f80c374b76950 : SUCCESS in 6m 17s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/5f4642ff4e0a445f9c9006546c345fb9 : SUCCESS in 7m 52s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3937729398d64ee4beded1e5c445cad0 : SUCCESS in 9m 38s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/6549fb9d2b524f19980448b7a6f58b91 : SUCCESS in 2m 52s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/bbca81bc11274ae0a4dc1e3ac647cf64 : SUCCESS in 6m 41s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/87d2eee8fd22418988970368f9ce0152 : SUCCESS in 3m 56s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/3b0a9e10939c41b4b2449d651d3452d2 : SUCCESS in 3m 47s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/0667c0dad7454856b86af64f1eb3e978 : SUCCESS in 11m 32s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/b8547c839b7f4ffda284cb3ce16a7451 : SUCCESS in 24m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/74375fe7278d40ecbf4e8fc9ef1f5d75 : SUCCESS in 7m 10s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b0f37e7d6eba49878bbe2c0e73b1238e : SUCCESS in 13m 53s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/7b7b611859d64cc7afee55af9006b2cd : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/376d1f9ab67c4cce8df9b3aec39cd1d7 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/02eb8647fc064a3785df878e80ee1cf5 : SUCCESS in 26m 33s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/598e06b56bdd488a8a0f4fd97ebc8933 : SUCCESS in 4m 05s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/22872bf114bd4992a445a42d8eecc429 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9024df100c4948d98e60efe6f136e973 : SUCCESS in 3m 55s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/100f6ec9cfff4c29b6703c1d22bd70ea : SUCCESS in 17m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/48d1418a185f4fc4b425a49bbfc1d16f : SUCCESS in 4m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4525a1e3bca5409f8b07286b9bf0ecc3 : SUCCESS in 1h 11m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3300df888d9a4437a7db40bbc319672c : SUCCESS in 1h 41m 21s","accounts_in_message":[],"_revision_number":2},{"id":"de87f5d67099493208cdc90e8318923c24c721f6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-10 05:34:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a096fb6c2b199409f46768aa475dbcfda736be47","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-13 18:01:30.000000000","message":"Abandoned\n\nthis is not needed now as placement new version fixed the issue.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e021ea3bd1637d83fe3e17caf2b6946b17537f24","revisions":{"2caf755950ac80873991a06cd5a62ec393c10553":{"kind":"REWORK","_number":1,"created":"2026-07-08 18:55:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/996517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/996517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/996517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/996517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/996517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/996517/1"}}},"commit":{"parents":[{"commit":"7a04938e7e6f130a007e17750e880b6ec3eadd07","subject":"Add respx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a04938e7e6f130a007e17750e880b6ec3eadd07"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-08 18:54:08.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-08 18:55:52.000000000","tz":0},"subject":"Test nova cross functional job with master placement","message":"Test nova cross functional job with master placement\n\nModifying cross nova functional job to use placement from source\nbecause nova functional test depends on the placement functional\ntest fixture. As part of enforce_scope flag removal (imapcted by\noslo.policy new versio 6.0.0), the placement master(fixture modified\n to remove the enforce_scope flag) is needed to run nova functional\ntests.\n\nChange-Id: Ieb110e4456768d57ea89f2b412c47c000f1d53ff\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2caf755950ac80873991a06cd5a62ec393c10553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2caf755950ac80873991a06cd5a62ec393c10553"}]},"branch":"refs/heads/master"},"e021ea3bd1637d83fe3e17caf2b6946b17537f24":{"kind":"REWORK","_number":2,"created":"2026-07-09 18:15:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/996517/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/996517/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/996517/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/996517/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/996517/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/996517/2"}}},"commit":{"parents":[{"commit":"7a04938e7e6f130a007e17750e880b6ec3eadd07","subject":"Add respx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a04938e7e6f130a007e17750e880b6ec3eadd07"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-08 18:54:08.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-09 18:13:44.000000000","tz":0},"subject":"[to be reverted] Test nova cross functional job with master placement","message":"[to be reverted] Test nova cross functional job with master placement\n\nModifying cross nova functional job to use placement from source\nbecause nova functional test depends on the placement functional\ntest fixture. As part of enforce_scope flag removal (imapcted by\noslo.policy new versio 6.0.0), the placement master(fixture modified\n to remove the enforce_scope flag) is needed to run nova functional\ntests.\n\nThis is workaround to allow oslo.policy 6.0.0 u-c update and once\nplacement is relased then we can revert it.\n\nChange-Id: Ieb110e4456768d57ea89f2b412c47c000f1d53ff\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e021ea3bd1637d83fe3e17caf2b6946b17537f24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e021ea3bd1637d83fe3e17caf2b6946b17537f24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
