)]}'
{"id":"openstack%2Frequirements~996790","triplet_id":"openstack%2Frequirements~master~I20604ce2df05fa335abfe6101c6d0cf722b170e6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I20604ce2df05fa335abfe6101c6d0cf722b170e6","subject":"update constraint for oslo.upgradecheck to new release 2.8.0","status":"MERGED","created":"2026-07-10 13:04:49.000000000","updated":"2026-07-11 16:54:09.000000000","submitted":"2026-07-11 16:52:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996790-new-release","meta_rev_id":"57a13586ae0e970783c9d06bbe41356edb2fd397","_number":996790,"virtual_id_number":996790,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-11 16:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-07-11 14:55:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-07-11 14:55:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-10 14:24:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-11 14:55:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d9bf2c39fffc0739defe847b647c86c582d7beb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-10 13:04:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83765f765bd1e4755f60515f95738b5c95bf9b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 14:24:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a9bf953e5ee43a0814692c7e9243669\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c91323ce7fd46f28c027ba7bd5bfc24 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62eea5bd89c94098a1eaf2b938bb846e : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae7898456da84f0890d0f76161f7a134 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/daa2855f2ca5406095149c0589d38aae : SUCCESS in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e0009526ff474dd68bf6161b983c9fa8 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84903013e9e74a16a3c02a0b039b9056 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d71301ad696d4e9cb0610504ad602c8f : SUCCESS in 2m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5b1ab300dec942abb6603e788755a9b1 : SUCCESS in 5m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5d5cc9434fc543e082391024306752ed : SUCCESS in 6m 02s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/87a1c99065e64dadb65d225f21f5d623 : SUCCESS in 7m 50s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/870ebe0bc1af483282fc0e8f1c59c7c5 : SUCCESS in 7m 15s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/b79eeba8678a466b8d409eac5399e620 : SUCCESS in 5m 57s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/203bad3d94a04f56b1deec1af312a2e1 : SUCCESS in 4m 04s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/652d41ef4a2d4e398ae6ae8493e793ac : SUCCESS in 10m 16s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/e5d754065ad04b39aeb8a78b8fffe82e : SUCCESS in 4m 46s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/4fc181347eb9450b8a56c0c2a8d0bdeb : SUCCESS in 7m 28s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/dd628277ade945418ba66dcfc3e40933 : SUCCESS in 3m 54s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/17ec618c67ab412c94677dc0c60cd323 : SUCCESS in 14m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ad89a752f73480fa50ad2eadf3d0957 : SUCCESS in 5m 16s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/789746b73de64114b91c31a374a89084 : SUCCESS in 6m 42s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/c88e5ed8a8534af984106fce798d7c07 : SUCCESS in 10m 07s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/6e4ef65846ed413ba3d21875f5f25e86 : SUCCESS in 2m 26s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/eab75ab002c44d2488a9b96f92e2f35f : SUCCESS in 9m 46s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/fd1587b8b322497eadf08b4a96f54ced : SUCCESS in 6m 40s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/c7215d97af8a496395ac8b4ebbe0c57f : SUCCESS in 5m 13s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/11e57737134b416a8d897da3da29c633 : SUCCESS in 10m 20s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/ef9343eb5339461b9f8cbc8b3112658d : SUCCESS in 16m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d5f51474bcdc4d23a9e000114a42a5d9 : SUCCESS in 6m 19s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a9c98896cf2a46fcbd5b85ab32e6f66a : SUCCESS in 11m 54s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/8b62adc6926e4d89b847ea4fbb1a0a64 : SUCCESS in 2m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e81b181429d4d97bb43f6cfc92eb2e8 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b4c78c338824d0189ee7db9f6e7b9d3 : SUCCESS in 26m 12s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/49a824162f3d43e785ff2fb0d48ea1dc : SUCCESS in 5m 37s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/9cbc53a8c4a34dd487a0812b8f138ae7 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c4da9a0592a4e1a9340d8a51daf1b10 : SUCCESS in 6m 29s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/99f88ac4dd9a4d02a6a6c0d0763e7112 : SUCCESS in 13m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/35ba6df992884a3eacee88ea2f113af4 : SUCCESS in 3m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8aa3e33c8abf4ac1b16ea0400fdf17e3 : SUCCESS in 1h 05m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aed72dadff9f4e01ada18ec0c9df5740 : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e7f34d00e55eedad7d8637d0f83ec5c9686aefb2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-11 14:55:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1c1946525045a67dd1846173eedf55b542fed3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 14:56:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ff2169075d59a8a6634cc9735114a19cb5fd844","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 16:52:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4af73286d124beea97aa4dfbed0accc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ea13baa96b4e67b747cf08b0eca18f : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0bbddbe39f94637afee9fb4e68866e9 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca946bf5b21646b2ae443fd2f7a6af0f : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d3f168f531445eda7c3635b72297117 : SUCCESS in 2m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf86e25bed1740b68e58228f88dd88e6 : SUCCESS in 2m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac9f99d2de994858a65b01737f08c505 : SUCCESS in 2m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9f98ae44873846aaa0c27473807965fd : SUCCESS in 3m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ef3caff473164026b421d463a426d520 : SUCCESS in 3m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f94279ccf8154a1196c073b5a46f2c8e : SUCCESS in 5m 59s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/476e4b2bdd9749a4b1e50e3c7658826e : SUCCESS in 11m 02s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/874b826c57924b44acb057bfbe3d45ce : SUCCESS in 3m 09s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/db543dc8933848bfbb4c08d6aa8a987a : SUCCESS in 3m 24s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/63676111e77a4d7a8f4f6ddc6f391a33 : SUCCESS in 7m 28s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a029d6455ae343148e1a8a3d8ba3fdb4 : SUCCESS in 18m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b619cf103ed4e10b8ed12a2f3d85f77 : SUCCESS in 7m 32s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/3cfcde8598fb44559ef809c928923301 : SUCCESS in 7m 03s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/bb1ab37aacf447baab60cb1b8cd8787c : SUCCESS in 9m 18s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/bc272dbcf8fb4af4ad44be3c3a9fec4d : SUCCESS in 4m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/a88c1d985c2e465a8dae2fad052c1029 : SUCCESS in 8m 12s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/11a479f7739f46d1889d9416403d23a7 : SUCCESS in 19m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cbf1132f989d46cdac0daa334324e437 : SUCCESS in 8m 56s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/f0ccb25544174a3c948725e6ae744f9a : SUCCESS in 13m 22s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/01aabad91d03403b91c64f6ea196137d : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9cbd8e308c844390a3ab6697d4dfb4b5 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c98d5df12fb342478639d4cb20c73c39 : SUCCESS in 25m 59s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/28627b428a644d73931e2adcef06c7b6 : SUCCESS in 3m 25s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/78386c8ff19f4da7bc9360571d71e743 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/529ffc548f3c450aa6f480cf9d884f09 : SUCCESS in 6m 56s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/35d4e79c22144d76b4fc60c5aad8f2cd : SUCCESS in 18m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/abacc51138c04f9c8d4304ab1670e339 : SUCCESS in 3m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8162280693b405a862d39c0652ec866 : SUCCESS in 1h 04m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f11046d0157b4656b25272e79deb7009 : SUCCESS in 53m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5f8d8b07f9e65a30713624e2b1ad40883894cae9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 16:52:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"57a13586ae0e970783c9d06bbe41356edb2fd397","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 16:54:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f72900aee0e74526997e7aa3940380db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20fb252f3aa4f2f99b75ebcd27d1e7b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fe52904855160472adc1322f0e4e5e559f3e67c","revisions":{"6fe52904855160472adc1322f0e4e5e559f3e67c":{"kind":"REWORK","_number":1,"created":"2026-07-10 13:04:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/996790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/996790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/996790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/996790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/996790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/996790/1"}}},"commit":{"parents":[{"commit":"3f2c6d5a06d6c12c62c71697a48961ba13be1e24","subject":"Merge \"update constraint for pycadf to new release 4.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f2c6d5a06d6c12c62c71697a48961ba13be1e24"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 12:54:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 12:54:46.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 2.8.0","message":"update constraint for oslo.upgradecheck to new release 2.8.0\n\nmeta: version: 2.8.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ifad96e2fd8fcfdd4499dbc38f3ac81030867c8af\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I20604ce2df05fa335abfe6101c6d0cf722b170e6\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fe52904855160472adc1322f0e4e5e559f3e67c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fe52904855160472adc1322f0e4e5e559f3e67c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
