)]}'
{"id":"openstack%2Frequirements~996797","triplet_id":"openstack%2Frequirements~master~I307533392395d44bb6fa1e292d8597b6b3452044","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I307533392395d44bb6fa1e292d8597b6b3452044","subject":"update constraint for oslo.service to new release 4.7.0","status":"MERGED","created":"2026-07-10 13:20:54.000000000","updated":"2026-07-11 16:52:32.000000000","submitted":"2026-07-11 16:52:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996797-new-release","meta_rev_id":"379d4d942046fd24399ebadb8b9f8f4ae3c35323","_number":996797,"virtual_id_number":996797,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-11 16:52:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-07-11 14:52:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-07-11 14:52:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-10 15:16:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-11 14:52:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"89c4a798eb5b95596c2e6204a95f2e86da95f729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-10 13:20:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba42d2caa87aef43499d6021fa5cc84bd1a26317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 15:16:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b2a1975ef204ca4b7f80a73dcee5188\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49a867fba1d419f9e6f263977f78fbc : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30e0c93f666d4688b515c5282fa4690d : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9646a73bc5f14239bdbedbbfe471e2cf : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5080b8bbd3394f2cb28a4e96bc181e10 : SUCCESS in 2m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae38ec7ba6034a53bef327ecf871be50 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b83d2440431043be8addf1a983448458 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4dcfc98b98f4ec49b0f95d337de6dda : SUCCESS in 3m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/84301548c11641cb85748e07f4533108 : SUCCESS in 5m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/07869df57aa342a5b4eb4b80292c15e1 : SUCCESS in 5m 49s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7042c7dbd0e1460d865c518c7b854458 : SUCCESS in 5m 23s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/1b2e6e7a9ca04757bcc927a3c18102fa : SUCCESS in 4m 29s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/2c0cea91a6d147d2b67eda70bf401c7a : SUCCESS in 4m 56s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/0e6681d967ee4663a27948d195b82ec0 : SUCCESS in 5m 51s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/414d3d6be8404cc18c57fdc61102d71b : SUCCESS in 10m 52s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/3fdab6f23ae24128b743ecd6ee8ad3ec : SUCCESS in 4m 54s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/100d1fa63aba466ca713104e7d1c1177 : SUCCESS in 5m 29s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/201043477e4a49249553efdae0b9e11b : SUCCESS in 6m 49s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/d5b5fa50e5204b949eccb96e4ba37ee3 : SUCCESS in 17m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/588aeb26121544638cc1f966b2c9cba2 : SUCCESS in 4m 17s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/bea1a3f6b3b74e93bdfe8d43d83942e9 : SUCCESS in 8m 23s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0b515e5b79f94673973ae9770faf1ee3 : SUCCESS in 8m 40s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/3966f8728e504d9dbf13ad4de5d81447 : SUCCESS in 2m 47s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/2ae1f97f18b349ea88ad7ea18628a1df : SUCCESS in 8m 59s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/da09aa11503a4c8eb29da2942e727157 : SUCCESS in 2m 44s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/fa769c1c2ac94ef8b9887540802d76c0 : SUCCESS in 4m 57s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/42e64152710f4ca9bd8394e6babb3ef5 : SUCCESS in 11m 09s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/47dba6cff8574c7f9418f2d94d80ae81 : SUCCESS in 18m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c78fbc0da9964bba9b97d146be9a4b9d : SUCCESS in 9m 11s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/1e9395fab7614bfd8753811101b86184 : SUCCESS in 13m 31s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/2e49e27944184ef6bafda311a1a732d9 : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/783a02aa87b4452296a7c570c918370e : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f093379f955041f586d93d9bd29ce0d3 : SUCCESS in 25m 21s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/30d1591bdca74d1a8bb789fce9c1736b : SUCCESS in 5m 27s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f50ac9f1d8454936b8bb4f8eecf671b7 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f76c3ccf1ce348dd8c173fa0048e30f0 : SUCCESS in 5m 57s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d71f18a901c8463b8ee581623b339803 : SUCCESS in 7m 40s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b325c415daa949f3815ca366b850d52b : SUCCESS in 5m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/209c7a4ab4fe4fa4a2e8ddb55c5d8829 : SUCCESS in 37m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/086589ad575245298baa623f9dfb6bab : SUCCESS in 1h 46m 12s","accounts_in_message":[],"_revision_number":1},{"id":"329f46adcd7218c55db304e32cdb77f10daeaab6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-11 14:52:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b004679973416c15e73b856249eb42fcb833a4ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 14:53:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33042197a58f708920f34225b20284fad0617f74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 16:52:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c247d73a36d946029cde7e1f1087a8d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dee3c6b29594bae8ef8800104de0d59 : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b57249732f4616bf14da3bd28ce06e : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1fe7f7f7000443eba899200069f7cc3 : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b04aeab4ed68486a8177775585c3316e : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1bdbc760b33f475488c2e6a188b75967 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/29425be72b3146e9854beb95b409c109 : SUCCESS in 3m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fa1e9cd90d024c0a82f52f4187e4ebb3 : SUCCESS in 6m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8ccad0916fb2489b89fffecf77166a77 : SUCCESS in 10m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4146c28fbb934f24a01aaad438c8cb40 : SUCCESS in 3m 34s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/41ce4940c80641bab0410d27ff99cd87 : SUCCESS in 9m 58s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/adf9c34fc47c4a379232231a9bffe39a : SUCCESS in 4m 37s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/6641c7824f454eebaf9be6744584b16a : SUCCESS in 5m 39s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9e09f7412e504903961d7e7e37a4599f : SUCCESS in 6m 00s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/d472026896a24af5ad58d76d7d9f175f : SUCCESS in 17m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88e0f64e0083448580ebac03c052178f : SUCCESS in 4m 29s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/0ca02116c9b34eafb9e365db4a413ae7 : SUCCESS in 4m 09s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/8bb4231335d54b4097c3db4866530856 : SUCCESS in 11m 44s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/af589d982c22454eb907e4d2e47bd39f : SUCCESS in 1m 54s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/62530ed9d16e42278587102346d2acc6 : SUCCESS in 7m 58s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/fb9f6c3dabf048ee9e7955e126e36d9b : SUCCESS in 16m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a6923df9aed944659362708c5df31810 : SUCCESS in 6m 30s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/33f9d2f4ac4e4f21b8aa42f00925076c : SUCCESS in 9m 42s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/d6e9e6aadb684310a1d51d83705b5a84 : SUCCESS in 3m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea8d182f93ba47e3a838d4b3ea88512b : SUCCESS in 2m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d584212ae9974ad9936a01c4e18494ae : SUCCESS in 24m 39s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/1f3bfa26439440369fce365113a46421 : SUCCESS in 5m 32s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/4854674d362c47b99d6bd0911d2cd03c : SUCCESS in 7m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c34b7134d46445ae8c8e1fee532248e2 : SUCCESS in 3m 55s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d494328b86b3491ab2a3443b66267fe0 : SUCCESS in 13m 46s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b522bfada36041448cfa21e3d2766fbf : SUCCESS in 5m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfb2dd49b4334fc8b087db781d20fb71 : SUCCESS in 52m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6588b5b6ecb3452fbd5ebbc1ba327718 : SUCCESS in 1h 45m 51s","accounts_in_message":[],"_revision_number":1},{"id":"379d4d942046fd24399ebadb8b9f8f4ae3c35323","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 16:52:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a108768c8f656ace859bc8a2d2d93d2ca78bbfd","revisions":{"5a108768c8f656ace859bc8a2d2d93d2ca78bbfd":{"kind":"REWORK","_number":1,"created":"2026-07-10 13:20:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/996797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/996797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/996797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/996797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/996797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/996797/1"}}},"commit":{"parents":[{"commit":"3f2c6d5a06d6c12c62c71697a48961ba13be1e24","subject":"Merge \"update constraint for pycadf to new release 4.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f2c6d5a06d6c12c62c71697a48961ba13be1e24"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 13:10:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 13:10:53.000000000","tz":0},"subject":"update constraint for oslo.service to new release 4.7.0","message":"update constraint for oslo.service to new release 4.7.0\n\nmeta: version: 4.7.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I3d419748c57ea54915d997d954b2914f2bf558a3\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I307533392395d44bb6fa1e292d8597b6b3452044\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5a108768c8f656ace859bc8a2d2d93d2ca78bbfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5a108768c8f656ace859bc8a2d2d93d2ca78bbfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
