)]}'
{"id":"openstack%2Frequirements~996798","triplet_id":"openstack%2Frequirements~master~Ida2436d17cb827d7f32ef94669b5530540e43c7f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ida2436d17cb827d7f32ef94669b5530540e43c7f","subject":"update constraint for oslo.log to new release 8.3.0","status":"MERGED","created":"2026-07-10 13:22:48.000000000","updated":"2026-07-11 14:47:25.000000000","submitted":"2026-07-11 14:46:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996798-new-release","meta_rev_id":"acc4d00882d7a80e37eba0f0ad183e8eacf60097","_number":996798,"virtual_id_number":996798,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-11 14:46:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-07-11 12:53:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-07-11 12:53:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-10 14:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-11 12:53:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3a39e6c5ba307cb1503caab60dd7aa10fd81db96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-10 13:22:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c728483ec7bdbeebcea79df8f97c8a6bdc9fa46b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 14:21:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dd27634390f49d9bee651959edb70ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40d4ac9ea0d41678569a73ebc0e65ac : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/641ed24e270a4c11800fbbac4db7ad07 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/def7c98be77748f590651bd3751f412c : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44d6540d6734463fa638f34874021767 : SUCCESS in 2m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dbe3bcf87e5e463e9b6ecaa582e706c6 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8163237ceb54f93a204b590e8aa3dcd : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7406a6157fbb49e1b060a3126e86ec63 : SUCCESS in 2m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/845e935e35734ad184273d9a98083287 : SUCCESS in 6m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/74969376bc9649acb1b33f5df9ba7ccf : SUCCESS in 5m 35s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e36388b3a9ea46049140d131a647d57e : SUCCESS in 5m 52s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/c54dabb7b5f245b8b502300392b272af : SUCCESS in 4m 43s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/d12ca9f9827d47fb9cacdaf81257f61d : SUCCESS in 6m 03s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/e7038800e78543bbaf902859f087b107 : SUCCESS in 4m 48s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/aaf775a62ff14016a08b10324510fc93 : SUCCESS in 10m 30s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/2207eac8e01b4071b77a01702dce1018 : SUCCESS in 4m 06s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/e5182d94fa8b49d6b099f0e91ed1396c : SUCCESS in 4m 33s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/81c532037eb24dc692e334332d3d9890 : SUCCESS in 6m 23s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/e754413045fb42da8eb920b124174f97 : SUCCESS in 18m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9db7749c70c0487f86c7d835702554df : SUCCESS in 6m 51s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/68efc577263e419a897bafd5b20394f5 : SUCCESS in 7m 51s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/6cbef51633124ba483959531cdff4267 : SUCCESS in 11m 17s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/70e0e60bf15a45f3a7703a1c38f48e2f : SUCCESS in 2m 37s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/995d3f51781b47a1a3874e05a0d917ca : SUCCESS in 8m 01s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/9a7f542f4c6a4832948b3ca095882739 : SUCCESS in 3m 51s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/aa65b38de4ea4099829ca7bbad098b59 : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/d5e6433937c94feeab8edd7d46ac9501 : SUCCESS in 8m 54s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/d737dbb0db3a4c2ebe64f4e01ccad203 : SUCCESS in 25m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4ad1e9770d8a41cba0ee3f2c4f636534 : SUCCESS in 8m 01s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/252e6802054d40248ceb33632e78e337 : SUCCESS in 13m 38s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/ed35738f43ec49b29e7e9847c9126ea0 : SUCCESS in 4m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d98fad69665745e08c1b570aa48d7581 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d7cbb8d1721441f3870cc43a0686e724 : SUCCESS in 23m 47s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/835d359f9cbe4b318d47f2372787ac99 : SUCCESS in 6m 33s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e9d454c3b19541d8a3ec95b0e2c6c11c : SUCCESS in 3m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e0b132f6ce5c424e8c54acf6b922fe09 : SUCCESS in 8m 10s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/72ad65dc2ab9462995bec6b1b1754a7a : SUCCESS in 17m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/120eccc935c74ff797c738acfea55c47 : SUCCESS in 6m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ec33beb5e664da89c3b2e86c44d1b75 : SUCCESS in 34m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/256fdb38b1ab4817a0bf4cfc66d9f59d : SUCCESS in 55m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6fd7b4a9cf0d2f6eec9caec9c01c2d6a8eb5bc16","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-11 12:53:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a044137436e629e59c0ef31ad417fc971b4bb17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:54:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"823367e62d04ae375364ac1d1f5ef22361230f6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 14:46:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a32c0ba03b2482db16c72c92ac766d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68a82e08b0ca4e19b0581cbe6db12fee : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec5671b8506e4440842772797d903948 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f63f135c3c049bfa467b73cc119b0ca : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c349384e29604a6d94b769b51c6ec9d3 : SUCCESS in 2m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44b4ee946c0343e1880b62548efe5964 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a21c4b156bbe4cb188bb288bac728f4b : SUCCESS in 2m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9fb4ede4acf3454ea5283abde6a75178 : SUCCESS in 5m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ffa7df4fd76a4cf2bcdbf23989a71907 : SUCCESS in 5m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4e9720fe43e549d98cef4924108cdded : SUCCESS in 5m 36s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/239fe37897e44be7aabeca11ff4a532d : SUCCESS in 10m 31s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/02f65b0ef2144109920f7389cc4c4ce0 : SUCCESS in 5m 43s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/b43a29b719174e7e8aaf37294a068de7 : SUCCESS in 5m 22s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e28a143665574e83a6fdb15a7cd5b236 : SUCCESS in 4m 04s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/ed850e1d984c4041ac842f8b39ebd721 : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ca0a7555c65426fa368c6931b3c0139 : SUCCESS in 7m 00s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/871d6cbd055f4d9e97b05717848c6508 : SUCCESS in 6m 59s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/2efcfc491574490a946d8f446c3466b9 : SUCCESS in 8m 26s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/b537e01993d7419794f8500ddbb47449 : SUCCESS in 2m 03s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/8c2758c50e7143f4885ee9aff83ece4e : SUCCESS in 5m 27s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/cceeb3da42394b42974573e0ce4f5848 : SUCCESS in 26m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a94a44b3ba534cf9af3a2ae463165ba2 : SUCCESS in 3m 53s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0431a85eeab54d8b8260d8183cd5d83f : SUCCESS in 15m 23s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/40820d8b7dd5441b8addb6c2b47f9073 : SUCCESS in 2m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8459fa2d318427aacac4ca6547dc263 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/916d5edc6ae04f96b4550b8a9dc9297c : SUCCESS in 26m 18s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/cef575f988ee4a49bc45c5523d6ec824 : SUCCESS in 4m 58s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/654145bd5c474dd794a6abe6576366df : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7781a8ed5aa649199d280367070ae517 : SUCCESS in 6m 28s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/29c2643badd34d2ca4cb6ae20b5f3349 : SUCCESS in 18m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4a9d5b0963bf4dee811bddf99216e67c : SUCCESS in 4m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3e5142e0e614468911f5730b39e4648 : SUCCESS in 33m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7150c6f1c6d2492781f5a0e57c3419bd : SUCCESS in 1h 47m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9812ad1349789a7f861c8ecec545ea3254e68734","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 14:46:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"acc4d00882d7a80e37eba0f0ad183e8eacf60097","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 14:47:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fc2b79591404262aa96c6894810aa91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a10fd17494462d8c4db3e69646c38b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7693ddf403afbd4dc4ed13dc70977af41842b599","revisions":{"7693ddf403afbd4dc4ed13dc70977af41842b599":{"kind":"REWORK","_number":1,"created":"2026-07-10 13:22:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/996798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/996798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/996798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/996798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/996798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/996798/1"}}},"commit":{"parents":[{"commit":"3f2c6d5a06d6c12c62c71697a48961ba13be1e24","subject":"Merge \"update constraint for pycadf to new release 4.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f2c6d5a06d6c12c62c71697a48961ba13be1e24"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 13:12:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 13:12:46.000000000","tz":0},"subject":"update constraint for oslo.log to new release 8.3.0","message":"update constraint for oslo.log to new release 8.3.0\n\nmeta: version: 8.3.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I8b406a604fdf510fef591a6eb6e9d3cf182ac50a\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ida2436d17cb827d7f32ef94669b5530540e43c7f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7693ddf403afbd4dc4ed13dc70977af41842b599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7693ddf403afbd4dc4ed13dc70977af41842b599"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
