)]}'
{"id":"openstack%2Frequirements~996804","triplet_id":"openstack%2Frequirements~master~I4c230eece59be210e7771cac084c1835529a51d6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4c230eece59be210e7771cac084c1835529a51d6","subject":"update constraint for oslo.limit to new release 2.12.0","status":"MERGED","created":"2026-07-10 13:52:04.000000000","updated":"2026-07-11 16:52:37.000000000","submitted":"2026-07-11 16:52:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996804-new-release","meta_rev_id":"a3a7210e89baacc1444336558cd6df95050ceeed","_number":996804,"virtual_id_number":996804,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-11 16:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-07-11 14:53:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-07-11 14:53:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-10 15:44:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-11 14:53:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1c272f6801d08cdf953cfc01f509d668f749bd6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-10 13:52:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6ab4b91653b9630a14f7b79a718724014bf5eb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 15:44:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afba7d87c8ba433d94d7f2ce32f0d739\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59780c165b4e487c94e251ef7a185df5 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfa24490ca60451a9331c8ec9dd3bff9 : SUCCESS in 2m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e73dcff71d524beeab260dbbd64313ea : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3d93703d1ba44f3ae26e6694b020f5a : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4ba134b2b6d744058f135d492992c03e : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c24714fa967d4ed7bf038787fac6e4a1 : SUCCESS in 2m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ac95eaf195d4c2682119d0011d5bbbd : SUCCESS in 2m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/91d11094b0ea465989c25162299ebf8a : SUCCESS in 3m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b339be05a3be47edb479779792048dc2 : SUCCESS in 3m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6bbc0275fe7042f19ce35a51c73df9b3 : SUCCESS in 5m 39s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/328cc93ec8c544bca9f721397947dcd6 : SUCCESS in 3m 25s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/49f51679c7d8435b82a8cf8a15aaf971 : SUCCESS in 4m 49s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8543e511c54842318739d8fa7e6f530d : SUCCESS in 4m 26s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/6ef0e4993e634fdb8a719d700984d8dc : SUCCESS in 6m 42s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/e68b95c4803e4e72af0609524db3d298 : SUCCESS in 3m 27s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/9fd34d94b7954d37810a645b04692ef5 : SUCCESS in 6m 25s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e11e9c02a88e4b638c63971fbbf90645 : SUCCESS in 7m 12s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/c29729fe9ae948dfb9295560d60cacbf : SUCCESS in 8m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7bdad3b1678a47c185cb34c4b9459558 : SUCCESS in 3m 52s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/9c9d03c056a94ad694cd381f36e7424c : SUCCESS in 7m 49s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/c5c335c4e2624008a3782a995ad47e3f : SUCCESS in 9m 33s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/ad5b3ce83e3d402b86033068ce63c3a8 : SUCCESS in 2m 16s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/ab0f87d0e2f241cb8cb945625c7fd465 : SUCCESS in 8m 02s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/b1103960892342318493b7f56dbbd06e : SUCCESS in 5m 37s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/9d32c48df9f9460eb0dae636fb7b4377 : SUCCESS in 5m 19s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/2dd2ddd5352d416ea9f44cb05a12d83e : SUCCESS in 8m 52s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/fe525a78d4a9414b8224486ea9e49d5f : SUCCESS in 18m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cc1d68d39b80463f9394296e1fa43cc8 : SUCCESS in 6m 35s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/c4ffba3140b24822acc0cdffd692b25d : SUCCESS in 8m 59s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1c7e623e987340bf9d483ccd1a502b8b : SUCCESS in 3m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b1d024411f6f49a8b405d3bfd3130b9e : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/023f90cdf29d4d65acf7290115749dde : SUCCESS in 24m 32s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/7b2e43b93e5c4bf2b40f225878935d57 : SUCCESS in 7m 57s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f9e67f241ba14ee888fa00079ed19d74 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50587be42cda4d48a06dd715e75d13bd : SUCCESS in 6m 30s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/3dc498c49e7c4543b8f6eaaea5e7cd51 : SUCCESS in 18m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fe04c06bfa2b4efd857688bf3169908e : SUCCESS in 4m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38b55877fd3744f2bba4d132958228ea : SUCCESS in 38m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe9fefa66e7b43d48300c8637dd1c71e : SUCCESS in 1h 47m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1d367ab6532aff2fe59441a35ea9c092add92b0a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-11 14:53:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fcfabefc0c4d0f4dcb2d80a07a1d5e7d60b6d4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 14:54:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36ef15a7f17be85a08e15381b53804d679884126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 16:52:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c680c729c65148d49627353bc4343a5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79688aacf2a34026b112aa0af63d75ae : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff097c87b88244b889ec175fec833054 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b66e73bbdabd48159b16767437d12653 : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18e61353e0e3498f9e180d833bcab6ef : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d73a3611ff8047999531e846ab39c9bc : SUCCESS in 2m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8f3ef56db60448ab84f363c712078bb : SUCCESS in 4m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/559bbbc9260544f4b3939a7ec1c4c81d : SUCCESS in 9m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/55827468ed914de5aacd50b881129389 : SUCCESS in 7m 26s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b8549ab0a5174158a998e9d4a1f40a20 : SUCCESS in 5m 51s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/e1db1248b738488493a56ba4e2e91249 : SUCCESS in 12m 39s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/eb63d646c96449fa992f450daebb8725 : SUCCESS in 3m 33s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/23895c70312c4233a9a05135ffb43e0c : SUCCESS in 5m 32s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e7b8dd2e62ff49259f60034cd45a51e5 : SUCCESS in 6m 25s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/d33d4ca168ff45bdb56643aed42e4e78 : SUCCESS in 18m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b2a7ab28a13443f9abf1e5ab9ed49009 : SUCCESS in 6m 42s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/58b9eda5d1ac4906a2967b26a322481d : SUCCESS in 7m 24s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/43deb3d1c70e4c8689b995b1c9b25bc2 : SUCCESS in 11m 20s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/eda5a68c80a643c0b095bd9d4a43b041 : SUCCESS in 2m 15s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/e7baa332b82146838df2115f8e199357 : SUCCESS in 8m 47s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c7c6886172014ccea6d58f023d2475ff : SUCCESS in 21m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2e03aa0e235742cda4564a621b170ac0 : SUCCESS in 6m 21s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/3d6b4bacfc9845d8997a2913c953273d : SUCCESS in 13m 37s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1eca03c09e20487190417575defc5a7c : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/285af14662544a318815a7fa95677114 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3f8a37e1dff4af98d6d4a2d1811104d : SUCCESS in 26m 10s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/66e7d6556fe84118b5c82fa166c41a14 : SUCCESS in 6m 19s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/bb0106295f1e4365a601d6855a8133cf : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/993cc732200f433f84a20f5833efe6be : SUCCESS in 7m 26s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/3d783d55b7ab48e488ad592a6736e2d9 : SUCCESS in 8m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0988eca8e088434897c6d9a865597bb3 : SUCCESS in 5m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3d0541e4cb242b3a26730560d997f9a : SUCCESS in 1h 04m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9ed9b60e31343c8936fd6b0751683f8 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a3a7210e89baacc1444336558cd6df95050ceeed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 16:52:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e83ce5da1d3f9b36f43e7aa70ed35d169d0722cd","revisions":{"e83ce5da1d3f9b36f43e7aa70ed35d169d0722cd":{"kind":"REWORK","_number":1,"created":"2026-07-10 13:52:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/996804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/996804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/996804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/996804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/996804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/996804/1"}}},"commit":{"parents":[{"commit":"3f2c6d5a06d6c12c62c71697a48961ba13be1e24","subject":"Merge \"update constraint for pycadf to new release 4.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f2c6d5a06d6c12c62c71697a48961ba13be1e24"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 13:42:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 13:42:02.000000000","tz":0},"subject":"update constraint for oslo.limit to new release 2.12.0","message":"update constraint for oslo.limit to new release 2.12.0\n\nmeta: version: 2.12.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia637ed16f4e48d58d01b2333a5594969a1d13c33\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I4c230eece59be210e7771cac084c1835529a51d6\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e83ce5da1d3f9b36f43e7aa70ed35d169d0722cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e83ce5da1d3f9b36f43e7aa70ed35d169d0722cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
