)]}'
{"id":"openstack%2Frequirements~996811","triplet_id":"openstack%2Frequirements~master~I69cb2435361ae7726f2c85734292a4e16b633c4e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I69cb2435361ae7726f2c85734292a4e16b633c4e","subject":"update constraint for oslo.messaging to new release 18.2.0","status":"MERGED","created":"2026-07-10 14:04:01.000000000","updated":"2026-07-11 17:05:03.000000000","submitted":"2026-07-11 17:02:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996811-new-release","meta_rev_id":"912846dd44329c0fb2951ec2db14af5ced8a5186","_number":996811,"virtual_id_number":996811,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-11 17:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-07-11 14:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-07-11 14:57:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-10 15:48:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-11 14:57:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3069e8bf424618413aac7f305e83cf2781452a79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-10 14:04:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb467c4494e98c12a1e105ab4d98b879ac966e84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 15:48:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31a8dd25aa8a46feb0e429f2e7c7f906\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4e514c69e374fb1b504c7e9d9f87739 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0153cef00f634ee7a871f97bb6acec33 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5bace16bdab6453ca62d7ee1e4d0762f : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26faf4ab061e43298febc4893c2bb141 : SUCCESS in 2m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/09c94fa135ec44ef80c5307ac87f92de : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba522562c39b4339bc100ac2af1c8dfc : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a4d8d7dfe3248c9a19cc4eed742a69d : SUCCESS in 3m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/306f7e7288bf4182807de5c7d0fce5cb : SUCCESS in 3m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/74625338efc54ca3a3fd5f16f1e55bda : SUCCESS in 9m 34s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1e7278c5a37543f49e80b5f442629bd0 : SUCCESS in 8m 36s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/b92bffac4a3e4e9481083055486fceb2 : SUCCESS in 3m 29s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/210c09c964894cc1984d18f1497fa044 : SUCCESS in 6m 47s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/805f6880696349f4bf180c43b4164723 : SUCCESS in 6m 56s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/0f61d46285544f51b11ac321f464b329 : SUCCESS in 6m 39s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/30dfe225645b4682b8dc7e132d0aa6dc : SUCCESS in 3m 14s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/bd885d28608c4f758e6d339c8422bb32 : SUCCESS in 3m 20s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/c527856c348d4b2983e7d8413dc5486c : SUCCESS in 7m 17s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/363c3ae0ba524e44ace7d30c4d238135 : SUCCESS in 18m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a4c87147dfb41c8ae52bd6c54c58083 : SUCCESS in 6m 50s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/5cdacc70fa554b368d1e96eef893f144 : SUCCESS in 5m 01s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/d9f072e368bc4489855c23e97ad6512f : SUCCESS in 8m 04s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7d7d08d13d8b43df9dead5cacdc47dab : SUCCESS in 2m 43s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/9551496e475648a4a373c15b1fe7da69 : SUCCESS in 4m 43s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/e184dde5f1f543158c37c158c07e04f5 : SUCCESS in 4m 19s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/581e59b26bee4996ace2de081bb61613 : SUCCESS in 3m 12s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/885420afdc464221bc57c89d759cd21c : SUCCESS in 10m 48s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/b20e157d92d94d5eb6471ea2afbae1b7 : SUCCESS in 25m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e1a4fcd6ae044806872a1af524642d7a : SUCCESS in 8m 20s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/90c5e46c62174a5c92088fcd88afa7e4 : SUCCESS in 11m 18s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/abeb1084d14447bba4cf2a2b46dd2d77 : SUCCESS in 3m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6f6a701529ed42c3a1b7b939a4de0d90 : SUCCESS in 2m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8a2efd6594334f1d90737087c18f26d3 : SUCCESS in 21m 50s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/6810b309b8234b298a86dd2703af24cc : SUCCESS in 5m 09s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/6a2a5ce20f3b41a2b7f00578fe6fe6fb : SUCCESS in 3m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b2a3c91a5c1340658c14b8b95711543d : SUCCESS in 5m 12s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/4e1ec020f6ef4cb587e798fc51f1634f : SUCCESS in 18m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c645e82d861b4da09157823e59d3f152 : SUCCESS in 5m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ae405c2a56a4008ae3264a6a3fa1bed : SUCCESS in 1h 07m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a97cf3ffe81e4177b91b8df4d38fbdc7 : SUCCESS in 1h 37m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ed3bff4b6d18d92a398d2cce54869dae6059e927","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-11 14:57:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b6fcfab7721539518dbe46343a9c4585c6a2d9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 14:58:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80b2391ea30cb89d2478d2613a2cf071e1e9e96a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 17:02:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49d2bcef1eb949fba446d82412728dc9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95f84392e1a94ce58feb9b4de75d3aa3 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e194fac1d9d445e4869a2b0fb01bf55b : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/feaf4d154d4f436a9b811da3c68e86c1 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02ab616db63e41f78ebfe92237219fea : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/391ee9ffbb754846b5c20c38bd2c2a03 : SUCCESS in 2m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/03f35b9e88554a9183b7e02470f74b50 : SUCCESS in 3m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9c2029691a5d42f0ae3fde3a7fe81f83 : SUCCESS in 5m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4172c324c7b64a5b9a9053f259d56334 : SUCCESS in 6m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0e4ee66d47f644cd9e7310bc818e26ff : SUCCESS in 5m 43s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/20831c91f0f5407392c50c149014ee32 : SUCCESS in 10m 53s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/e510b262e2e34b47b9a14ac0f5fe32a9 : SUCCESS in 5m 48s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/7b486efe13d24ae7b56f66bbb25d0a87 : SUCCESS in 5m 24s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/19176b90921241fc93787235908c2d8f : SUCCESS in 6m 37s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/203431f5099e41508f3cee7b46e2494d : SUCCESS in 14m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0b8021b3698246dc9a75176a8951907e : SUCCESS in 6m 32s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/2eed404f09094419b65b7a7d20b37cc5 : SUCCESS in 8m 20s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/42242cca5b8c42aabbc1f286e5412d15 : SUCCESS in 8m 28s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/18377117fd4c47e79cbdc5ce8db5ff8d : SUCCESS in 3m 17s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/c15016dc727a446093a1ebb9befbda7e : SUCCESS in 8m 07s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/1a9625c7ca9b4596b72cdb478b7931ee : SUCCESS in 21m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/001cf5726b1a44cfba0234cb13245312 : SUCCESS in 7m 19s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/4d61e7b397194ae88bb4ea1237ad2e3c : SUCCESS in 14m 54s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/fbcffa13f4df4e69aa527241cee8b62d : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/98b8bc1c5cb44ee5b8e55ce9c4541ffc : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a586dc9a563f43bb91ed241c07ef2189 : SUCCESS in 24m 15s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/faed537e10774dbc9e4be075c379f940 : SUCCESS in 5m 22s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5e0ddf49418544999be9e7c772a38d57 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/819d6faaf427465e8142c545314c8ec1 : SUCCESS in 7m 12s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/bcb2fff837c24719ab6385dd6023d34e : SUCCESS in 20m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8ddc21ee77df4ff8beb667674629042d : SUCCESS in 5m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd0b7677b3d243729599a5ded21edcbf : SUCCESS in 30m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82b3f855066644b488a3d98e0bceb719 : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e022281bbaa3d79f490aff679f4580c1ce374635","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 17:02:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"912846dd44329c0fb2951ec2db14af5ced8a5186","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 17:05:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfcfed99bfd2458fba8906ab9b26883d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d445a68077546e2826d5915233adb06 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e3ae7f4950363c9ba3c747e2486c8262de7b830","revisions":{"2e3ae7f4950363c9ba3c747e2486c8262de7b830":{"kind":"REWORK","_number":1,"created":"2026-07-10 14:04:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/996811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/996811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/996811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/996811/1"}}},"commit":{"parents":[{"commit":"3f2c6d5a06d6c12c62c71697a48961ba13be1e24","subject":"Merge \"update constraint for pycadf to new release 4.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f2c6d5a06d6c12c62c71697a48961ba13be1e24"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 13:54:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 13:54:00.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 18.2.0","message":"update constraint for oslo.messaging to new release 18.2.0\n\nmeta: version: 18.2.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I211922b9938490a7797ed0feaef0ce2139319789\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I69cb2435361ae7726f2c85734292a4e16b633c4e\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e3ae7f4950363c9ba3c747e2486c8262de7b830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e3ae7f4950363c9ba3c747e2486c8262de7b830"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
