)]}'
{"id":"openstack%2Frequirements~997349","triplet_id":"openstack%2Frequirements~master~Ie787bb8fe80be941e7ce2b104d1795b6ee0c9f55","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-07-15 17:27:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie787bb8fe80be941e7ce2b104d1795b6ee0c9f55","subject":"Bump python-openstackclient from 10.2.0 to 10.2.1","status":"ABANDONED","created":"2026-07-15 09:30:21.000000000","updated":"2026-07-15 17:27:17.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b4e7ae2a2f082c3421d1d43ff3d898f2e273ee44","_number":997349,"virtual_id_number":997349,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-15 10:36:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2026-07-15 10:07:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-15 10:07:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-07-15 10:36:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4282b42d40c2e1d8715cccb43828f1466148f908","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-15 09:30:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20bbdec616a1c205f57160c023267033bb68da09","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-15 10:07:06.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf6244e01ae8233ee62450fc4376d078ea110e89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 10:36:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cda280495ae4c74917a48276ef0ba22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/291773622c0d46d7a01b10a31d45992d : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b662566dd0b64fd8ae9c36c19c26650c : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ddbf242b63a47dc9fe28fed818c6a91 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a69d666cc5654a329a5016f66f748e78 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d2744dd2f96043459f6c69209555d79f : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5c9290b2fc647c48adf8c27514e829f : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8da137eb36534d6e977e59e2e7773a0d : SUCCESS in 2m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9380da1970324c4dbdaa450b3abc6109 : SUCCESS in 5m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a28c4d7ee8e040969c89816f997d9da6 : SUCCESS in 6m 34s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1fefd0f1d6644a9cafed030fd3879fd5 : SUCCESS in 5m 28s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/6fbfd369a0034f68a17763974e6e7222 : SUCCESS in 4m 17s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/0393ec1eba014a719f1ecb66cae59694 : SUCCESS in 4m 31s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/f7b7b7db58d043d2921b23cbc8d1f489 : SUCCESS in 5m 23s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d0f9220758fd432096e208192604e56a : SUCCESS in 9m 55s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/313bb19bc988415d8f221cb71d49876e : SUCCESS in 5m 00s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/1c4f80e1b26e4c9aa8c5c547bd6e75bb : SUCCESS in 4m 44s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9314d61f7ac1492b87f21aa16e1ef2f4 : SUCCESS in 6m 10s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/75710a58ebf44320b4ddfe3ca63b2c7b : SUCCESS in 18m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d947d0eed2014a329c98b97c066d0156 : SUCCESS in 6m 52s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/3df7e19956164861b93ac52c9de8b057 : SUCCESS in 7m 23s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/76aa68a25efb4777b8f086d63e3a1b7b : SUCCESS in 13m 36s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5ab78755bf904721bfd9420b7018aa9b : SUCCESS in 2m 27s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/c0187c38b9cd43f4a26e7772fd0eee1d : SUCCESS in 6m 34s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/0899876d35254f8ab384496fc47b76cc : SUCCESS in 5m 01s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/b63189474fba4a7aa3afb8b4a86d5bca : SUCCESS in 6m 19s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/5d9ef4f1dee24a2685318421893892cb : SUCCESS in 9m 56s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/020570186265431d8a3b28643c631a6e : SUCCESS in 30m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/476973460d694e12b68e49b910f8f43e : SUCCESS in 8m 01s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a9f115b799284f29b8290eae24a9b66b : SUCCESS in 13m 06s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/588c57071043434cade48a49e4ea9a04 : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba7a74224e6c4c9da0e7402593f5f8d2 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/700badc89e5241229d3abb1cba89303a : SUCCESS in 25m 25s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4e406d263be44580bfc3ca4c12016dd5 : SUCCESS in 5m 25s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/befed69d7f534a9c9015f2b4fb867acb : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f7d745002b64698b9cd1527ceae821a : SUCCESS in 6m 50s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a179c658fee54ff1bf830bd3f1f11526 : SUCCESS in 15m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/637029c7be304bbf9eb21da0624ae431 : SUCCESS in 3m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c572f8b64a140e2880bb390e7a1a87f : SUCCESS in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/780f2a9cba19496bac5bb9b1d962026e : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6c4c6f36da57418cdfc923786e177a4140135004","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-15 17:27:02.000000000","message":"Abandoned\n\nDuplicate","accounts_in_message":[],"_revision_number":1},{"id":"b4e7ae2a2f082c3421d1d43ff3d898f2e273ee44","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-15 17:27:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b3808898ce96872aa519ba31cc9151595e97799","revisions":{"6b3808898ce96872aa519ba31cc9151595e97799":{"kind":"REWORK","_number":1,"created":"2026-07-15 09:30:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/997349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/997349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/997349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/997349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/997349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/997349/1"}}},"commit":{"parents":[{"commit":"3eb6830a0bbb7e4f105e32ad19e7fa5373f1940a","subject":"Merge \"update constraint for tooz to new release 9.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3eb6830a0bbb7e4f105e32ad19e7fa5373f1940a"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-15 09:27:21.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-15 09:30:15.000000000","tz":60},"subject":"Bump python-openstackclient from 10.2.0 to 10.2.1","message":"Bump python-openstackclient from 10.2.0 to 10.2.1\n\nThere is a regression in 10.2.0 where the client will try and probe the\nblock-storage point even when volume related options are not used[1].\nThis causes a breakage on clouds with cinder disabled.\n\n[1] https://bugs.launchpad.net/python-openstackclient/+bug/2160039\n\nChange-Id: Ie787bb8fe80be941e7ce2b104d1795b6ee0c9f55\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b3808898ce96872aa519ba31cc9151595e97799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b3808898ce96872aa519ba31cc9151595e97799"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
