)]}'
{"id":"openstack%2Frequirements~997772","triplet_id":"openstack%2Frequirements~master~I0530c6e21f69ab1bfa223e743ef030b195c0a1f7","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-25 07:30:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0530c6e21f69ab1bfa223e743ef030b195c0a1f7","subject":"Remove python-vitrageclient","status":"MERGED","created":"2026-07-17 16:59:55.000000000","updated":"2026-09-25 07:31:27.000000000","submitted":"2026-09-25 07:30:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997772","meta_rev_id":"12d93a7d153a0ae4a8aa33d2809e89c15be4a7c4","_number":997772,"virtual_id_number":997772,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 07:30:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-25 06:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-09-25 06:37:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-17 18:32:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 06:37:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"6a0dcd9f701b8eba6aeb71ac4ebb3f340dee82f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-17 16:59:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f7884d9fcc32f481fef7674824fc3968f379fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 18:32:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc7d1c4aad0149db86dd3870380ecfe9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0e1634337284c9ab86af343aa50859d : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdd293710624b0bb26fa32058c59ff2 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31ec5b349f4d4bfa8a0bc7655d8d7cac : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1717d1c2c3574f33a00b2e63efb263d0 : SUCCESS in 5m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a7b1277a8e88441bbc332b8fdef1c572 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94b5fe6a07344d4080be3969158118cb : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4ec80825790471bb0114be5514191ec : SUCCESS in 3m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/35799927cb104788b921059a8ed4fca9 : SUCCESS in 9m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/77988abf8755479e958df25675a9b7e7 : SUCCESS in 5m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b3616067887d4571b89252021335b523 : SUCCESS in 8m 08s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/72ce2a83bb1049b5b30b161076bc60c6 : SUCCESS in 6m 22s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/c5843276af104c91a94245bc67ab44c8 : SUCCESS in 7m 05s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/975021a45df2457698ef8c295fe54a25 : SUCCESS in 4m 56s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/5191c282b0ac478bb6f0fc1a60f9844d : SUCCESS in 9m 26s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/7558d3e851714b0d86082524ecc652bc : SUCCESS in 4m 47s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/1fdf493cf8c1498d946c6af7f85342b1 : SUCCESS in 4m 53s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/1aaadffeea174035bd3c8b0c2c04cc82 : SUCCESS in 6m 04s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/ddf5b394df7c4a34a3db1d115374bb11 : SUCCESS in 18m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61736b6e13fd4f62968ff50fc47a90fa : SUCCESS in 6m 01s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/20f7d879e8e04db9b78fbd44c03806c8 : SUCCESS in 8m 47s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3bcc91af701145aa9546bbb1460b3ce3 : SUCCESS in 10m 44s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/05dc3cdb606b4ce5b7868337bd84310f : SUCCESS in 2m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/3db9d010581d4334ac2060bf9dc356c8 : SUCCESS in 9m 44s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/4e29441d9824429695df9f98b3bbb9c1 : SUCCESS in 4m 04s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/42bf404ebad249c9aad0ec214580eef9 : SUCCESS in 6m 21s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/e5b2cfb4423543519d160c363c8c652b : SUCCESS in 10m 09s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/11d0ea33c49f451f9566cc26256ff690 : SUCCESS in 27m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f957ea0f95aa4faf9b97f511d1b25116 : SUCCESS in 5m 55s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/028cd9e9ad8142b484c753cddc40c18b : SUCCESS in 12m 32s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/4c75c66e8374423fba884fb9eb4fcad1 : SUCCESS in 4m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/647e910bb2924b4fa3ad0e46d28bf2cd : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6a8c4266977423eada6fb2f016b1c65 : SUCCESS in 32m 28s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/89ebc86d98ee429a93446a6a4028cdac : SUCCESS in 5m 20s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/b9c27cce27b047c38be42c6e16735baa : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f91f46ae2ee4d608b5cb0530fa98d8e : SUCCESS in 6m 08s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a4b3437dc86341c9946db0186fcc594b : SUCCESS in 19m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cb7b7f80b088429493f1a5cb7e625309 : SUCCESS in 5m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3175b462ed5e47b0aa475adc287214a9 : SUCCESS in 1h 08m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3ba76714a28447caddf86a132580d13 : SUCCESS in 1h 26m 50s","accounts_in_message":[],"_revision_number":1},{"id":"315bac91cd522f459c6f5c3294808c3fda2d330a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-31 18:37:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4f74f1a2eda7b4cb0babdbd3a50cf63a09172b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 19:27:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd79e78dfd44c35a33fd1a8ca9df4c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/506261c439884746bc13b35e041588b9 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0306a5e6dbd42258d1e279279c4ee4f : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/101cba5584094bcdadd3bcf9ae11b15e : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec97673c7d1c490996f04f937923ce40 : SUCCESS in 4m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ec539f91ab7843f5a6ac1c2309478330 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4bd245c768ad43cf918c0105e56238df : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8299cb5ade5141eda62a0856d475fd2c : SUCCESS in 3m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f6cf7154259b497295684445ce5e82ed : SUCCESS in 4m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e5c05aa3a11949108397b24c98e5685e : SUCCESS in 6m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c9d6554bdcb2489586e931e5c9fd8951 : SUCCESS in 9m 50s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/22295192d68d421196d06220ab5faeb1 : SUCCESS in 8m 11s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/9d472340094d486a81903b40998ac205 : SUCCESS in 3m 45s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/5996388ace8646e48a689c3ddf1e6496 : SUCCESS in 4m 38s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/9004a121fee5401580101543cbdd4867 : SUCCESS in 5m 29s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/dee9ac81de2e43e3a74a85a0db99a9d3 : SUCCESS in 11m 53s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/3129279d99374c6a884a5489224a42a4 : SUCCESS in 5m 43s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/eef3cf17758440bfa9bc751e467f4d99 : SUCCESS in 5m 32s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/c31d296dd0d54790b007655e871c92d5 : SUCCESS in 6m 39s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/d8f5277d19a84852820d50ca64a5e239 : SUCCESS in 23m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e6bd1d3fb5744bab86f44875210c538b : SUCCESS in 7m 10s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/afe1b375f3b44e008a9166a06625800a : SUCCESS in 5m 45s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/1c2cfec5b253478cbb298847164d7e78 : SUCCESS in 10m 37s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a77303046aaf4eda96c8d9b5c0c8367f : SUCCESS in 2m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/acf9e6221e9342afb71cca8fd4a3bd93 : SUCCESS in 4m 57s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/9f26b4e3b3644d19bf49c33f78365382 : SUCCESS in 4m 15s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/c38e74f70ee3411f81c38a11caf3733c : SUCCESS in 5m 09s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/5f9be7ee63514573a8a80480cff6a968 : SUCCESS in 10m 36s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/1f86eb2095da4f9ebac93f86f6d4c1bf : SUCCESS in 26m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/41f125947653401b83aa0b64fbffe97e : SUCCESS in 6m 39s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/d2f47b77c08d4d69acb6e18d5104b653 : SUCCESS in 11m 34s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/0c2870fa46b34d19b7cc233f724df90a : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/89d2a504533448fbba5863212b6abd11 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a02851e3055e4b4fbec88d12cd1ce1ad : SUCCESS in 24m 37s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0422a13b565a4557a59e0612fcef036b : SUCCESS in 5m 35s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/583c4eb8548c423ab4450ccec54d2fa0 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba9a24acb7e0446da1f428b4c4d5cba7 : SUCCESS in 6m 20s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/e4a7c0132e514e3783b1f6a85dec0478 : SUCCESS in 19m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/db480f8d6cdb49f292d6eeba8f054921 : SUCCESS in 5m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d2850e6798b4638b11c6b98d3f9bfa3 : SUCCESS in 45m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4208dc360934e1dbc8863b0ed2e0f72 : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":2},{"id":"83af04d86108cb83164a8a227385c965c4aacab3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-25 06:37:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6b7bc65b3693f8ac994a51cfb3467611dfdf591d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 06:39:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9a6b10e53f9e27432e8be5e3d969413c7f1f6b6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:30:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fe0a3a8840d4ade87395bc359b5cd0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34908d7ee3a042bf8bdca3594409ba66 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c71ffcf4a1e1499ebd350cfbebfe3ff9 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3048738dba8c41ac94618b6b52180148 : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/35d3cc411c394bfd8e57165ba53437ca : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ca1c67fe4ec4d8b946c02ae1e367a8c : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ed73761ea7649528d804bd08725fb6c : SUCCESS in 2m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3a38d39d84934f01b51ecb12aed04fcc : SUCCESS in 5m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/02b368a784f34c7483b245b12a7d4a16 : SUCCESS in 6m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/60cdab209be34a63805507ff6d0b2b90 : SUCCESS in 3m 17s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/0c4e965da9304ec989d23239bbbd528e : SUCCESS in 11m 23s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/5f02a3a9809444268fc0db6dfd1dcb78 : SUCCESS in 4m 46s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/3eaad2422c4443b29a5389d59aa1d122 : SUCCESS in 4m 19s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/01e8e19bd82846d8b7aa970ed84e9eb1 : SUCCESS in 7m 10s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a363d235cc5d48d3b7ef3a76bb15da2d : SUCCESS in 19m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/16a06c7ec7b34972aec5196a795af6bb : SUCCESS in 4m 18s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/14badfd3f6754924a291755f5b348c4e : SUCCESS in 7m 08s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/79af230b66764807a1cecac6028b09c4 : SUCCESS in 11m 20s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/325109db9e6c493dbb5e5e5caafbd06f : SUCCESS in 3m 27s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/94e01abdccbd40fb8f987b68869c5e57 : SUCCESS in 8m 27s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c03b86d42346436aa0d6300f236628b9 : SUCCESS in 24m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2241c1638e3043f5af88675bf612b421 : SUCCESS in 6m 38s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/31f52ed495ce482cb8b4a5bc4b74989f : SUCCESS in 13m 43s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/ed7399824c2e4a27a2725c8167d6218d : SUCCESS in 2m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cb482e2a7bd34b7098529750efe7d889 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53a51a176baa4b7b91e5873194f7fb98 : SUCCESS in 21m 21s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/b7c8ccd5e2ff44b4bc0af16230228a31 : SUCCESS in 5m 34s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/ecd1f73391c8406d9a90c6cf6b9a4aea : SUCCESS in 3m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ce1097658344d67a55da117a123ccf7 : SUCCESS in 6m 48s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/915c3f285df8443ca23f39639ee27468 : SUCCESS in 18m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/38f470ca20554fceb212238d8a790178 : SUCCESS in 6m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7755fcc46aad495387d0c22fe80ab486 : SUCCESS in 32m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d269df8cd91f4c24b516504d41d43666 : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":2},{"id":"512aebb0371bafd9e0c9c0e9ce6927b4ee1837d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:30:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"12d93a7d153a0ae4a8aa33d2809e89c15be4a7c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:31:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3564cec8d4a94632a61d1f7fabb8dfe6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d05e12232f37426e996712461ca97b7c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"33a97c3c6313dcce08d92fb0bf31627e9b384fdf","revisions":{"1882043662f6792b1b1748b071bb2e48175c4de7":{"kind":"REWORK","_number":1,"created":"2026-07-17 16:59:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/997772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/997772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/997772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/997772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/997772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/997772/1"}}},"commit":{"parents":[{"commit":"1244391f2eb1a5b626c84ea9623d631ce5820ff7","subject":"Merge \"update constraint for oslo.policy to new release 6.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1244391f2eb1a5b626c84ea9623d631ce5820ff7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-17 16:58:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-17 16:58:31.000000000","tz":540},"subject":"Remove python-vitrageclient","message":"Remove python-vitrageclient\n\nThe vitrage project was retired and we no longer expected new\ncoordinated release of this library.\n\nDepends-on: https://review.opendev.org/c/openstack/heat/+/988034\nChange-Id: I0530c6e21f69ab1bfa223e743ef030b195c0a1f7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1882043662f6792b1b1748b071bb2e48175c4de7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1882043662f6792b1b1748b071bb2e48175c4de7"}]},"branch":"refs/heads/master"},"33a97c3c6313dcce08d92fb0bf31627e9b384fdf":{"kind":"REWORK","_number":2,"created":"2026-08-31 18:37:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/997772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/997772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/997772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/997772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/997772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/997772/2"}}},"commit":{"parents":[{"commit":"b9e908e228b5fc554b55ede5460dcd06f1cad7c4","subject":"Merge \"update constraint for python-watcherclient to new release 4.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9e908e228b5fc554b55ede5460dcd06f1cad7c4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-17 16:58:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-31 18:37:39.000000000","tz":540},"subject":"Remove python-vitrageclient","message":"Remove python-vitrageclient\n\nThe vitrage project was retired and we no longer expected new\ncoordinated release of this library.\n\nDepends-on: https://review.opendev.org/c/openstack/heat/+/988034\nChange-Id: I0530c6e21f69ab1bfa223e743ef030b195c0a1f7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33a97c3c6313dcce08d92fb0bf31627e9b384fdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33a97c3c6313dcce08d92fb0bf31627e9b384fdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
