)]}'
{"id":"openstack%2Frequirements~997923","triplet_id":"openstack%2Frequirements~master~I7111873be93101383b536d8538ef0461e4c8e194","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2026-08-14 14:57:20.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-08-14 13:03:55.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I7111873be93101383b536d8538ef0461e4c8e194","subject":"Remove unmaintained python-consul from global-requirements","status":"MERGED","created":"2026-07-20 04:10:21.000000000","updated":"2026-08-14 14:58:16.000000000","submitted":"2026-08-14 14:57:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997923","meta_rev_id":"7dc0db65f1c131636f75d3c9a8330515a9fc6885","_number":997923,"virtual_id_number":997923,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-14 14:57:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-08-13 15:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-08-14 13:03:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-08-07 04:33:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-08-14 13:03:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 04:16:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-20 07:15:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-08-07 04:34:00.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-08-13 15:32:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"0e2e1309b368f9c516f435f2e5aebb85bbe03a86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-20 04:10:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8225376540ae218de9000adf4cb92aaca2fc8093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 04:16:33.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/masakari-monitors/+/997922 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"a229b1c80e6e9dcb351597dd4235b3f95d0b2f69","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 07:15:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dff066a55f277a10f2bf38cc688ad54d683e00ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 08:46:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d01a4b3cc6a41ba890e413da613edb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e97b3c0a41c4d4d9b484eb3387c5f7d : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb1c5f3d72904b528fc576ecc1b5ea01 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02e204cab6ca42aa8099047b2a91db7a : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/589fa4f9594b42258e0a791bc4268022 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b88d6206d8c340f0b1e9ebad42076f96 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d858c01cad22438e8533f7da84184dd5 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/222131618f6f4f269698e6acb23d3d62 : SUCCESS in 2m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/064f05a7253d445a9d5d108f9a1cdf4e : SUCCESS in 6m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2ab4f49f818d45fa87c1de7438baf6dd : SUCCESS in 5m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0d9b4bde8ed247838e32cc6cb274766b : SUCCESS in 5m 36s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/b79d4f40b5f640b6b0e61fefb077669d : SUCCESS in 4m 33s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/8d8ded44a5404af8951d5cc17fca5e67 : SUCCESS in 4m 41s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/a037ad8b92234cada0d09297fc1ee65b : SUCCESS in 5m 37s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/b36e15301c45436d9c447f277045634a : SUCCESS in 11m 16s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/f844760f5aa242f0bf4d05213fc87266 : SUCCESS in 4m 49s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/2b3da1db32ec4630a8087b177bdc875e : SUCCESS in 4m 27s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/54cf3b1540c74bd0a08f763a3d32dfdb : SUCCESS in 7m 18s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/844792b26656421d9113d265f70f01a4 : SUCCESS in 18m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/920e26d5faf145f78cdf31c7b69842dd : SUCCESS in 6m 22s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/d6b69b536fe446d18d9725a05cdc3629 : SUCCESS in 6m 29s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/b29957bf139b41be8ac61f7b6a357484 : SUCCESS in 12m 21s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a98e1be0a16444209e59ea69036ef02a : SUCCESS in 2m 37s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/818514ddda664baba1e7ca1ede0a991b : SUCCESS in 8m 37s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/24f6d1d5b5574f8bba6cfa864b2b1feb : SUCCESS in 4m 11s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/01896cb09a3f4d4486570e09a5d7463f : SUCCESS in 4m 24s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/f52ca0d962294334a4fddda1e685766b : SUCCESS in 9m 51s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/cb539b5f5a0d4b78a4ced2191beaa8fa : SUCCESS in 25m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4b237d76c82847ae9b203baf14cdd044 : SUCCESS in 6m 28s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/2ec005eaaad647d0959b92acdbdcdfdb : SUCCESS in 12m 32s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/818e27ba31b946f68a1ae42dda97b596 : SUCCESS in 4m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/022836fffcfe4eb19b0e18e9d20d3e61 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/556c7eb9407d4f00b4ffa356b1c8474c : SUCCESS in 23m 27s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/3f41ab0b282d4d1bacc7bd9e4e953e48 : SUCCESS in 5m 01s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/985fd998707b4b1f86d2719cdd986f86 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f8e69906dcef4cb8872c06ec63836fd7 : SUCCESS in 6m 29s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/071572c516c9429d8c2cafc5849705c8 : SUCCESS in 18m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8cbe7d4489a04e24bb1f6679706bedf6 : SUCCESS in 7m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30c31d208a89480aa23b441a46b297c3 : SUCCESS in 58m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e92f801ce9f4da5bb64623e983456e1 : SUCCESS in 1h 20m 28s","accounts_in_message":[],"_revision_number":1},{"id":"40ee3824cb321c8f748f11f83c008343ac394dba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-22 05:17:04.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"af5079aad44d14d132bb30a04cfe9bb56ce30ee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 06:53:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/376104bc9eeb4fcb90d6dc152db0457d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/258ce3b4aa3c4dd0acc63da88b1282f9 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e2f946e44e544658fd53c5120b43d2e : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce730e7c51114420baa079f8b7489f06 : SUCCESS in 6m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1a873e616b24ab38e2b85c56b9dfac1 : SUCCESS in 2m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/85b11e1c842e478cafdfa2e42ce82b92 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/995c5f9e429e48fb9c54da5867637b02 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7abd37221404eafac65ab8b70d6f467 : SUCCESS in 2m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e2dc522f379f4d69ad18e62dc39b58c0 : SUCCESS in 4m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/42eff52403c9428fbb93841bf7004f37 : SUCCESS in 10m 32s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4f7f946f24ea4e48a385d1f394a0b37b : SUCCESS in 3m 38s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/2e89c6a594ea4d5e89e910738da475f4 : SUCCESS in 4m 27s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/3c0b967f334b412daeed6f54acc029ec : SUCCESS in 2m 47s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/66943eb994d347bd8f737c59d07d1693 : SUCCESS in 5m 00s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/732a5eb3109d46b6b15f78055291d47f : SUCCESS in 10m 56s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/9a95e1b125e84379b88c90b4c792f2c0 : SUCCESS in 4m 20s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/2a9636c2de6544289bcbf86e7923207a : SUCCESS in 4m 10s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ca0ec81767fe4598b9b32bf3f3c70255 : SUCCESS in 6m 23s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/3f0cd6bdba434726975534e58cd5210c : SUCCESS in 7m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c71ec8e498846a68cc906d0eac40df8 : SUCCESS in 4m 35s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/f3429c679756451789e6bfb3330860ea : SUCCESS in 6m 50s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/933621bba642459d9a609fa8987d59f4 : SUCCESS in 9m 22s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/9c7759faea9f40ddad7fd1401eca89cc : SUCCESS in 2m 54s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/6e09b41100f0490587a43c91647e1bd5 : SUCCESS in 7m 21s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/e3f05f32d8e24e3c99cda3c4e4667746 : SUCCESS in 6m 48s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/e635b51d6aff478ca7299e5855275be6 : SUCCESS in 4m 44s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/bdf180d8144d43179d0a2941fae3fed4 : SUCCESS in 9m 48s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/a37ab4cbdc6546f8b86a0a70aedcdaa6 : SUCCESS in 22m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5f22874dee924825864c037a2b0c7e64 : SUCCESS in 6m 25s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/37ef25aac19d479c905e030c0d03172a : SUCCESS in 13m 25s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/45364784fdcb43f7aa01d429eb5726e1 : SUCCESS in 3m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4d95dcb159a43dda653c3f1df6a5021 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9c87be011ef0483fb7450ee12b67c86e : SUCCESS in 24m 14s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/6040d2de549a4e4c8430e78b4d8a84ec : SUCCESS in 5m 11s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/35a18ba6c87345b3a8b260a164a144ed : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5900c8e3ef57475bb66a72d6ad22e699 : SUCCESS in 7m 08s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a71808afb5e54499a184f69aecefb5f5 : SUCCESS in 13m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/30aa7cf8c8e34b91a32757bc0aa9bfe9 : SUCCESS in 5m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05c4a92e609a491dbf0bdbd1315601f2 : SUCCESS in 57m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae10b2582ebd436d8c4a64782eb4db37 : SUCCESS in 1h 29m 54s","accounts_in_message":[],"_revision_number":2},{"id":"4e1557a74111bdb8999f785b0385412864709531","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-07 04:33:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f76ef33622d969ce87f652db3880ca4e06af5cc8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-13 15:32:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"062cb24dfca825c9663f0f53eb56905a754e8a86","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-14 13:03:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d46e8fb3c75bed2db1b29d9e1a92d61f9cd18c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 13:04:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"68ab77bcb7f8a9085cb2ad3b2c2839fc3e0322b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 14:57:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/106a0ce3bda541fbbeaf8c09428eeeb6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73fba60010194383bcab118a8ec3e6d6 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7976fa872aeb4758b22c0fa25551cbdb : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe0ba49822404b5d9bf64f8780fdd31e : SUCCESS in 2m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7836efcd85764514bb10d625ad951942 : SUCCESS in 2m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e600090ece14cbbb77f2cdbf0f75d8c : SUCCESS in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/87b4e7d7433441d1b404258fce4375f4 : SUCCESS in 3m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c2524151036b429696af247593b71472 : SUCCESS in 9m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f61f56632e9f40c7831a2b8489af5e61 : SUCCESS in 6m 02s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d4e3b238cf7a4173b6cd75334c1e30eb : SUCCESS in 5m 07s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/2c42267cadd04f4b98fa707ec8766cbb : SUCCESS in 10m 52s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/06961bceaee94315b4eee1932333e41c : SUCCESS in 4m 26s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/64ddd496102142e38bc66200757fd0c4 : SUCCESS in 4m 43s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/2a422d445dfd4d72b9680f3e523e5423 : SUCCESS in 5m 42s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/2c04b25751e04ef2aeb2e25e82080401 : SUCCESS in 7m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/901d31c70e2d463a9d2b83519414f9de : SUCCESS in 4m 11s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/4cb0155bbfc746da8e0f6eac5fc9b6a0 : SUCCESS in 7m 18s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/cb91c4839e8c4709a980935034f45116 : SUCCESS in 7m 26s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/50b3f89596b547e3a11e73eed1efc3d2 : SUCCESS in 2m 27s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/9effa597cc88426881255469927c635d : SUCCESS in 5m 03s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/dfcce22829a94b8b89ba90ac6e8eb369 : SUCCESS in 16m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a8d6bf99e74543a28ee74f48bc03f85b : SUCCESS in 6m 50s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/c8eb5dc4e6a446cfb517ae77e503f460 : SUCCESS in 12m 21s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/0feef84642cb4ab195e59fe63ef3db03 : SUCCESS in 2m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/973c16bfdbfa4a88801fae6f7913f23c : SUCCESS in 6m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0c9d516ecbe493cbc89eba77c4dd893 : SUCCESS in 25m 38s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e4d4984c6c094af19560937e15453e2e : SUCCESS in 3m 35s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f6e6be94f3404c8a8ff09f773b41d15d : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/13937f86d2e544f2bb7606ff40636915 : SUCCESS in 7m 01s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/337379193fef43ac972c3a7df3b41e4d : SUCCESS in 12m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a056767b490b464081d04059999b3058 : SUCCESS in 7m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8030ee9c18fc42e4b3284175951f9436 : SUCCESS in 1h 00m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85803fda5674488ba2e99412d6075444 : SUCCESS in 1h 47m 42s","accounts_in_message":[],"_revision_number":2},{"id":"9d1954dbd8dd5181a0b538fc3765b430502ffe2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 14:57:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7dc0db65f1c131636f75d3c9a8330515a9fc6885","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 14:58:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9cfcef5c9d6409a977a2940d81d0131\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c579bfe417a4ad09e83e64927b7b516 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a61dc45bfcb9e26a58549e4b1ba85f7c1a9234fa","revisions":{"66c0231622a34c9eec320d704d1ffc91dfda6d2b":{"kind":"REWORK","_number":1,"created":"2026-07-20 04:10:21.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/23/997923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/997923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/997923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/997923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/997923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/997923/1"}}},"commit":{"parents":[{"commit":"8e0ac76353a2f701f572469e44b417a6aaf26083","subject":"Add py-consul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e0ac76353a2f701f572469e44b417a6aaf26083"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-20 04:10:08.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-20 04:10:08.000000000","tz":-240},"subject":"Remove unmaintained python-consul from global-requirements","message":"Remove unmaintained python-consul from global-requirements\n\nThe python-consul library is unmaintained and archived. Its sole in-tree\nconsumer (masakari-monitors) has been migrated to py-consul, the actively\nmaintained fork.\n\npython-consul2 is retained as it is still referenced by the LY Corp\nDepends-On chain.\n\nDepends-On: https://review.opendev.org/c/openstack/masakari-monitors/+/997922\nChange-Id: I7111873be93101383b536d8538ef0461e4c8e194\nSigned-off-by: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66c0231622a34c9eec320d704d1ffc91dfda6d2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66c0231622a34c9eec320d704d1ffc91dfda6d2b"}]},"branch":"refs/heads/master"},"a61dc45bfcb9e26a58549e4b1ba85f7c1a9234fa":{"kind":"NO_CHANGE","_number":2,"created":"2026-07-22 05:17:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/23/997923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/997923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/997923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/997923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/997923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/997923/2"}}},"commit":{"parents":[{"commit":"c18899585bc434315e34603e9bed55342ae43329","subject":"Add py-consul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c18899585bc434315e34603e9bed55342ae43329"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-20 04:10:08.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-22 02:46:37.000000000","tz":-240},"subject":"Remove unmaintained python-consul from global-requirements","message":"Remove unmaintained python-consul from global-requirements\n\nThe python-consul library is unmaintained and archived. Its sole in-tree\nconsumer (masakari-monitors) has been migrated to py-consul, the actively\nmaintained fork.\n\npython-consul2 is retained as it is still referenced by the LY Corp\nDepends-On chain.\n\nDepends-On: https://review.opendev.org/c/openstack/masakari-monitors/+/997922\nChange-Id: I7111873be93101383b536d8538ef0461e4c8e194\nSigned-off-by: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a61dc45bfcb9e26a58549e4b1ba85f7c1a9234fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a61dc45bfcb9e26a58549e4b1ba85f7c1a9234fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
