)]}'
{"id":"openstack%2Frequirements~997978","triplet_id":"openstack%2Frequirements~stable%2F2025.1~Ie6b3cd5c8dcf2321ed3240e6200bf6fdc42749bd","project":"openstack/requirements","branch":"stable/2025.1","topic":"new-release","hashtags":[],"change_id":"Ie6b3cd5c8dcf2321ed3240e6200bf6fdc42749bd","subject":"update constraint for ceilometer to new release 24.0.2","status":"MERGED","created":"2026-07-20 12:38:43.000000000","updated":"2026-07-20 17:51:28.000000000","submitted":"2026-07-20 17:51:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997978-new-release","meta_rev_id":"c3dbb8da9c4c54341989a82c301e0036ba7f4072","_number":997978,"virtual_id_number":997978,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-20 17:51:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-07-20 15:45:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-07-20 15:45:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 14:38:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-20 15:45:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"b1b33c79ccc324b488d78fa92b5abdda86d8fca2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-20 12:38:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b557460065727997bf1362e32a6338a311c135fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 14:38:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fde835530fb1438a958662c39ca9481b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b6855f9d64d42299312b3d24dba459f : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee8b734d5e3547dc8ee1f389600a0291 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4c07fe5dc3f49318178ab13ab923e79 : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8466b9fb34304dc6b45e4047a82e4360 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b3c3fa65271b476c823d108b7182e732 : SUCCESS in 3m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cfa3f8877ea64985a0cd3592c868f04a : SUCCESS in 16m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f1ef081eb6d8427b94906d896a173120 : SUCCESS in 9m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/625ff55a8a564279a5440d49c19f4841 : SUCCESS in 19m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ead131bd432a46a4a4ab1264503d109e : SUCCESS in 20m 04s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9ec682c4c6ae48f59d1f10da31041cec : SUCCESS in 5m 47s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2d47ee4f5bfa410ab352fcc1ed00f714 : SUCCESS in 6m 00s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c4cec2b7004b4644808362a8b5eb0055 : SUCCESS in 4m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2ddcc18321f24a9cbe866b207d8f658b : SUCCESS in 13m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/31c59ccd07974d86b3a246fd08722cdb : SUCCESS in 6m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0bb33a7180004ce3b6e5f643c54b6ef6 : SUCCESS in 5m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b1a6fa4307174327a1c16887f4587c9b : SUCCESS in 8m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dce11b6cfafc4b4c9c81983a123750b9 : SUCCESS in 16m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54a5515232cf488c9317b8e4bce843e7 : SUCCESS in 7m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/be84be72cfc8470484ff25af4b775101 : SUCCESS in 8m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3e70227be93245928e8bc9a23aa75108 : SUCCESS in 12m 58s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b87f4e1f04b64ce3be916939d2b98681 : SUCCESS in 2m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1e35a48bf1054eeb94aa71b690fb1225 : SUCCESS in 9m 20s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ed438c0606a145cf911878ebd6ea61ff : SUCCESS in 4m 49s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1ae9632a26a044a2b8ae0005c726461b : FAILURE in 4m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/dabde6c1dd2a4a6394f06505e36fd265 : SUCCESS in 12m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/df4f641a1dc6406eb4f7de62216725e4 : SUCCESS in 21m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/81cc24e9074c4ffd9265b971ed1aefdb : SUCCESS in 9m 18s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/aad91f79dbc84299a373d4b53fad415e : SUCCESS in 13m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7d7fdb68c0e74220b38a14b059ba6c8c : SUCCESS in 6m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5fd998217fa14578b2535f2216d79cbb : SUCCESS in 5m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/da04e8e9794943dea65ec20230a9e661 : SUCCESS in 28m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2d45924eaba748d3ba8c1ffb656d0fd1 : SUCCESS in 9m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/df9e129627dc4382913740b65d3cc777 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3631e3e67952444dbaf9a7d6c3b3cd5a : SUCCESS in 7m 32s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ead7418a0f854223bc7f36c8decb5abb : SUCCESS in 13m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2aa3760c30c347689407ab6929e8042c : SUCCESS in 1h 01m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f87266f4bf7c4ae8898d1e8797a21191 : SUCCESS in 1h 53m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3bf8619b1214ca2c27e64b525c0fa06a9213fb40","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-20 15:45:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1bf5269240a4033d07a68ec02ee4ea60d5661d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 15:46:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"356776e30281557788431e520bcc288f95177a88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 17:51:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/065b76a550e14b7b8901a6eeeed995ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0704ca730dd048cbb560a4ba2c13a7be : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7f95afbcbb5432ba5ccc93297fed619 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db6172f5d7cd43c49c016f6b93205a06 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c600cb96ff6a46f2a6f11a49609efcb9 : SUCCESS in 6m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92532e7e956249a59d96bdd47a9e634b : SUCCESS in 4m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/eadcba3f5a6c45248cc3b0e68bfbc98e : SUCCESS in 17m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/420e7b8b0afd4b9eb6bbc7551794fdd8 : SUCCESS in 11m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1c3819f9dbcf43969475a768e569410a : SUCCESS in 18m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f6a1462f7dcd4d49a650bfed16cc5b51 : SUCCESS in 21m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bd85e4a34d2a4316a9e4bd6f5fc25d29 : SUCCESS in 13m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/854df7e83a864c78920e5f8fabc07d81 : SUCCESS in 5m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1be967bd970f48edb3daa2dc7cec4266 : SUCCESS in 7m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/932c6a2683204c2ca96bb03e8baf3be0 : SUCCESS in 9m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/217d84f4c19f4fb7826a2e043ae93d18 : SUCCESS in 16m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f82b8bde4364dc08d223c4eb39d9007 : SUCCESS in 13m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cef25af231aa4c13a2d74ef7aeb0abf3 : SUCCESS in 9m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/895065fee3ea4f25849afd93de66a089 : SUCCESS in 3m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/437e1df33fe24f5a948260d956fde1e1 : SUCCESS in 11m 37s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f8280cdaaf5e4934839a04d23af7e845 : SUCCESS in 21m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/31f4bc142fc141ba9547626f48663c87 : SUCCESS in 9m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4d4796892030452bac0decf8ce6bb5aa : SUCCESS in 15m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e7ade34bbaa642ecb8d643baa145ac8a : SUCCESS in 8m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9d05284c7e9d4d60b93e03391bc4a696 : SUCCESS in 5m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d58a50f436f847e09accc54036d5739c : SUCCESS in 28m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/571d9af54b7d44fb879e35011bd6d677 : SUCCESS in 9m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f72a277b85484b5e85a27f2456183b2a : SUCCESS in 8m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/423ca7fc4b7546f59ec9d1830070cfba : SUCCESS in 15m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f71f07e72724fd2a69e3796c50bad35 : SUCCESS in 39m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34eb3a2ed75745269df3619f02462c3d : SUCCESS in 1h 34m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c3dbb8da9c4c54341989a82c301e0036ba7f4072","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 17:51:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d66fc1dde4649824afa118480a856d545daa87a","revisions":{"4d66fc1dde4649824afa118480a856d545daa87a":{"kind":"REWORK","_number":1,"created":"2026-07-20 12:38:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/997978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/997978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/997978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/997978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/997978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/997978/1"}}},"commit":{"parents":[{"commit":"29314ba6f30cb991027a3dcfabbed8d800827ad7","subject":"update constraint for oslo.messaging to new release 16.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29314ba6f30cb991027a3dcfabbed8d800827ad7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-20 12:28:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-20 12:28:42.000000000","tz":0},"subject":"update constraint for ceilometer to new release 24.0.2","message":"update constraint for ceilometer to new release 24.0.2\n\nmeta: version: 24.0.2\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I210271cb20d0a5ecfc16d2c2f0a961cd6093f94b\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Emma Foley \u003cefoley@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie6b3cd5c8dcf2321ed3240e6200bf6fdc42749bd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d66fc1dde4649824afa118480a856d545daa87a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d66fc1dde4649824afa118480a856d545daa87a"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
