)]}'
{"id":"openstack%2Frequirements~997980","triplet_id":"openstack%2Frequirements~stable%2F2025.1~I5c8137f75105a2d6f186a3670efa86a8a3b50b76","project":"openstack/requirements","branch":"stable/2025.1","topic":"new-release","hashtags":[],"change_id":"I5c8137f75105a2d6f186a3670efa86a8a3b50b76","subject":"update constraint for python-swiftclient to new release 4.7.1","status":"MERGED","created":"2026-07-20 12:49:03.000000000","updated":"2026-07-20 17:51:36.000000000","submitted":"2026-07-20 17:51:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997980-new-release","meta_rev_id":"01f04e7d5025c2c61ed3435bddd3c11a2146e298","_number":997980,"virtual_id_number":997980,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-20 17:51:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-07-20 15:46:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-07-20 15:46:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 14:51:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-20 15:46:00.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"75da8e4d2383183cb7913838ac55217cf141d285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-20 12:49:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4d481fc597953f11cd3f6b28a624f21bc3a7c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 14:51:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6ac2ace58d44745beb1f568621b07f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca08604e207949a2936843b26ab5f157 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a558df4d0d44709bda41f7334325537 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95d6fbc9d95945dea17ba7f0f87fe6dc : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1de71b4a89ef477cad84bb4b3f98d1b3 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/411bd6cae40b4a2abe2ddbd95d8240b7 : SUCCESS in 4m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/07eba338444543afa234ef9db7c9321b : SUCCESS in 16m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/04de7da6b5bb4c5994df40d26b50d439 : SUCCESS in 11m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6a624b3739b04ca687ef5b39992b674e : SUCCESS in 20m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1783a1709a5a4702937190c87685543f : SUCCESS in 25m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f899b6109e604eb588f43b49a67b332d : SUCCESS in 5m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5b08a5dd6fc1483e8b0b749ee01dbbfa : SUCCESS in 4m 34s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/212e7d23118b47a3847e05a025192603 : SUCCESS in 5m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5f5826ae4b1945e097f042c887963d64 : SUCCESS in 11m 26s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e69c0bb230474feebc667876dfa2da58 : SUCCESS in 5m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/95c3ee2966cb428ab72830ef2bec7305 : SUCCESS in 5m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d32e00bbbc274884ad9d87ccd8514961 : SUCCESS in 8m 12s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8ac6d85e6d9b42f6bc083ca5f489a4b5 : SUCCESS in 15m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b73f9ef5bfc488cad564476ff15aadc : SUCCESS in 6m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4ed696b889644ad594ac438e4b8644df : SUCCESS in 8m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/163fe255283d4281865af9ffb8112ffd : SUCCESS in 14m 06s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3baa3c6ef42c4467a8d0e6c226bfe8df : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/70cf74bc4280409dadba144151e855a9 : SUCCESS in 11m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/76b3e070df8d434e8e4ffb485484cb45 : SUCCESS in 5m 56s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c2e534e3e54d4338b9d325272753beb7 : FAILURE in 4m 36s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/fe0363ffe0d442b9a6409e580ab3d11f : SUCCESS in 13m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/704723faff5141f3ae692204699fde22 : SUCCESS in 21m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ed7032d47f0e401883fd99f4534e9490 : SUCCESS in 8m 36s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fbf41f21d84c4656a47de038dec16718 : SUCCESS in 16m 34s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f97070459c2047fd83ba27b578f93b79 : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/360d3d60d67248308b1574b4f6bb647f : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95ec930f59d947b08759d211f038087b : SUCCESS in 27m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1ecab301ff2a4563a7d9766288328a03 : SUCCESS in 8m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e60fca32760848fdb0bbd8cae021943c : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bda2094b1a8643b58dc4f5c9e1eed8f0 : SUCCESS in 7m 02s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a1f1465927b74d05924aaa411b28586c : SUCCESS in 20m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1811177e5f1e400ba040d9a235cdd22a : SUCCESS in 1h 05m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2565206bab954e5088af99e799dce2ac : SUCCESS in 1h 55m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7e3078e2a6c3d107ee9b46fdb89391fe4fcc43e1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-20 15:46:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e01f1a7c309a9d86f6f344b6e9115abb41d90c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 15:47:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"667d8ad974a59fd460e6247003127ee3cb9f0b3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 17:51:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a280706a07d3419cbc885cc484e5628d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe58fedfb0b145368a370a87c7099838 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/899e97c85c6848a98be62d26d2118f28 : SUCCESS in 6m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86c59ca1256c41098338ba92b5e8da7b : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc9ceae0f1c145c28a72fe201d1ff596 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5401a714d1a4bf2a7828d2b2659c13d : SUCCESS in 5m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/81386dacbd754ffbb1e9ecaabe18f0d6 : SUCCESS in 16m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/072b6da7283f4a9289f72c9b54987dbc : SUCCESS in 10m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5d5995f8e6174be9995758cad65e709a : SUCCESS in 8m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bfc44b04ff594e958b8c24a8c05d0f15 : SUCCESS in 20m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/188185c625144ea9a86735449ceca437 : SUCCESS in 13m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b5db2cd3bb3d41398fa915e3127e1637 : SUCCESS in 5m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cddfb390ccf346368a365121689d00ee : SUCCESS in 7m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/38694df6fd134237a7275451c6169589 : SUCCESS in 8m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ec8628bdc32c4f549475c4e89cd5a7b0 : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56109a2cea1949deb2d580adc01b0506 : SUCCESS in 9m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1c39de6034644a26a1ae870fb18f302b : SUCCESS in 8m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3cd8c72c955e4be3b8e160285ba55388 : SUCCESS in 4m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6e4e3aada3d54664ae3a29e1fade8fcf : SUCCESS in 7m 33s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/eefc61c6892f4c08bef71335d89cc5eb : SUCCESS in 26m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a27fb87a774040fdb2d9fe3f51a12d7d : SUCCESS in 9m 13s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/750fd219ddfe4674bcfd6294f7f945a4 : SUCCESS in 16m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3e944d54bd3e4bdaa96e337fc3746eb3 : SUCCESS in 7m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/142c98b4f0b84ea3a44262d025302c10 : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47da8f04d37a461086e00c49ac1706e5 : SUCCESS in 25m 52s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f5f5160b3652468694e6f543057c26a6 : SUCCESS in 7m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d65f1dfb09fb40a39e2ca6e4ab5c0c22 : SUCCESS in 6m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bf01aaf70a5f45339e39d6946f210916 : SUCCESS in 7m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d42d5a109c424a559a68afcafe21b6e0 : SUCCESS in 37m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7af61c7c672e4e55a2eb2f19bd4c93ee : SUCCESS in 1h 20m 29s","accounts_in_message":[],"_revision_number":1},{"id":"01f04e7d5025c2c61ed3435bddd3c11a2146e298","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 17:51:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd1e54d213187378c8edcbff43320b14820d927f","revisions":{"bd1e54d213187378c8edcbff43320b14820d927f":{"kind":"REWORK","_number":1,"created":"2026-07-20 12:49:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/997980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/997980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/997980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/997980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/997980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/997980/1"}}},"commit":{"parents":[{"commit":"29314ba6f30cb991027a3dcfabbed8d800827ad7","subject":"update constraint for oslo.messaging to new release 16.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29314ba6f30cb991027a3dcfabbed8d800827ad7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-20 12:39:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-20 12:39:00.000000000","tz":0},"subject":"update constraint for python-swiftclient to new release 4.7.1","message":"update constraint for python-swiftclient to new release 4.7.1\n\nmeta: version: 4.7.1\nmeta: team: swift\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I38685fdba4244c22ce0cef7de4b1096d677fb1f5\nmeta: release:Code-Review+1: Tim Burke \u003ctburke@nvidia.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I5c8137f75105a2d6f186a3670efa86a8a3b50b76\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd1e54d213187378c8edcbff43320b14820d927f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd1e54d213187378c8edcbff43320b14820d927f"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
