)]}'
{"id":"openstack%2Frequirements~997986","triplet_id":"openstack%2Frequirements~stable%2F2025.1~Ie2491776f4ae701f14452957485a0fbe7f9facbe","project":"openstack/requirements","branch":"stable/2025.1","topic":"new-release","hashtags":[],"change_id":"Ie2491776f4ae701f14452957485a0fbe7f9facbe","subject":"update constraint for os-brick to new release 6.11.1","status":"MERGED","created":"2026-07-20 13:16:47.000000000","updated":"2026-07-20 17:51:32.000000000","submitted":"2026-07-20 17:51:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997986-new-release","meta_rev_id":"fe6bca6200eb053f9165a00772b2c96d4c3d7b5f","_number":997986,"virtual_id_number":997986,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-20 17:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-07-20 15:45:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-07-20 15:45:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 14:49:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-20 15:45:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"47801d92430073ec4182c0f3be6d6e15c96e7b9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-20 13:16:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81218687b708bd409149d4c527b93deb44a02687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 14:49:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/383ae82de847439d8ddf50820f88d7d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2510582dddfa4f488c0c01f596511335 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82349d2ecdf54dfa9f0709914dd768a3 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4f17e1b8a7547a3b328cf3d55d85b80 : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b42494f852014b9ca920d284e069749d : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/137bfcf801d74ccdb66afc6172169ae4 : SUCCESS in 3m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7e49eefa96cd4986a5d37f47ab73afae : SUCCESS in 16m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/81dfa826066d4ababb08b16845612fb3 : SUCCESS in 9m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0c8d63c5b32d43a49bb9f221e354a7c1 : SUCCESS in 17m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/24404a83776a41a2b719972a8bd5a3a6 : SUCCESS in 20m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5ab13b3ed9624796ab7a4e9ca304db2b : SUCCESS in 4m 47s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/be7926c6037944dcbab3ed6ce4da42eb : SUCCESS in 5m 16s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1b8854b12fa84fc1914c4d6f71ca2510 : SUCCESS in 4m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b33ee872495d4c6896cb5b4fd4a719fb : SUCCESS in 11m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/59ec7e18bd5b4c9ab7e37fad1fc1fab3 : SUCCESS in 5m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f6db7203b962475e84013eb926f84a27 : SUCCESS in 5m 16s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1053f8d0c3c743479a34ef007be60df3 : SUCCESS in 8m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b9a32ea3f7674988a7f4978cc84cc231 : SUCCESS in 15m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b254bed32204011a17eeb65a5e4b3cc : SUCCESS in 6m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8fdc41aa640b4759aa1543ffaf775c18 : SUCCESS in 8m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7693389349e145b4b3d86501a1c41725 : SUCCESS in 14m 32s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/103d75286ed541feb4732d6b5c445719 : SUCCESS in 2m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8d587fd693b348a0beae96aa61e6ba33 : SUCCESS in 10m 54s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/154e58bcefc84136abf5faa9a2b770f4 : SUCCESS in 5m 31s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cd50e62427e042538cbb1779083c6101 : FAILURE in 5m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7c5f87b59de149fcaa80fbe4b193a8e0 : SUCCESS in 13m 12s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1182745273c54f2b9a3a359f86cbafef : SUCCESS in 21m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b74f2c18f4ce40e5af36742e32274d60 : SUCCESS in 8m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a56ac3c0951f4a4782f070ea42f9d21c : SUCCESS in 13m 10s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e7fd135ad14c4bd8b846fcb9f7f290af : SUCCESS in 6m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b3999446d2943169e2b8700ff729238 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1cb033dda4f84490a9cca372a2d928d6 : SUCCESS in 27m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/04d7a38fb5a7455bb836626cb732df61 : SUCCESS in 7m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/594748ee72934f11b0410b7d65e15b8c : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cfd653b201a740b291fda75b85677862 : SUCCESS in 7m 47s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9cb1eaf6fb94436cb254f9cc5648dfb3 : SUCCESS in 18m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07f25b38492648e5aa5de9cb5a15f730 : SUCCESS in 52m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2a6b00dedda48a7839b5245050c3b6f : SUCCESS in 1h 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a1790eec953dbf619a0f9b85bc81949c8c231118","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-20 15:45:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac2edc66ab25b070c7689aad39c131c0b21932ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 15:47:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"185012b9d81416fd263702c865a3efa0843ca2e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 17:51:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb03af81c154085b14395ccc23a10d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92adbf99c064427ea720db8c665b902f : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dae5d87e90e54b26972cabd11f9de1e1 : SUCCESS in 6m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/701689b86dfd4649b9752bec9da60adf : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2d0fb9459524dc390349e6f4b9379f6 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86d5becf24714a4bb641a243f201ec01 : SUCCESS in 6m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/879499af3d5b421fa1f2e63dd255058f : SUCCESS in 14m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9c9dd7c0be92453e9d948c98f26058fb : SUCCESS in 9m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e90bc0f090b54ae288277a7b4475908d : SUCCESS in 22m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4cbc499cbb4a474899a16df47b55bf09 : SUCCESS in 21m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/176c2aa641bc49dbb47f43c04e0050ea : SUCCESS in 12m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/305ac87a68d44d0188c6a5bfb7748bd1 : SUCCESS in 7m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/eb515a4e6c394825adb7e0f1aa5a9c2b : SUCCESS in 6m 38s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e3c7b742759349b1b721ad38219efb7c : SUCCESS in 10m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6ba05f5cf19d42c8a52933db458fe306 : SUCCESS in 15m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f5e803751164b21a1d838cf53537376 : SUCCESS in 7m 12s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/de40123e7c964be0be9aba2d36130b0e : SUCCESS in 8m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2def5a5398c44219b0ee499eb206bdbb : SUCCESS in 3m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c6bf35ba354c47a4ad8e2b6731f423ab : SUCCESS in 12m 07s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c4fcce0cecaf471aa8ebfb8b8743324e : SUCCESS in 23m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a4ad19a0c54b4e3fbfec2188f0dce290 : SUCCESS in 9m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5697ebfa1a7c4612b7bbbb5971b199a0 : SUCCESS in 13m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c887ee223f634e3d88eb65f2caa2a6c8 : SUCCESS in 8m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/863c7970a5694ddf8f2f6f56bbe4b618 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d7f5592ed394a61bb85d92cd04c77a8 : SUCCESS in 23m 28s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/67d9518a0ac24a64886f5767c7e4f266 : SUCCESS in 9m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a7d8b10c85944985a92b1994ec6d6cf5 : SUCCESS in 5m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e8c81fff3873470e99d22afe32a7be5f : SUCCESS in 14m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d79890aa7a3042e5b851fb76f58dcbf0 : SUCCESS in 1h 08m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f837503ae5d64b37b7cf3bce29197b37 : SUCCESS in 53m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fe6bca6200eb053f9165a00772b2c96d4c3d7b5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 17:51:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7fb773f92c6f0c72d37f93810f1234a3dbac3fca","revisions":{"7fb773f92c6f0c72d37f93810f1234a3dbac3fca":{"kind":"REWORK","_number":1,"created":"2026-07-20 13:16:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/997986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/997986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/997986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/997986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/997986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/997986/1"}}},"commit":{"parents":[{"commit":"29314ba6f30cb991027a3dcfabbed8d800827ad7","subject":"update constraint for oslo.messaging to new release 16.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29314ba6f30cb991027a3dcfabbed8d800827ad7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-20 13:06:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-20 13:06:43.000000000","tz":0},"subject":"update constraint for os-brick to new release 6.11.1","message":"update constraint for os-brick to new release 6.11.1\n\nmeta: version: 6.11.1\nmeta: team: cinder\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I076e010ed1a31fa5b78efc2c5a93a8456ec6412d\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Amit Uniyal \u003cauniyal@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie2491776f4ae701f14452957485a0fbe7f9facbe\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7fb773f92c6f0c72d37f93810f1234a3dbac3fca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7fb773f92c6f0c72d37f93810f1234a3dbac3fca"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
