)]}'
{"id":"openstack%2Frequirements~997987","triplet_id":"openstack%2Frequirements~stable%2F2025.1~I372bdce97ef2859c4c9b7235b5cee14dafd3f5a4","project":"openstack/requirements","branch":"stable/2025.1","topic":"new-release","hashtags":[],"change_id":"I372bdce97ef2859c4c9b7235b5cee14dafd3f5a4","subject":"update constraint for oslo.config to new release 9.7.2","status":"MERGED","created":"2026-07-20 13:23:00.000000000","updated":"2026-07-20 18:10:51.000000000","submitted":"2026-07-20 18:10:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997987-new-release","meta_rev_id":"6daac7bb900e7734fe772207eedd3b51e33276d9","_number":997987,"virtual_id_number":997987,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-20 18:10:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-07-20 15:46:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-07-20 15:46:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 15:28:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-20 15:46:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"a7bd476cb04c0d2c15dd8638efa60a165f433a84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-20 13:23:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0328bd08ff4945b9a5fbf4336b07e9bbc5f07a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 15:28:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bdf36da064445d18ec39efc8686399e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/017a455eccb4406888e67b28743154c6 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c110a88e0fd94257bf5543b5eb179a9e : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3b64d95c9834d71b6f3eec67e22a311 : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7007746efc40438fbfc693b020823c62 : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1f74b2d32b0745869cdfdba310300f96 : SUCCESS in 3m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a28a773cd9304a37853eb5f1a8734657 : SUCCESS in 18m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9b50f3f3244545b39410c5a255f26813 : SUCCESS in 9m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a6557768496e4a95bf29ecdca0fc3577 : SUCCESS in 17m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c9b537fa39534f039e629820553ceb22 : SUCCESS in 20m 03s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/18c60ffdfca8440f958c2e2f863d8d85 : SUCCESS in 6m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fd56fd0c77584978b5ff9c3a6520aa70 : SUCCESS in 6m 45s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e044711f12b5471a9f59af794f86b19a : SUCCESS in 4m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ca9f7a3716b941fa99e59212199c3922 : SUCCESS in 12m 11s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6682d0e9efc04f9b8b9a300c8f3cdb82 : SUCCESS in 4m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5df5ed3dff7f4c3a9ce7d8a779893161 : SUCCESS in 6m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1b9130e6e43c4cd6bc96abf457ff8a86 : SUCCESS in 8m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/159d6517e766494293a83459e0d97da3 : SUCCESS in 14m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e18de29086e44209bf670dfa23aa8198 : SUCCESS in 6m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/15c8aa006b0e4c5582e6fd7ac8ecd13e : SUCCESS in 7m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1a445d324b904b888894785e70facd7f : SUCCESS in 16m 13s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c5f7df7bfbd04e6a92bf23c13825978f : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d245843d94c84796ad0d6021f538811e : SUCCESS in 10m 12s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bde959bbb83b4351b14b7db7bfb8ba64 : SUCCESS in 4m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ca33b437555d4508a8c4b8b03ea7b4cd : FAILURE in 4m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/15e259e30e3c448ca9c76af0693bb0bc : SUCCESS in 12m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ee4b26fa3c0245c2be133f8fc59ae096 : SUCCESS in 21m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/aa8925f4f5944cbc95ccfc01a4a10451 : SUCCESS in 9m 02s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/19202a60775d459c9d26eb81206ce90f : SUCCESS in 14m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/14937a04d5214e55a94dec141ea947a0 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e467327334244d4182e0a8992bc9dddc : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d6c0981cea7849128947dcde4d69474f : SUCCESS in 26m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0bd56cdbc5c241f0835aa4f234e9f161 : SUCCESS in 7m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4a84fddfa2f545059162abe769cfd105 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee3e83300ffb454da46bae5a45dbf3d6 : SUCCESS in 5m 50s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bef52864bc1f44968ad2b984a4433fc9 : SUCCESS in 13m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/468459d76fd5473bbbb4df3752fd736c : SUCCESS in 57m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7310733a26ba436db829b4f413a57404 : SUCCESS in 1h 57m 59s","accounts_in_message":[],"_revision_number":1},{"id":"30d25d6504fa64ee286b69216a4f734d67f72f33","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-20 15:46:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f30bcbeb953478c9d8e296f332e2a6bdea07cf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 15:47:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"604f3ef43330b2dc840fb0d51dfdb342408c068b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 18:10:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7454589cf58c4941a7a78e888a5d05b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83d2ec20980f4381af8028d191f564c1 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4276a71f12842cfaf8dfe4922c4c40a : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/416d6d1354b04249ab4fc157d2c9a4fe : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14986f2181c74b5da8ed6ebbc31ee238 : SUCCESS in 7m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d461006236ce4179987c9f62aafe4e6c : SUCCESS in 2m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b976ff4e5e8c43b8b1b38299d6d04fb8 : SUCCESS in 8m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/edcf94c7260047dca2bbccb3040224d2 : SUCCESS in 11m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2299a937c23346ddb9df61c9e8d4d9bd : SUCCESS in 20m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2a64f5ba4514472fbe51cc5688302ec5 : SUCCESS in 21m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/aa532a9c97464e4e9c968810831dd293 : SUCCESS in 13m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/dd18cae557c44a908f8aec3596e4fc1b : SUCCESS in 8m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c23543d920b64fe4b07f8905c918d1de : SUCCESS in 7m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5e2de265ade84509bd4ff27946def0ef : SUCCESS in 6m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/09a5e573e4e54feb911515b9f5c8c99d : SUCCESS in 13m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1d4ab2554734b9e9c60686aad70d7d7 : SUCCESS in 5m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/327c281a13c743689bcb55d4c1ee93e8 : SUCCESS in 11m 07s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fd672d86251549828163bdf258197763 : SUCCESS in 4m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e2246234d0c440acb14c23e198b44b42 : SUCCESS in 10m 56s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ad7cca89f496420fbb0e159e12546a0d : SUCCESS in 25m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8be48877e0c642a0838752180460f5fd : SUCCESS in 10m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f3735856468949d480fed18c6b1725f1 : SUCCESS in 12m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3f25eb41724346e2a194667a15d10c3f : SUCCESS in 7m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/788b1f65b79247b3a0cb84b1bcebdb20 : SUCCESS in 5m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0883358619d44dc2905952c612522781 : SUCCESS in 31m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a034ba4fcb8a4f729cf787a9c0a86497 : SUCCESS in 9m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9f60d1b2f4d24eb3bcbcfa40cdac9d05 : SUCCESS in 6m 44s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fed1de52f2eb4ccb91f981dfb99499e6 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d05185e51e34abf8118222d7c0e8719 : SUCCESS in 1h 02m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3eb9ef403ad740f7927edf3f656e1c6d : SUCCESS in 1h 53m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6daac7bb900e7734fe772207eedd3b51e33276d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 18:10:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59c9ebcfb1540911f73d46b6d391f465263a8908","revisions":{"59c9ebcfb1540911f73d46b6d391f465263a8908":{"kind":"REWORK","_number":1,"created":"2026-07-20 13:23:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/997987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/997987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/997987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/997987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/997987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/997987/1"}}},"commit":{"parents":[{"commit":"29314ba6f30cb991027a3dcfabbed8d800827ad7","subject":"update constraint for oslo.messaging to new release 16.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29314ba6f30cb991027a3dcfabbed8d800827ad7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-20 13:12:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-20 13:12:57.000000000","tz":0},"subject":"update constraint for oslo.config to new release 9.7.2","message":"update constraint for oslo.config to new release 9.7.2\n\nmeta: version: 9.7.2\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia7ece8e3b5ac0211d3cf0a7a954a52e6ca4284e4\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nChange-Id: I372bdce97ef2859c4c9b7235b5cee14dafd3f5a4\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59c9ebcfb1540911f73d46b6d391f465263a8908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59c9ebcfb1540911f73d46b6d391f465263a8908"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
