)]}'
{"id":"openstack%2Frequirements~998092","triplet_id":"openstack%2Frequirements~stable%2F2025.1~Ic9ff50303fe399af29ee7559ebfd1391a0f5433b","project":"openstack/requirements","branch":"stable/2025.1","topic":"new-release","hashtags":[],"change_id":"Ic9ff50303fe399af29ee7559ebfd1391a0f5433b","subject":"update constraint for python-neutronclient to new release 11.4.1","status":"MERGED","created":"2026-07-21 08:48:56.000000000","updated":"2026-07-21 12:10:20.000000000","submitted":"2026-07-21 12:10:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998092-new-release","meta_rev_id":"8ac1aa93b56c3b1260b10cdc05dd639a872b78c4","_number":998092,"virtual_id_number":998092,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-21 12:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-07-21 10:33:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-07-21 10:33:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-21 10:28:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-21 10:33:14.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"9115dc7ffe7ce429e24c7a86d94c334d241f4f60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-21 08:48:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"901dbda5af971cd028f31122d6cba95b38537ed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 10:28:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c803cd34efe46d7a5588e2651db07f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/911c0034210a46b584646ed3b2eeee54 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90fc3b3e02434e60a08023a32853a15f : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dbd5e221a8d44b45a5f1a4b20770453d : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e8d8a86c3054b3a8becccf694552381 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5eb51470ebb348a5a4a76d5d66e3ff40 : SUCCESS in 3m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fecec4b210964d8a82b11b11d0eb0f55 : SUCCESS in 17m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2e2f9ee983824782a9e00a6637017a84 : SUCCESS in 10m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/06f00a49c8bc440a83f3e5bd5ffb609f : SUCCESS in 18m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/28c06c42ccbd4d9ba25c20d18bb602d7 : SUCCESS in 19m 27s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/149f7b64d1314a2083f945244636ac46 : SUCCESS in 5m 05s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/198eefbe49424030b587d2f9bf1b8e63 : SUCCESS in 5m 53s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/94963383f05e4a8b845ec6aff23c244b : SUCCESS in 4m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7f27b2dfa3d34b8ea3c8319f77bff61f : SUCCESS in 12m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/08c4444901cf463b8119de9718adb5bd : SUCCESS in 5m 11s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1ff898f6b2bd4505941d2f487e26653b : SUCCESS in 6m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8f1d18ff743c483c85662cd45bc7bb24 : SUCCESS in 9m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6c5c810cceaf4a22b57ea84b8bf7c61d : SUCCESS in 15m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ed01b15160548fcb810560a03afc21c : SUCCESS in 8m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8432c9055dda466bb54777f3320dfefc : SUCCESS in 8m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d23cb626a44f4078ad0a532c58ca7c2f : SUCCESS in 15m 34s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0db24d96f4884c55a498369d4f82dd76 : SUCCESS in 3m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/286e168f5a384bdbb5ccbe8c7cbd7f8b : SUCCESS in 15m 04s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f068157763c14aa2858e9e9d381be1c5 : SUCCESS in 4m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e173212644534cb4838b6ad1fe2a034e : FAILURE in 5m 55s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/fda18b34ab5741a19ff2dee57e83cc3c : SUCCESS in 12m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cda70808699f4caeaf7c9d367f1d7d5f : SUCCESS in 27m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c9f0620620214b20be02d8a131e7738a : SUCCESS in 8m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a317997d1f544107a63e878d7f19f559 : SUCCESS in 14m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/79fcdf90adc04f2aa4a924fccd0b76e6 : SUCCESS in 7m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/70c4db1b92434557a90128f34bd4e140 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be37ff869e294403b1a1051b8bc8027d : SUCCESS in 29m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/23ac8c51df254967833361f0ac44c080 : SUCCESS in 8m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fa70f7fb61834fa6ac7966e71efe1234 : SUCCESS in 6m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da383620bc0248f2bf0022d7b1e72af4 : SUCCESS in 6m 00s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a2df9e9091fe40f6861fb3a8c34f2d59 : SUCCESS in 14m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/adf438994ebe42f6861d3eb993e62760 : SUCCESS in 1h 04m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54de483cfe70484b8b22244eccc12c6e : SUCCESS in 1h 36m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7e472e2193cb2b904c5daff667deb56c843923b7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-21 10:33:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07ec95d36e21dad1fe1d78fe82becf8e4c54d8da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 10:33:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fee98d43346c3ba2ed0dfd169bfd91ea88101a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 12:10:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a00bc71f9234fdc9ec8a4910697c9a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab2b1938e7da4a9d89300cc0a8a4cd75 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3985b36a4c47466283d2edd94c4a0ac6 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80c8ebeb6dfd49f3825e181390840b04 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa8c2c8b8e254debb545ec7bf3d03c2e : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fefe8182f7cf4f289d353c0f2203e1a2 : SUCCESS in 3m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9ba8c56575ff4882bb09021027a0c1b5 : SUCCESS in 17m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e05c7c3f038b4e6d97e6ccee5dc939f2 : SUCCESS in 10m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/751d734e16f3497cad0fdb8b1807d7a9 : SUCCESS in 18m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7b9272cb0ad64505b6f2e1916b8e88e8 : SUCCESS in 21m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/dd334609cfae4e44a3955a948c494c7a : SUCCESS in 11m 28s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/80b52d55e3164aadabe19836836678db : SUCCESS in 5m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1ac0c2a3ccb349c89d4577282ae82465 : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c69ef696cb0d492ba143426703538918 : SUCCESS in 8m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/67ad2b0db5ea4e74841c984df84e1d9e : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e185f88bf23b44c089d0e8fc69724a75 : SUCCESS in 6m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2732d96994684ebbac643301ad7f1d91 : SUCCESS in 9m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/970bfcef72184c58a69b2903e67b6980 : SUCCESS in 2m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6760736890fa4640b814c7da13b85aca : SUCCESS in 10m 51s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/525496ab88f94c42853e371176d4efc3 : SUCCESS in 22m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f54fefa21b4049f389f355f5518246ef : SUCCESS in 8m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/866e498854b24bc7a4f77b55f4d655da : SUCCESS in 14m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f5c11c1736914006853ce70cca65e131 : SUCCESS in 7m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/54113f033f9643d99b9c50db411f20bc : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8041de8f4fb54b25afe219bd06240460 : SUCCESS in 29m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/79da1cc0ec6f43a98207eb266d8451c2 : SUCCESS in 7m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c4b78d222c43407cb73c47ae7260ceff : SUCCESS in 7m 14s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/43c90ee360bd42e98ae2654530af6678 : SUCCESS in 14m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30dbb7b2605d44bda5b60531a2054f15 : SUCCESS in 1h 03m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef77f178081043b5a1dfc8b4142086ac : SUCCESS in 1h 31m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8ac1aa93b56c3b1260b10cdc05dd639a872b78c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 12:10:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8dfc7f29afb1eb1c5af247189c141370b82fc525","revisions":{"8dfc7f29afb1eb1c5af247189c141370b82fc525":{"kind":"REWORK","_number":1,"created":"2026-07-21 08:48:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/998092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/998092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/998092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/998092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/998092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/998092/1"}}},"commit":{"parents":[{"commit":"316aa24519e7a317073dab1bcbdc2b2f5b4f04ce","subject":"Merge \"update constraint for oslo.config to new release 9.7.2\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/316aa24519e7a317073dab1bcbdc2b2f5b4f04ce"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-21 08:38:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-21 08:38:55.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 11.4.1","message":"update constraint for python-neutronclient to new release 11.4.1\n\nmeta: version: 11.4.1\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I157a6fb54898fb70707ba3a10dd044eeb53dde5a\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic9ff50303fe399af29ee7559ebfd1391a0f5433b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8dfc7f29afb1eb1c5af247189c141370b82fc525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8dfc7f29afb1eb1c5af247189c141370b82fc525"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
