)]}'
{"id":"openstack%2Frequirements~998094","triplet_id":"openstack%2Frequirements~stable%2F2025.1~Ic8deab46e0368ee62c1f49ffbf7409a22cf42d47","project":"openstack/requirements","branch":"stable/2025.1","topic":"new-release","hashtags":[],"change_id":"Ic8deab46e0368ee62c1f49ffbf7409a22cf42d47","subject":"update constraint for tap-as-a-service to new release 15.0.1","status":"MERGED","created":"2026-07-21 08:51:53.000000000","updated":"2026-07-21 14:52:53.000000000","submitted":"2026-07-21 14:52:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998094-new-release","meta_rev_id":"6ce193db345b92ccc02a0b39fece736093eba557","_number":998094,"virtual_id_number":998094,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-21 14:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-07-21 12:51:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-07-21 12:51:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-21 10:42:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-21 12:51:08.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"48972e1d9a5cedaab6daa07957f555411d5bc08e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-21 08:51:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f22f6f63b7e37a35e9e1bd61c363e4888b8375f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 10:42:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e9ed0f031f544ab9370998706299310\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7d95a41d2dd4f92b1d39880fbb067e6 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdf19287a974431fa1536d7c17b10534 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5332cc12a9374130b063c8406caa0bf9 : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31cdf7ae32b945189f2dd8fb762094e1 : SUCCESS in 2m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/314d5cf001c948c7a5211e33422b2ec2 : SUCCESS in 3m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/56a8f81a4c2441c19bf249fe6126009a : SUCCESS in 19m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/488d49c62a35466f87bd20868967cb5d : SUCCESS in 10m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/45bee50983504a04bc9d943d14835c74 : SUCCESS in 17m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/af40d9e976b940dab464b99f11967ac8 : SUCCESS in 19m 24s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/37d12ff85c284427af7375722d7a38ea : SUCCESS in 4m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4fb3b24853ef46b39755b714ea7a9fcc : SUCCESS in 5m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1c331660f93844a6ba52911b31fc5ffd : SUCCESS in 5m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7cebfca1ca2c44f5954546bb73f33be9 : SUCCESS in 12m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2470a6fd4bde42cab3dc6cbe44b14abf : SUCCESS in 5m 00s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6bb9378a5e844df3a75a8fa4bb50ebfc : SUCCESS in 5m 59s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/70ebbc4870324590b5e02a441f965adc : SUCCESS in 9m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cfdefc466f574190b7f907fac9ba8676 : SUCCESS in 15m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af22ffe1149a475f91ea81ef5f208f7f : SUCCESS in 6m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cb27d5ab083148a0b97d0940d0e825a4 : SUCCESS in 6m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b0e8a1bc6ce64d28819579aa98e96cec : SUCCESS in 16m 51s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b55b694e573a4d9ebd758f1c674ad17e : SUCCESS in 3m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9c40ae0215df44109942a0cae6491229 : SUCCESS in 10m 27s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/73b881ac67cd44678d47fb50232650ad : SUCCESS in 4m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8a43b9b945214bb8b88333f3ef414418 : FAILURE in 3m 40s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/53de5c9aca73454c84b1524465b885ca : SUCCESS in 12m 21s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3142230622ae4f099e8871646ee82014 : SUCCESS in 23m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/840cada9b89a47c089cd59f26da12502 : SUCCESS in 8m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/54145a44c48c469387282c96e2879fd7 : SUCCESS in 15m 25s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/44546e6db65b40db8aa7898c5522f675 : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/22bf95d4aeb14df18f2060bb30ebf29d : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2fe7b45b5cca4ee9b5b15270ef02af98 : SUCCESS in 28m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bc83c1a5d98c485f8d6ca6601c20b54a : SUCCESS in 6m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/29f2eb62436a458da0ec4ef1650cedae : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4fc807f7c0f04c2d816c2835fe88a64a : SUCCESS in 5m 52s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1224b769feda492688b5ab2b61056f4b : SUCCESS in 19m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b82f2fe065664e03a3c35101315ea177 : SUCCESS in 1h 05m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a86e9dcd3d7946a7a1a37a31f2c25ecb : SUCCESS in 1h 43m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7095d30b263b9ba0d6404b13c5278f87e423b03d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-21 12:51:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01ba3a83e183713d984db6c959ada075f1650123","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 12:51:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d943e425f911ea9bc902b02e6019c65edbc2ada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 14:52:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74c85954e5e84fd5af05fb95eb3aee44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6effaacae9047749bb5224d1c6daa03 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a35779f002da4819bc5bca4f31893fad : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50766274bf0a423faaa458684c887a9f : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5a3158e658148a7922db86e7106313a : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/83145078ed1c4087a7b3f6d2d35583d5 : SUCCESS in 4m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c68b0987a8ab47269f07c859c0207399 : SUCCESS in 16m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/10a2ca3fef3244c28669df8b1b073d8e : SUCCESS in 12m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/214034c154f4434eb903be6895af5424 : SUCCESS in 18m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ebae6dc711294ee29893290ae0c7d191 : SUCCESS in 25m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e91b42808d5f4539974e3a6d043eb2c6 : SUCCESS in 12m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/48088bfde7cc41b989f45e7466ed1591 : SUCCESS in 5m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f977edc7a59a415b8328d9c90ea9960c : SUCCESS in 5m 38s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/76a91c42d50343f288061b77a8ca8fe4 : SUCCESS in 8m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0996840edf674108aba09a821c2ea684 : SUCCESS in 15m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4dea06c03b0f4d3fac217ca6ff1a8934 : SUCCESS in 8m 20s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5caa42361c6a42dc8a50a8c77649ae0f : SUCCESS in 8m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5b85737a8b78480ca0aad322f63717d2 : SUCCESS in 2m 31s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a4747e38cf114902ac6213fe2bf36372 : SUCCESS in 6m 46s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f9c23bc476b04db38d47b8484975b8e9 : SUCCESS in 21m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dfb7168078cf4981ab43a46acc9d653e : SUCCESS in 4m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/95e12b5f99af427294637a9bd08cd9eb : SUCCESS in 12m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a188af783258465ca30c60e7fbec61d6 : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/04f841998ab24939b837225dede70d0c : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8587a1a8e724e80a26861f20be7dc4e : SUCCESS in 23m 09s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/41fe1cef672341ad8b0d3d257b45153f : SUCCESS in 7m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1cdd6f3961384dafae405aabc0d889a7 : SUCCESS in 6m 06s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/69b8744eb9db48f5b93e160dd054acdf : SUCCESS in 7m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84acf361271c43d2a440b8dd72e87146 : SUCCESS in 51m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac49ef95c7934c11b7c2ca4d519f29f9 : SUCCESS in 1h 56m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6ce193db345b92ccc02a0b39fece736093eba557","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 14:52:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be80517795ad712893388565fb76284e1651e662","revisions":{"be80517795ad712893388565fb76284e1651e662":{"kind":"REWORK","_number":1,"created":"2026-07-21 08:51:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/998094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/998094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/998094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/998094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/998094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/998094/1"}}},"commit":{"parents":[{"commit":"316aa24519e7a317073dab1bcbdc2b2f5b4f04ce","subject":"Merge \"update constraint for oslo.config to new release 9.7.2\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/316aa24519e7a317073dab1bcbdc2b2f5b4f04ce"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-21 08:41:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-21 08:41:51.000000000","tz":0},"subject":"update constraint for tap-as-a-service to new release 15.0.1","message":"update constraint for tap-as-a-service to new release 15.0.1\n\nmeta: version: 15.0.1\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: stable/2025.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5a5ca73759e00c453874350d3f610d7503496dbf\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: Ic8deab46e0368ee62c1f49ffbf7409a22cf42d47\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be80517795ad712893388565fb76284e1651e662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be80517795ad712893388565fb76284e1651e662"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
